)]}'
{"id":"openstack%2Foctavia~726494","triplet_id":"openstack%2Foctavia~master~Ia343d93accc0b313d979029584ab68f270421126","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia343d93accc0b313d979029584ab68f270421126","subject":"Properly validate AZ-profile networks","status":"NEW","created":"2020-05-09 01:28:38.000000000","updated":"2021-03-24 16:29:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6e0713418c1c5a7405a5bf962469b6243c72d6a1","_number":726494,"virtual_id_number":726494,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-19 10:48:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-19 08:41:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 04:02:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-11 21:12:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-12 21:08:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-19 08:41:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"95057f098380b7cde371e5ff4de145256ea8be74","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-09 01:28:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f80d82edf3c30344362d29e3b98320f588161fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 04:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7853a1b87fba4543941994a0cefb023e : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d8a98c8a0624b64a3aa679091457f8e : SUCCESS in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1be248ff4a4f2aa297954129d514cd : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea08fb5def1842dc954355427aa7f789 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a4156fe86a54a63b8ed7918f2c91161 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a781530873542249280bcedb95060ec : SUCCESS in 11m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/40cc6c8b05a146a5b86a33294524c0e2 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62124be2e75146ecab65301cbaac515c : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1972e025d1994d8492cd63d522bf5f2c : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cac9b1b7dae4409abd70d6d6776f2281 : SUCCESS in 46m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4760bb61e609497696cbe63be0cff73e : SUCCESS in 1h 06m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2b60e90f5944f38b3820518c9043631 : SUCCESS in 49m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6076d65f257643b9b0417ba72d7fb613 : SUCCESS in 47m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f8cc98c91a64a9fbe0dbaba7521feb6 : SUCCESS in 47m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/39068afd360b491c9e5c45f10b1ea63e : SUCCESS in 1h 46m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4b71c7d82d24c09b7192efcb01aeb30 : SUCCESS in 1h 01m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/582509f71a4242d387fc3b21fb2a6da7 : TIMED_OUT in 2h 32m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/20bca6b414a747d392dd6a24a4a16593 : SUCCESS in 1h 07m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2bf1a06112f2432baa8082da276da35ba1dc5bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-11 21:12:05.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+1\n\nShould have a release note.","accounts_in_message":[],"_revision_number":1},{"id":"19bc1ca35c78e330f27abfc94905409ca32d867e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-11 21:25:09.000000000","message":"Patch Set 1:\n\nReally? IMO this is just a bugfix? Not sure what to tell people... Bug only existed in ussuri briefly? Assuming we\u0027re backporting.","accounts_in_message":[],"_revision_number":1},{"id":"2cb8f4332cab6bfbb7d2e7097bc6bbdaa28e20e3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 21:08:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7deb157329ac848396fb2f9581fbd5935cdca5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 08:41:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f15cad87eeec4b8710f53e9822f76f21d240fb47","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 08:41:52.000000000","message":"Patch Set 2: Backport-Candidate+2\n\nrebased on master","accounts_in_message":[],"_revision_number":2},{"id":"28ac7942b6a5c247c21a7331284d58b1be6fac7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 10:48:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5516ceb95509453097ce4b0676570b55 : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/565e8c1b8bac48dc8df8bfe9f33aaa06 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0326eb95f1f420584c5c95b13ee2b6e : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0b1e0db9ef0470ab109cfe440f55eef : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58f402865cab441eba9f576352f99005 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917acbc54aa84889aff6bde0cc471d78 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef55869ac47e4a149d0bf04df6ed77ce : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/728729b458174dc0b302b7b604daa450 : SUCCESS in 7m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ffad640baa034eff98e392655c46d0ef : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c5f7613dae31463eb5116cfbb5c089fa : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d97ccbea569e47968fd5d497dcbff689 : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92d0a6b3d6fe48b3b0923d714a152446 : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5701afd63e347bab4ca425491e6c6f6 : SUCCESS in 1h 53m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26304886006644aa96904c19b087ac81 : SUCCESS in 40m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1b4c4da474c043178ce2af33434c6ab4 : SUCCESS in 35m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea911e263d4e4b9ca2652e9c7459fb43 : SUCCESS in 33m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b557d27a5b474bc1aa07ba3da27a72bc : SUCCESS in 52m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f693935db904a06a18aeabc1fd190ac : SUCCESS in 1h 47m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e5dc1d601f045dd947cc540f30b62d1 : FAILURE in 43m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29261658588242d0a6139ddc823c999f : SUCCESS in 1h 47m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/217ec5b9b0014ec1936c5b328e9181fc : SUCCESS in 1h 40m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0b8d9ec6eead4461a1c8dff525ae121f : SUCCESS in 2h 04m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c5fce6e6de544aeba334ae1b109d4ccf : FAILURE in 27m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e0713418c1c5a7405a5bf962469b6243c72d6a1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-24 16:29:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd191a2ab2eccc0aa8337632c735ced57da339e3","revisions":{"1bb0ffc4c14f533ddc69c397d261038239aecb29":{"kind":"REWORK","_number":1,"created":"2020-05-09 01:28:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/94/726494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/726494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/726494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/726494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/726494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/726494/1"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-09 01:27:01.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-09 01:27:01.000000000","tz":-420},"subject":"Properly validate AZ-profile networks","message":"Properly validate AZ-profile networks\n\nThe v1 driver was accidentally skipped when the validation was added for\nnetworks in az-profiles. Also improved testing for both.\n\nChange-Id: Ia343d93accc0b313d979029584ab68f270421126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bb0ffc4c14f533ddc69c397d261038239aecb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bb0ffc4c14f533ddc69c397d261038239aecb29"}]},"branch":"refs/heads/master"},"bd191a2ab2eccc0aa8337632c735ced57da339e3":{"kind":"REWORK","_number":2,"created":"2021-03-19 08:41:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/726494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/726494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/726494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/726494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/726494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/726494/2"}}},"commit":{"parents":[{"commit":"71c7402bc964ebac8361b7522ae98c58c171eccb","subject":"Merge \"Fix provider driver quota handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71c7402bc964ebac8361b7522ae98c58c171eccb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-19 08:29:07.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-19 08:37:02.000000000","tz":60},"subject":"Properly validate AZ-profile networks","message":"Properly validate AZ-profile networks\n\nThe v1 driver was accidentally skipped when the validation was added for\nnetworks in az-profiles. Also improved testing for both.\n\nChange-Id: Ia343d93accc0b313d979029584ab68f270421126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd191a2ab2eccc0aa8337632c735ced57da339e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd191a2ab2eccc0aa8337632c735ced57da339e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
