)]}'
{"id":"openstack%2Foctavia~726567","triplet_id":"openstack%2Foctavia~master~Ie71db27dc383c93496c1dfd69f486a4fd02b597e","project":"openstack/octavia","branch":"master","topic":"bug/1877818","hashtags":[],"change_id":"Ie71db27dc383c93496c1dfd69f486a4fd02b597e","subject":"add the verify for the session","status":"MERGED","created":"2020-05-10 05:02:43.000000000","updated":"2020-06-17 16:12:37.000000000","submitted":"2020-06-17 16:08:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726567-1592410082445-a663f06a","meta_rev_id":"fc36aa6f175cfe1742d01231cc6fc462d96db5ed","_number":726567,"virtual_id_number":726567,"owner":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 16:08:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":2,"date":"2020-06-17 09:05:24.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2020-06-17 11:21:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-17 09:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-06-17 09:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-06-17 11:21:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-06-17 09:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2020-06-17 11:21:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-10 09:54:34.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-05-12 21:31:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-13 08:54:17.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-05-22 04:35:17.000000000","updated_by":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"reviewer":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"state":"REVIEWER"},{"updated":"2020-06-17 00:12:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-17 01:19:24.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-17 08:25:13.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-06-17 11:21:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-17 16:08:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f303f8c60cd941694cf94df74ab00e504111a11d","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-05-10 05:02:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"646a23e95caf7fd104656b70fa03425c57ce6b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 06:41:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc01bfc39bef494a85a76526fb0e5d67 : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb18882f61e2499491ced3843eb54d6d : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a33eab4afb914bdf9438f72b502fccbc : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df93238542a44ab7a2b75602b1a6194e : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3aec676282f8405e9e22946b9492704a : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1a233d876d400aba24070fb3e9ac8e : SUCCESS in 11m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/df69a9356b134692b91a6521b7a1b7bf : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4d74f75b840e4a73a5f16b94fea2e8a9 : SUCCESS in 9m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/631e2685e9eb495ca7763387c583ad2a : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6929ae65c9a2400282e5d0734ba4004e : SUCCESS in 42m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/364b929470f84bf5bcd360cc414490c0 : SUCCESS in 1h 03m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea72e8869014495abe95463bd6c7291f : SUCCESS in 45m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/34823d6e373b4afaab8ce0499783cbc9 : SUCCESS in 43m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd5a349aba8f4d3983463d7204d508e4 : SUCCESS in 45m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3b8f575c82149cf8f62bbcf8bc41c5d : SUCCESS in 1h 37m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e59116f347d45548cad7f40ab3f8179 : SUCCESS in 1h 07m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7395ae5d712d462187f39cad4f1a03d3 : SUCCESS in 1h 10m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b33eeda899fc40698473370a266bc832 : SUCCESS in 59m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4929813b2977c9f1599c427581f52c3a21c3f84b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-11 21:06:46.000000000","message":"Patch Set 1: Code-Review-1\n\nThis fix looks ok to me.\nCan you please open a story (https://storyboard.openstack.org/#!/project/openstack/octavia) for this bug and add the \"Story: \u003c#\u003e\" \"Task: \u003c#\u003e\" to the commit message.\nAlso, please add a release note using:\n\"reno new fix-barbican-client-verfiy\", update the \"Fixes\" section, and add the release note to the patch?","accounts_in_message":[],"_revision_number":1},{"id":"934eca8af1fce8dc4a669228ea7349d0e0a1d46f","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-05-13 04:08:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74b375d7f6224bd5d6de9cefa8b68f336d985931","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-05-13 04:10:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"753fddadc0243683e146a96085446b450d85522f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 06:32:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb841860d6214e50875a60a6a53ac392 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/700ce7b9651a495b82ccdd49763cbe9c : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c324ebebd0492887ee954e1018d29e : FAILURE in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f62ccf4e64847ac88af7f815d8e7a6a : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/136d87fddb5a42568988c5c9796455e4 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b2386fc8ce46f28c20eabb860f0ab6 : SUCCESS in 14m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b0a3ff76d34426b8510679b572a1668 : SUCCESS in 5m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/097bc08d9cdb4d78babe0d013c51abb8 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bcbec0a266cf4086a506ccba7a96d0cd : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eca5f07d88d84d73aef9e5f6c86cffdb : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d30b41f41a2424eb023c6dce2148cdb : SUCCESS in 43m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/298c0e13b98849148aa5c10c13d8ba21 : SUCCESS in 1h 04m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/65cd13f788194ec9942488f8ec9663a6 : FAILURE in 52m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c77983058e984d23afc30e1ff130512f : FAILURE in 54m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/529392eb4705428986e2b9831cee9f9e : SUCCESS in 1h 11m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10d9fe2a91b74a11880bbde3be5a9698 : SUCCESS in 1h 40m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8212b46b4acb40f28c1b375d338f296a : SUCCESS in 1h 44m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d9267793ceb04c4b9237ec02e6f561b4 : SUCCESS in 2h 14m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86d262acadf846b1b84ba08593fdaf40 : SUCCESS in 1h 35m 51s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_quotas.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_amphora.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py","accounts_in_message":[],"_revision_number":3},{"id":"5cc3c6a6885b07a8646e3216cbe5090708c40dca","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-05-13 07:15:31.000000000","message":"Patch Set 3:\n\nwe need to fix the issue on master first","accounts_in_message":[],"_revision_number":3},{"id":"4f43c3affa436e8cf599e886bc93d71e9c8822fb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-13 08:54:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2ccb565ca80130e97c00d9ab8e9f6ae1dbaba9d0","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-05-20 02:22:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"676890228228a397abfc628f9c7e4cbce2961226","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-05-20 02:22:54.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nthank you","accounts_in_message":[],"_revision_number":4},{"id":"09554e26d69e1b0da821116569bd36629e0e89ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:47:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee3b2cf4b3c141089db9c589c77d5012 : SUCCESS in 9m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7b3062aaa4546ae90d9f6fe1909320b : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a1299017714bac9f093c934184cc6b : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de4387174ff54dc58464caf7b33a728e : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26ba783d59744957a2c3428e20e860b2 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2009702329fe4ec08feb1e278d619786 : SUCCESS in 14m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8937b12e094d4c3db071fd92422d8d40 : SUCCESS in 7m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/41d0c69672c44e328d0cd0baa0077312 : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4051292c2f74ea8a8b21c88fc756128 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22263e77a9974cfa8f4c1957c394c4c6 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4e9b20127924b09b8c14e9448a455e2 : SUCCESS in 41m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b80ce409ff00442c8dc209f25adc464b : SUCCESS in 1h 39m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df46484c4b5c42deb008af472c1601bb : SUCCESS in 50m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/71aa56fb2532477da1cc8fdccadd5acb : SUCCESS in 53m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6004255618734ce98c891a03b58311f9 : SUCCESS in 52m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2ad3207da904d40a92715566f056899 : SUCCESS in 1h 28m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/377ad279da5f479590faf281edc3e719 : SUCCESS in 2h 01m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9a2af30b1f84450e9f999751ad4e5ad3 : SUCCESS in 1h 23m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ff9adaec416543e0afcd0cb8b03569f1 : FAILURE in 2h 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"91feb180000b75ca57e8f6a08535c643e0309a6b","author":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"date":"2020-05-22 04:35:17.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"946a00cffb861a0d03da68ea19c8967026838c42","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-06-15 09:56:45.000000000","message":"Patch Set 4:\n\nIf anyone can help me merge this?","accounts_in_message":[],"_revision_number":4},{"id":"886a59d922240864cea7bd148cacd0ab8545a4d8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-17 00:12:08.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"c4eceedba914696abdffac0b22db3a12268b1b81","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-06-17 01:19:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"247765ffe9d72ad7d358c592335462cd034b398c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 01:19:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"74e7d25325df0ded7054540567661bb4fa77834c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 01:39:49.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7520a0d835314121b8552e36a09002aa : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a818ed2684f460eb4a9d53a061fc94a : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7feb3103abab40999f96f60737e1dfd8 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89f693d03516454b9bc9ffba0683988c : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d5f871ac4244278a7af889bcdf86c2f : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff87d90315b4f969d076b269f372f67 : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75743a506cc546b8b58ab1201c06a074 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d10b80645e774207a6200fc70926d634 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8058093b1b043138a6f34ca1f66832b : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83c3de7ee2f641878889babd3f126da5 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/939630e6f4d34d579d3321e6c65af0f9 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53c0a3924d20472885bfc791d424fd8f : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1510839d1c834b07adf1655ff3223ee4 : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":4},{"id":"9088ad9c28ac42c3be284fd038921cc0e8e590b5","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-06-17 01:56:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d4f91bb663cbd41ab875ba5fd7be0efed86727aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 03:04:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a036a7484e85402eb764438ae98352d0 : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4886f465ec004cb996e1317fb7f9ff1e : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7531d9d87868475aa8c0ae29d6a8039b : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0325e336bd2148059d14cfc9f571d93b : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb1a6a5da7644bbbbab986265012b8bb : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92bc152fe0514ccb82a2fec6a1a7eff5 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7eeff191c844ca5a7dd5ab5cf1c704c : SUCCESS in 4m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ff012510bcfc48e08ee582f1ade26b33 : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/19b8dc21ce1f41dab30ffc38c612eae0 : SUCCESS in 8m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2659326f44e04162943e6f514bfa15ac : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/650ee59a7f054f6ca669b602d3dc89e6 : FAILURE in 24m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5249a51757b4bdb8a948419a87359d2 : FAILURE in 31m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7557859adbe498bbf5ade9ee01cd9c9 : SUCCESS in 1h 06m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6646027fb8c7404c951f16275f7ecd2c : FAILURE in 41m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d04b253ffa544be2bfa29497ece4140e : FAILURE in 30m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/15c3a3a9879e4d53b5e299c6c077b80d : FAILURE in 11m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ef941cb86194d1f9132fc60904a01f3 : FAILURE in 34m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66d9ccd7f4d74e79b5e9e41c64e725c1 : FAILURE in 25m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/afce7e0b0290493aa41bf5a8ddf7c6c3 : FAILURE in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ddee6c4e0e2d9b4e122f16918416bfb7bcde4fd0","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-06-17 03:09:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"80e83442e0c9d4b87eb58d1962f4a5c76b7ec9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 03:57:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc59bbfd8b164cc4a097d22f899415d8 : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12de9a272e7a439fb3feba553bd4a295 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00044ec27993450eabcae73cc2571a7d : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7138ce39501a4688a774ab18ea9371ef : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31a50060475e4f659672de9ce25c1758 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace132a411934f018d673d44f11cf639 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cc5d782693a45f9a7fa96150d37276b : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/faec54eb4e2b415b9d62d30206bb4f6b : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b79c82f50b3b4421b38c4ead3f8025c9 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23feec35f857476ab26f5b5ed7db7df0 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/938304bf8b2746f2b6cf2952cd9a9f87 : FAILURE in 22m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/742412cc869243759a9b1f5837d9cf8c : FAILURE in 27m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a74896d5c344dc6a1a524726ac21edc : SUCCESS in 36m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9a9b059ac24b4f62b80de738f3770047 : FAILURE in 46m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57710bceddc143229f5a60325067559d : FAILURE in 26m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c896329bea05450f824d83908b779d52 : FAILURE in 9m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e0b7de3ceb364f22af673ce1faa1e8fc : FAILURE in 30m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4f8864fec1245239b3b9adbbb898de5 : FAILURE in 24m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ac51ab2f696646088cafc363ce5f0664 : FAILURE in 37m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cd8e8608c9ef3399a0a4d8f08a10c98ac1fbce6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-17 08:07:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"fe788f0a1168d175b10d1f3319039f8cf5ea2faa","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-17 08:08:03.000000000","message":"Patch Set 5:\n\ndevstack@o-api.service: Failed at step EXEC spawning /usr/local/bin/uwsgi: No such file or directory","accounts_in_message":[],"_revision_number":5},{"id":"03498abd03f17b5e2f5c02fa8194a822f40efcd6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-17 08:25:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0ac17149988c8dc10726f420b805c37a062a27d9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-17 09:05:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5e5d3ec115ad3eda408fd5d6116ba389bb58ff59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:11:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77d4494a05e64d1dbc484ae7328ff67d : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbebe29b328b433d9dede3d79ad0d75a : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b34a6ca885448ca2907bf434ede30f : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de479c2c45894d5f9686ba91dfed17dd : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97b3ceefdf2c4610880ef9d4b35afc06 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bded42f413c49dea1c6386af9687c19 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/717ffe5fddcc4b6796600fdb7892d433 : SUCCESS in 8m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db5ea80072fb4ce2bb4244adcc8247b6 : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/87bf5227652d44ab86d0a128b0486a52 : SUCCESS in 10m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e903a3f8326c449e81f791c9f2ba4661 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86ad566c0702428fbf7ea8e3539be374 : SUCCESS in 45m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18ca1c019c2b441799271f6c3b739b30 : SUCCESS in 1h 52m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/acfa532eb28642e7916106f2a87c0750 : SUCCESS in 1h 04m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/54fe6c291cac4f5ebefd9893811cd4a3 : SUCCESS in 58m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d98366e580e24d9788c9631a9ce13a62 : SUCCESS in 1h 04m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0faa82d7ed0c4d038a157470b6e01af2 : SUCCESS in 1h 35m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a14024e1b92347bebcee970b03f59448 : SUCCESS in 1h 05m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a03fe91c3824bcbb6e0a361ae1662f6 : SUCCESS in 55m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8669a3defe2a42e2a5b28908b6733681 : SUCCESS in 1h 58m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"74cd8d5fc87109eb45e0b8bb96632c383e4e8ed4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-17 11:21:41.000000000","message":"Patch Set 6: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5a2f829f9f5bc7431ac7e20e1a591fc08ace1ff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 14:04:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d8263ed54afb57c8cd4fcb2bef78c5e8bd256baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:08:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaae408b2897433285727cb2bfbb0648 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699ba1f25be54f65b5d6d87a3489bdfe : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16c53a6cd1f04afd9622e0e142a542a4 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4e9f9d71eae42149472c97dd500fcfa : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bef5df62d7d4b2098b99d76edc2b60d : SUCCESS in 15m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66af455a4bc34a08a55a001a9e837f74 : SUCCESS in 11m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac8ef216812b4d589c34f359dcb4f4cc : SUCCESS in 10m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f70dabe2c524bf5a105b6551a883a28 : SUCCESS in 46m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01a44f96153b4c76a65d766c8a8ab54d : SUCCESS in 1h 54m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ace0da5b0e646a68b51c4d277c92d29 : SUCCESS in 1h 05m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/573def7b25dc4eb69362360ef538ec39 : SUCCESS in 52m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2daa95164074c08970028575234480d : SUCCESS in 1h 01m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5faf30c870b84ebab10f0a1bba69f489 : SUCCESS in 1h 31m 16s","accounts_in_message":[],"_revision_number":6},{"id":"9edc5415b78cf3e04a548c153b13c1571a4fa13c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:08:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"c432a245af35bfed4babc09db8001d3c658da131","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:12:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cff2a098449429fb0b13efd47137143 : SUCCESS in 4m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1757cc69f37410d9851a2a0321c860f : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","revisions":{"96dc870502d974b873e30e7571f10dfa65fdbec6":{"kind":"REWORK","_number":1,"created":"2020-05-10 05:02:43.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/67/726567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/726567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/726567/1"}}},"commit":{"parents":[{"commit":"5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd","subject":"Merge \"Migrate grenade job to native Zuul v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 05:02:20.000000000","tz":480},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nCloses-Bug: #1877818\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96dc870502d974b873e30e7571f10dfa65fdbec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96dc870502d974b873e30e7571f10dfa65fdbec6"}]},"branch":"refs/heads/master"},"ebf617a63053683c25d4455d9f99c3d894666992":{"kind":"REWORK","_number":2,"created":"2020-05-13 04:08:23.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/67/726567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/726567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/726567/2"}}},"commit":{"parents":[{"commit":"5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd","subject":"Merge \"Migrate grenade job to native Zuul v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-13 04:06:58.000000000","tz":480},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nStory: 2007662\nTask:  39754\nCloses-Bug: #1877818\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebf617a63053683c25d4455d9f99c3d894666992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebf617a63053683c25d4455d9f99c3d894666992"}]},"branch":"refs/heads/master"},"a5ebf59d7c86280959c608c027d2b02c40ca0ade":{"kind":"REWORK","_number":3,"created":"2020-05-13 04:10:49.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/67/726567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/726567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/726567/3"}}},"commit":{"parents":[{"commit":"5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd","subject":"Merge \"Migrate grenade job to native Zuul v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-13 04:10:14.000000000","tz":480},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nStory: 2007662\nTask:  39754\nCloses-Bug: #1877818\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5ebf59d7c86280959c608c027d2b02c40ca0ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5ebf59d7c86280959c608c027d2b02c40ca0ade"}]},"branch":"refs/heads/master"},"b4f75ac614b338ee250d29a81635486ccadc5045":{"kind":"REWORK","_number":4,"created":"2020-05-20 02:22:14.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/67/726567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/726567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/726567/4"}}},"commit":{"parents":[{"commit":"bcef33a3ff4b5cf67964919c9826514f8ddceca1","subject":"Merge \"Add TLS cipher blacklist to octavia.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcef33a3ff4b5cf67964919c9826514f8ddceca1"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-20 02:22:09.000000000","tz":480},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nStory: 2007662\nTask:  39754\nCloses-Bug: #1877818\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4f75ac614b338ee250d29a81635486ccadc5045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4f75ac614b338ee250d29a81635486ccadc5045"}]},"branch":"refs/heads/master"},"cd206467e6f7053d0729facbb0cd76f5c81637fb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-17 08:07:33.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/67/726567/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/726567/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/726567/5"}}},"commit":{"parents":[{"commit":"24acbe099bacacbe1840d8964bd32309d47a7081","subject":"Merge \"Improve terminology in an old spec file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24acbe099bacacbe1840d8964bd32309d47a7081"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-17 08:07:33.000000000","tz":0},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nStory: 2007662\nTask:  39754\nCloses-Bug: #1877818\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd206467e6f7053d0729facbb0cd76f5c81637fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd206467e6f7053d0729facbb0cd76f5c81637fb"}]},"branch":"refs/heads/master"},"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-17 09:05:24.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/67/726567/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/726567/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/726567/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/726567/6"}}},"commit":{"parents":[{"commit":"e5951ced5fe51262a57bf6a9717e75b8b6c63ca5","subject":"Use uwsgi binary from path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5951ced5fe51262a57bf6a9717e75b8b6c63ca5"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-17 09:05:24.000000000","tz":0},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nStory: 2007662\nTask:  39754\nCloses-Bug: #1877818\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
