)]}'
{"id":"openstack%2Foctavia~726741","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Iedc1d5c5b603753b171fe17816b4d9c3aff1a16c","project":"openstack/octavia","branch":"stable/ussuri","topic":"native-zuulv3-migration","hashtags":[],"change_id":"Iedc1d5c5b603753b171fe17816b4d9c3aff1a16c","subject":"Migrate grenade job to native Zuul v3","status":"MERGED","created":"2020-05-11 07:33:01.000000000","updated":"2020-08-17 11:24:20.000000000","submitted":"2020-08-17 11:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726741-1597663290312-0b925b31","meta_rev_id":"dcd0ea618891dcb1d4db4a5d33695ead6fd6f675","_number":726741,"virtual_id_number":726741,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 11:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-16 12:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-08-17 08:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-07-20 21:39:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-08-17 08:23:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 16:09:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-17 13:19:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-07-16 12:26:35.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-20 21:39:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-17 08:23:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-17 11:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1db833ef8974321f77e25cc789cc0d956124d4da","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-11 07:33:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5fcb0cd935c0e8117e27d2488be02797067fa4fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 09:43:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4b01fdb89a344feba200ee7182ec42f : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ded5d0cc7c894136910ebb6065d475c4 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffeaf7d21a194e80b6ac1d5ec3eed3da : FAILURE in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aed1afba2dc4c43b893224495605466 : SUCCESS in 9m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/192dfe90ac294a43a3a6edbf26e4d109 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13c91c214034ded87b8290543e5c991 : SUCCESS in 11m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ca5138d9f4954fb481d95da8854f8955 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/073c905bf443488faeadfee2042cc0dc : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/67eda9aaa8de473c9eb53443663b923e : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff97729f63574992b524f4b6fce42c7e : SUCCESS in 46m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fdaad2682764ce490b74c8dc4d9fac9 : SUCCESS in 1h 55m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ae86c96a15644548a92adf52fbde1b0 : SUCCESS in 1h 04m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a33ba384fe594641a5029208b079bdd5 : SUCCESS in 1h 12m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c30995ce6619442d91cef002ffdb88f5 : SUCCESS in 1h 06m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c36ad31ce234aa7a2251566ddbf6e02 : FAILURE in 1h 09m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/36a21a3861ae4966a5a871736e5499f7 : SUCCESS in 1h 46m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5fae1d3c93f646b1825d0b19353a1644 : SUCCESS in 2h 05m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e7935db1724d48d9814483dbed6e0628 : FAILURE in 1h 19m 25s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"c66328b88b1ca8d8c41b0cb289e9034285d1482e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-11 09:51:30.000000000","message":"Patch Set 1:\n\nLooks legit.\n\n2020-05-11 08:43:29.800 | + /opt/stack/new/octavia/devstack/upgrade/resources.sh:destroy:101 :   openstack loadbalancer delete --cascade lb1\n2020-05-11 08:43:31.744 | __init__() got an unexpected keyword argument \u0027tls_ciphers\u0027 (HTTP 500) (Request-ID: req-5e29e4ee-a251-4d74-af97-6d4e3801967f)\n\nhttps://zuul.opendev.org/t/openstack/build/8c36ad31ce234aa7a2251566ddbf6e02/log/controller/logs/grenade.sh_log.txt#25589\n\nhttps://zuul.opendev.org/t/openstack/build/8c36ad31ce234aa7a2251566ddbf6e02/log/controller/logs/screen-o-api.txt#1787","accounts_in_message":[],"_revision_number":1},{"id":"e4437c0daa48f455fe18e67bf9f3d5e1d9721714","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-11 09:53:22.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/#/c/723810/ will fix pep8","accounts_in_message":[],"_revision_number":1},{"id":"d605acd772509ffa336ed69b9bd6b955c2251f7c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-20 15:09:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e71a19faefa9ee036341b64adcc27829b4916807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:12:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72da41116b6542aa8799ed687bd3dc89 : SUCCESS in 15m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/715284a31e024a70a5f3c889aed36449 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e8328c78904d8f98ba22f979c8ff1a : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd7409e7265042a6a36907970d48216e : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1f1b5de69a94d47a1259a118702366b : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47efd467dfb4cdea106493dac5f3a3c : SUCCESS in 13m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6820cec1c9a34eb698fece82fa543167 : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f416d87d9a7d4579b53875bb3f87f781 : SUCCESS in 10m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/132d41412199489a85ed3a66d2163716 : SUCCESS in 13m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0378e81e093844738d7349a12c95fe27 : SUCCESS in 43m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84bca69689804e76832a43e8c7cb9f78 : SUCCESS in 1h 56m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eeec3121af624579b5db0d9999cd7cf9 : SUCCESS in 1h 03m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b15f2a4076a848fd827bef0ece485345 : SUCCESS in 57m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95a8248e15a5410895dc119b9414c456 : SUCCESS in 48m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab745463ec4e4896a8eead3b82c2c285 : FAILURE in 1h 20m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9431e0ab4b34448be2cbd310962c3ac : SUCCESS in 1h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e25ecc2869449e18bb5f27ead2e77bd : SUCCESS in 1h 42m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72937631f2b8463b8d458446423c1523 : SUCCESS in 1h 06m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdb7dd7067db929422242288d99d0e39d6bf410b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-22 11:14:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"af108ca01d998274c8d391dc1a83f0625867242e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 13:31:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e65d5cb3f5a14164bf1d16a8a18bf1ae : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/188ea9fcd4674e4689be258b9e727345 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34e87f3a3d254ecabf2fda552cd87d79 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c032699ca1454864aa4d38dbf80932a1 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6183606d065849519586c9413c789e30 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7b3cd63e0b47e9bf4aa05168d5d820 : SUCCESS in 13m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cdb2a5d58e1e4ed48fdaa49303a735be : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/065410db4c8840b18d09ad92af27d823 : SUCCESS in 9m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4de2a7441cff4b7a98d5d1736ca5e818 : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fe47a8bd17f438bb3b662c375d01472 : SUCCESS in 46m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac2bc37c34ab49cbbec19cf29266d58c : SUCCESS in 1h 02m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7e9dec819e94a388e0a4f460d87f03f : SUCCESS in 50m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/109c223e44d7401cab7bb3ff1465b184 : SUCCESS in 48m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85d4a21c83d94a66a72ae0e7cdcc364b : SUCCESS in 57m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98a9946bdebd414880e52ad1d53730ed : SUCCESS in 1h 52m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5be0a64ebb4f4021a1089df8878f467a : SUCCESS in 1h 09m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b445dd5d612c4832aec2fc5e874bbc22 : SUCCESS in 2h 10m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/25b9c520c05341e199606563efa67a30 : FAILURE in 1h 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ebc6a0904dd5e5af6e141715dca9ba1fce46d09","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-22 17:47:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f4def612202d9dc2def071e9e8ad9de3e07d6c14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 19:41:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e8c25dc6fb41beb109b54015620c5e : SUCCESS in 11m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/548369533bf347f1bb5d8a957281843f : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccd6fc6b5534cf1bcc1ad1e3b3f44e8 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/261e44be2b924920991e13af38702437 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/186c71accd7145898a2deb738f0820b7 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac2d995f0d941a8b52b5d5b69b5901b : SUCCESS in 13m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1467129e12a146c5b2be9f99b987fa2e : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1ffb935bd84b4b03a9abd6308dd49a51 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a7589dcdbb2464a9f578b8d8f49f17d : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2867af9798c14441bd95173e6fd551d1 : SUCCESS in 45m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c2034a4adfb409ba001300b8e9df72a : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b4298478729148d8aee11774433c4bbb : SUCCESS in 49m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cc3495067bdb410fbe83d11581c612ac : SUCCESS in 46m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51021ba46ef84fee921780811eec0e30 : SUCCESS in 1h 02m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/550769bdc60f4f56b179594cf8345b27 : SUCCESS in 1h 47m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4d90a0eabeba48208f868f6d9119279d : SUCCESS in 1h 37m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53b05ec919114033b16ac2f85b2923f3 : SUCCESS in 1h 10m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fad67674e9414910836a500bc28a8335 : SUCCESS in 1h 05m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5ece48aa7bb06e0daf8af52df5b1457fe7096cd4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-08 16:09:20.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"ba8972a5c917ab1850199fc636ce4f1635f6db9a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-11 19:11:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ec3c8954073db3814d348ebf571c6c1e23e372e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:12:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e34614278d730f3414f95879fd6d47056b60ad87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:11:14.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f30c8bfaa3c34e60970d703b37b21e94 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9c9e8602c94118918973c123de1232 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74deb07cc6804424899ed7c8b7b906ef : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6089a33e488b4e418221bafe769a6fc1 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f8e0d2e06544b396952761772d36f0 : SUCCESS in 14m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/344ae1288c5145d1bc450274ce51d7b1 : SUCCESS in 9m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff29d36ab4e0413daec0426547541bae : CANCELED in 30m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b22adb5f047b4812ba0c896d75c96d8c : CANCELED in 30m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/581ad896855d4062b71ca98e2ef1caf1 : CANCELED in 29m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3b30d0b23bdd4d678b3986663db7ef6e : CANCELED in 25m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54a11c5b8653456dabb4f2c4bb946cf9 : CANCELED in 25m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/36eee46cd7ec4d0195468f574cec2d0e : FAILURE in 29m 54s","accounts_in_message":[],"_revision_number":3},{"id":"bbaaaa4061799aee20fb74e4f650029114730208","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-12 08:09:08.000000000","message":"Patch Set 3: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"40a6b8fe9dc7a378ffad105e09e8b73e2b294015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 10:05:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e4731aa63d04d7f91774b2c4b704d58 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2be878ff00542d99c74d388ac8a53a9 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259767a195df4be09e25af8a6a2d8113 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31d68bd5bd30475aa98da4d5e75e5dec : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c48f1a620144c0bbae7962bf6ca5518 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f28b54095a04e939dadd00420a84a95 : SUCCESS in 12m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2711f9032649424697963a24b38d58f4 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d44e84e56bb24060a4892c98397c7ae7 : SUCCESS in 8m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf998a15a8ad43d88c7e20df6d664ca4 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3f0f35b8fb34fbeae136de38ccbfb21 : SUCCESS in 40m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de3a475d1cb145cd8b195dcce17f78a7 : SUCCESS in 1h 51m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55777c7be5da4f10a24fc3ae9d9fc206 : SUCCESS in 52m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d04ec59226284e0fb29cd79dd58d68df : SUCCESS in 36m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/515b86d1e7bc41fc959cb3c3e2bb3cba : SUCCESS in 1h 01m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0851e3b68074d40a9687424cb3c6eaf : FAILURE in 32m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a6caa5c7c07042578996c53ecc226f9f : SUCCESS in 1h 11m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0855ed05687443ba07297e90d04279c : SUCCESS in 1h 41m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/79687f5901fa40fba826d5937a3fe456 : SUCCESS in 1h 04m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78fcf61e9f80b38623664b0e58833c96aa495197","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-17 13:19:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d68d7742d4fc19c40754b8fa0f2e0133208fed39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:11:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/619f680d16014590932aa8b36c874599 : SUCCESS in 15m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad26149187074e18a661e3a8a882ea16 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875841338488430d92e955f958236ad4 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd4d25ad31c94b008d7cc8b6c97958e2 : SUCCESS in 8m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dd417a2ac14434cbe9603c69a4a37a5 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd88d4ebebf4d0c917a6be4bc19bbfd : SUCCESS in 17m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c31f1fa0b7a94f86a71af55ba48fc7e9 : SUCCESS in 8m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7615cad8c5fe4a9eba3256d44577ed4f : SUCCESS in 9m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0141f8ef55134172815191a082aa9dbb : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd9bc9d8129f44d184e3268f6af51f91 : SUCCESS in 39m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e1377c6db3c4dae8e1c93daa767f7d1 : SUCCESS in 49m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/21a769a2620e4f99af652ffa2ec36b5b : SUCCESS in 34m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4a3ea2a5b4fc4b5b937c46576c6065da : SUCCESS in 50m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98e3363d679645e38878ccfd7cc391a9 : SUCCESS in 59m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05201a162e0e4d47846302d77130a5ae : FAILURE in 14m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aad29965b6e141cfa54b87f9d76559db : SUCCESS in 1h 46m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb51c8a4ae5c49a2be77c60a76687079 : FAILURE in 55m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2564520fd9d74e41a9e571f2c3e4495b : SUCCESS in 1h 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"03eacdbb686ca5f1a177ce2b6a6b2c894d2b3088","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-17 22:18:48.000000000","message":"Patch Set 3:\n\nrecheck\nI think devstack is fixed now","accounts_in_message":[],"_revision_number":3},{"id":"b70d9bfb02a4748d14922d217d3ce8e5111461ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 00:41:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79d574894ee64216a6187a2add4daa8c : SUCCESS in 11m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c038d4faf674fadab881f818de5514e : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4820f880ba71400aafe0bfca948a208c : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/725b1669d8d245c2a0c4f7e1d850dcfb : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/beb0838503b3403590476672a88a396a : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd6a6ef9a78405e82247329ce7806b5 : SUCCESS in 12m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3c46c9c3ea424665bd22179b5bf75689 : SUCCESS in 10m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2474a9e27dea4cf8b9aaefa3a7995686 : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd1eb749fb0645af8b4811cbc7b525a4 : SUCCESS in 10m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1860f7814ec64c0ea906824a95a0ff43 : SUCCESS in 40m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e01538a7f8c3498289219a1e037610ec : FAILURE in 2h 17m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/874b753d4ad5446d9a7680477241b49a : SUCCESS in 57m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b8bfc1376987485b8a3a50ad1c1e73d3 : SUCCESS in 53m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/629bbf82a1d7401cb42837a266ef351d : SUCCESS in 42m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ac99080bc7f42eebcaae4b0ff7c6a87 : FAILURE in 15m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce936e4b49c84e819c9786e9182d587b : SUCCESS in 1h 48m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80ed7c45db7b496f99dcc3eefcf1ed06 : FAILURE in 55m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e8ad7b06bc954c6698249fec6b35229a : SUCCESS in 1h 47m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cea7e9241ec1b49ac890bf8dea906e8fc2aa775f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 12:54:24.000000000","message":"Patch Set 3:\n\nrecheck\n\nHmm. Devstack\u0027s broken?\n\n2020-06-17 22:37:27.048536 | controller | /opt/stack/old/devstack/lib/infra: line 32: virtualenv: command not found","accounts_in_message":[],"_revision_number":3},{"id":"4a4ff6518a430b61df6ab5787a7f020ba5d2e59e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:06:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaa26e09db7e4efdaae680ce6d5b94c2 : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1910cc3d3c654a3caf8b29ec35fc3f5b : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d2e063d1144807aaefd5112e03c24b : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c7606eb845b4a3f87c2b657859880b3 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67ebad3bd45248c0863874e23178e92c : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d504e37883a64388a9d8d4baed21e48d : SUCCESS in 18m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fd44ca5c254b427e81b31330743bfa43 : SUCCESS in 9m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b0e18c5929e14323b394b578ad70720c : SUCCESS in 11m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66d2beb0600e4e8bbcefb2426cd76197 : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98654173484d4aa3ba99f23eed7b01f5 : SUCCESS in 47m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5d0fa3a5c154f64b87ae92ee963ccbf : SUCCESS in 45m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/54ffa4fa14b3470486ac6705cf292e17 : SUCCESS in 1h 13m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/30ee58059e5d404fbc80f78ec55e4e47 : FAILURE in 1h 55m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdffe5ca1f9b4506a54524551145ff68 : SUCCESS in 1h 07m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca57fb737a46471a947362672eea9208 : FAILURE in 28m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a72937f8c7c348939012e0e6f780f023 : SUCCESS in 59m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec8056175e96405eaeb5114f6913ccef : FAILURE in 53m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d1f45a98b8d4428eb4b96bbf3ecc3dc5 : SUCCESS in 1h 46m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb68a9e2b838a7a0be894d202b82167eb619c77a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-29 16:55:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6cc9944e2e42db06da61e73326c15d7b5959e58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 18:49:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/933e32c7c2354744b28176491c12d8eb : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfb32eeafd0f4157814b4e79d2bb59a1 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17cb89d4cdd54ac0b43f9c4267f19977 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb1c4f58a3b1466bb41aa63f2f979951 : SUCCESS in 9m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8eea9874f38e49259b53ed5d7b3e498f : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de08886d5ad470c8ea3c517615ef8c6 : SUCCESS in 11m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/63eaae3b39374b09b0b5680aa6a113e9 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b6983a4376e04c2ea5bf6d830bc12ecf : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94c1ff6b725247e78c030acc2f506535 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/101cd8e873974e0280a90774ca1d5b8b : SUCCESS in 46m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dcb7db09e384c628395f3f7379f1644 : SUCCESS in 1h 48m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/475863c0a85d4374bec3f4ec89225cb9 : SUCCESS in 1h 00m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f9bcc4c36a594758a9732aeee28a39eb : SUCCESS in 46m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28d79c75ff494c0ab29d03f241233acf : SUCCESS in 1h 09m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b99abab72e74cae9562bef41433b341 : FAILURE in 32m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dad665565510452cb77e43ac1df790e7 : SUCCESS in 1h 02m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4dfba4f9ea34cd6b7cf43b909da855a : FAILURE in 58m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2857fe3322654e159f68300528c4dfd6 : SUCCESS in 1h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d8dad129172543734956db323aaaf28ec31ba5e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-15 06:18:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"46d0750a525666853bcdfa18a5d75dc81fa5615a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 08:31:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22de2381f2e24d6c944eb475dfc85e02 : SUCCESS in 11m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8b17bc231594dba9ff44ec0b600296b : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872f44b711ae4b5c9c0a4761dc100523 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e5fd0a7547a462ea7a10d1e2f8aefdd : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8febc90dedfd4505ad9661097643d6e6 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a82e5889bc6492ab2ac51d242e6584a : SUCCESS in 14m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5cf35b6affa64679b3fa11dbdc07e1d4 : SUCCESS in 8m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/54d10ef5426049e5821eea94596be80a : SUCCESS in 10m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0739e759235b45df9f59388a7868e814 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9766d93eb3604627af7f447d14bcad48 : SUCCESS in 51m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8e5923fc4ab4eaeb88d3b14c4e7e98f : FAILURE in 49m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18efc587fe1a41c784d1505d75c21eae : SUCCESS in 49m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/67e8508f055042599c0c8cf78b0a283f : SUCCESS in 1h 01m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8afe29b770d74ef0bd03a0709d857fdd : SUCCESS in 1h 09m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/00bf7abab968473bb41c6f9d0a3d65aa : FAILURE in 32m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a2e47bb64c4f4f2c8c32879e94626b27 : SUCCESS in 2h 04m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e98c1e2ceb64b03ab86e037e730eecd : FAILURE in 53m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19b100de67dd49bfaaa7f8697a3a3324 : SUCCESS in 55m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4fd2576f0a09204cff00a6b0418afdb4cf2c107a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-16 10:14:13.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e24ad89de03f5e05aa281713154b44f79dc15ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 12:11:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/788b557112164746a2503b141de2c128 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/941e75d5e0b64236ad74449d640f68e7 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ba52a34b0d4deca2e8c65fcd78f191 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e71b53055db0460c8e522b8911bec37f : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc2279f9fed1437b81ec10329efcf327 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f28da06f96a449990ef945510e67563 : SUCCESS in 12m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5e2098e8636a4028936cd584c1b0338a : SUCCESS in 8m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/58eb776712214c4da5462d27aedcf42e : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b899e1b0e4547e2816e6d5a3414da8f : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1513f83553334066942353e1b548cc43 : SUCCESS in 45m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb2a2095b2fb49e0b58fc0606f0e64df : SUCCESS in 1h 03m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e33bdf82c2434973914431910cda4884 : SUCCESS in 54m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/155b3763b8c848adbc3bd5d78d88e3ee : SUCCESS in 52m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e031a6eff144049908de9af331a79f8 : SUCCESS in 55m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f2dd976c8b714eb18236c32416d5ab78 : SUCCESS in 1h 52m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/db1d20ebd5ef4605a2d719112fc7e922 : SUCCESS in 1h 16m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d87cc0ef9d64070b574b75170e885bb : FAILURE in 1h 02m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3f27948da217408bb32d874897f54401 : SUCCESS in 1h 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0fe1a26beb912a8e0340cd4f5800b2d58806762d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 12:26:35.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck\n\nTentative +2 based on recheck working consistently.","accounts_in_message":[],"_revision_number":5},{"id":"e8433a561f97770e9442a0af984cd469d46ae68e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:28:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b9109691fa44e9cbe7bd2104ba8cedc : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47d0e4ad02f94591834a8aa1923f713c : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d4fb8f1ad14671b2b8dbaf9b564c89 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4c317d3f7194fc786b276976bd55365 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc8b35d6cb434614967b7e7b64011724 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d375b8920cd34e55a068942db61d3f2a : SUCCESS in 11m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2f5bf7f010234a7b8c71ffd69f140ae2 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0a3a621ca74b400e8ac704186a325381 : SUCCESS in 8m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5897f94f14a748b2ae7a7404d80cb80a : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a831359042dc4e88a9f6387d2ca0242f : SUCCESS in 46m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afb0f07621ea4dbd83a6d9a40584ec9f : SUCCESS in 59m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1436acabf5ca412a95a1adf72eef650e : SUCCESS in 41m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f0ce3ff5aaf641639591d9ba254477d0 : SUCCESS in 36m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a41d8a7fff0485b88ad96e2e6ae79f9 : SUCCESS in 45m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca431ca805b449fcaa97c912aa7d06fe : SUCCESS in 1h 56m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c8e5cdc96614aa8b6b9d6722e370b67 : SUCCESS in 1h 13m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1945d64ddf7e48589144ff149f012c2c : FAILURE in 1h 03m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/484afd6135a34657bda894d1fde5a108 : SUCCESS in 1h 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"787f1a5e2e51dd1905c38ec65b669a4029cb7330","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-20 21:39:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"10a54f036f223480a2d5493d325df8d48fafa7de","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-17 08:23:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6721a88211273157f74aabe5bd8b73747a2215d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 08:24:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3baa51a59e7a6c43434012ba8c0f556bf334d410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 11:21:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/506b5955b1dd4bf381abcba75a8530ec : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e2748b8506d4f8cae93c6109a634718 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb227cd8763144beb77db38d3470b2ca : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/227f90039f714aff82329efaac2c21fd : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07ebeb82d92487b8ce8aae52d2d0965 : SUCCESS in 12m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19bd5c8d809a45ecbd827ad05d48e7a8 : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/69aa2c51b33f4e83bd715305041f6625 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48c806b4ae334649956ec370a26ddad5 : SUCCESS in 56m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11aa8ee129544ed79d2c0fd9a2c4479d : SUCCESS in 41m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d933ab11b85341928fbb70dabd3f2147 : SUCCESS in 39m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee379674fe1b4e09a15c7943a2214892 : SUCCESS in 40m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc4587851b4e4cd894914ef4a7aa8010 : SUCCESS in 1h 56m 36s","accounts_in_message":[],"_revision_number":5},{"id":"45853ae3cc8b875630ffa4e3e6c2e5b3842b0770","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 11:21:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"8d38b1873d294ded0c324e5d8a6bf2353f357a36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 11:24:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72791901275493aa3261a53ebe72eca : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fc6e0b0a8ffc9281f1aeb676097d7713e023907e","revisions":{"0b957331122efecf728afc24ce12b908d5cbfcd3":{"kind":"REWORK","_number":1,"created":"2020-05-11 07:33:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/726741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/726741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/726741/1"}}},"commit":{"parents":[{"commit":"0e5372fb3a2e1a34e598da1991a41739547035f5","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e5372fb3a2e1a34e598da1991a41739547035f5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-03 12:14:53.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-11 07:33:01.000000000","tz":0},"subject":"Migrate grenade job to native Zuul v3","message":"Migrate grenade job to native Zuul v3\n\nThis patch also switches the job back to voting.\n\nChange-Id: Iedc1d5c5b603753b171fe17816b4d9c3aff1a16c\n(cherry picked from commit 8b8965bd7b89c78fe505ee2054ada5da30badcd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b957331122efecf728afc24ce12b908d5cbfcd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b957331122efecf728afc24ce12b908d5cbfcd3"}]},"branch":"refs/heads/stable/ussuri"},"e55bcb6b73cb043d6036961205d9e7b3cda7390f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-22 11:14:21.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/726741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/726741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/726741/2"}}},"commit":{"parents":[{"commit":"97975298f26e0ea9d841275686408984299350ce","subject":"Fix the grenade plugin to also upgrade octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97975298f26e0ea9d841275686408984299350ce"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-03 12:14:53.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-22 11:14:21.000000000","tz":0},"subject":"Migrate grenade job to native Zuul v3","message":"Migrate grenade job to native Zuul v3\n\nThis patch also switches the job back to voting.\n\nChange-Id: Iedc1d5c5b603753b171fe17816b4d9c3aff1a16c\n(cherry picked from commit 8b8965bd7b89c78fe505ee2054ada5da30badcd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e55bcb6b73cb043d6036961205d9e7b3cda7390f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e55bcb6b73cb043d6036961205d9e7b3cda7390f"}]},"branch":"refs/heads/stable/ussuri"},"c9e85e48323dfd252a927801fc318940171d5deb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-22 17:47:30.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/726741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/726741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/726741/3"}}},"commit":{"parents":[{"commit":"d926779724c7ce03dfbb3294d22ff4cd9d162724","subject":"Fix the grenade plugin to also upgrade octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d926779724c7ce03dfbb3294d22ff4cd9d162724"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-03 12:14:53.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-22 17:47:18.000000000","tz":120},"subject":"Migrate grenade job to native Zuul v3","message":"Migrate grenade job to native Zuul v3\n\nThis patch also switches the job back to voting.\n\nChange-Id: Iedc1d5c5b603753b171fe17816b4d9c3aff1a16c\n(cherry picked from commit 8b8965bd7b89c78fe505ee2054ada5da30badcd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e85e48323dfd252a927801fc318940171d5deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e85e48323dfd252a927801fc318940171d5deb"}]},"branch":"refs/heads/stable/ussuri"},"3a7f0a6e755fcb98e29c47712ad1f1f7daa88276":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-15 06:18:15.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/726741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/726741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/726741/4"}}},"commit":{"parents":[{"commit":"af902186c029f66540b5299621d6af74b64b5e19","subject":"Merge \"Fix netcat option in udp_check.sh for CentOS/RHEL\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af902186c029f66540b5299621d6af74b64b5e19"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-03 12:14:53.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-15 06:18:15.000000000","tz":0},"subject":"Migrate grenade job to native Zuul v3","message":"Migrate grenade job to native Zuul v3\n\nThis patch also switches the job back to voting.\n\nChange-Id: Iedc1d5c5b603753b171fe17816b4d9c3aff1a16c\n(cherry picked from commit 8b8965bd7b89c78fe505ee2054ada5da30badcd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a7f0a6e755fcb98e29c47712ad1f1f7daa88276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a7f0a6e755fcb98e29c47712ad1f1f7daa88276"}]},"branch":"refs/heads/stable/ussuri"},"fc6e0b0a8ffc9281f1aeb676097d7713e023907e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-07-16 10:14:13.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/726741/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/726741/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/726741/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/726741/5"}}},"commit":{"parents":[{"commit":"af902186c029f66540b5299621d6af74b64b5e19","subject":"Merge \"Fix netcat option in udp_check.sh for CentOS/RHEL\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af902186c029f66540b5299621d6af74b64b5e19"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-03 12:14:53.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-16 10:13:09.000000000","tz":120},"subject":"Migrate grenade job to native Zuul v3","message":"Migrate grenade job to native Zuul v3\n\nThis patch also switches the job back to voting.\n\nDepends-On: https://review.opendev.org/#/c/741421/\n\nChange-Id: Iedc1d5c5b603753b171fe17816b4d9c3aff1a16c\n(cherry picked from commit 8b8965bd7b89c78fe505ee2054ada5da30badcd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc6e0b0a8ffc9281f1aeb676097d7713e023907e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc6e0b0a8ffc9281f1aeb676097d7713e023907e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
