)]}'
{"id":"openstack%2Foctavia~727280","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I2a913b1ecea4247d2e33b8d68c7aba66c516fa06","project":"openstack/octavia","branch":"stable/ussuri","topic":"cap-flake8","hashtags":[],"change_id":"I2a913b1ecea4247d2e33b8d68c7aba66c516fa06","subject":"Cap flake8 to be \u003c 3.8.0","status":"ABANDONED","created":"2020-05-12 17:12:54.000000000","updated":"2020-05-15 15:00:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0925e318f67b06cb687bf1ba32de7c7f82cf732","_number":727280,"virtual_id_number":727280,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-12 21:01:37.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-12 22:41:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 22:41:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"706a83a08d3db7e1e39c57c174f668d7c74ce38f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 17:12:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c788811d19b5d937197bc8f830ca8a7934c9794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:32:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d578fe2f89604d9ab6ef7d655f52c313 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6a610ef00d549a68b30d90270ba9d3e : SUCCESS in 9m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25a60b083e2b4bcc92b7b591dc8ca868 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9a25b9991e4f7196f774db5fd3c0a0 : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2363aedd7f4e444d9967295c9b842714 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5681f01fa92427fabcf08b9fa8226f2 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adde7f5ca3dc41c5a126a3064d89ece9 : SUCCESS in 14m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4d393ca36604a1aa200984efb126c66 : SUCCESS in 6m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/543da2406fa24b8786b01b156644c471 : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c74893557c7441f280faef31fc32a3e7 : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ba05d0b90c34be38011468bd5c83719 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c7c0a27ef66422597230d797607227a : SUCCESS in 45m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30509b0014de411bba23660ed9029bad : SUCCESS in 1h 54m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1dde8172f12b4eb4a1f2146048681256 : SUCCESS in 59m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/39de7707e63f4349af80571aa3251aae : SUCCESS in 54m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b1489a262ac4ee388d85a3613ed4e76 : SUCCESS in 1h 00m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d106f2c6e7494492b08dc647841e66b7 : SUCCESS in 1h 42m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4139dec2d449462184194b70073df274 : SUCCESS in 1h 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b0e409ff97a24014b939a88ba174f6a7 : SUCCESS in 1h 41m 50s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"5f7a9a9f82595e7e66fa1335c14057946c093fd5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 20:21:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57d5c8a1f2b47db695d3602f432de988eb465988","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-12 21:01:37.000000000","message":"Patch Set 2:\n\nSigh, I didn\u0027t really want to backport all the style changes, not sure what else I need to cap to pass pep8 here.","accounts_in_message":[],"_revision_number":2},{"id":"b1b10bd54457f62b1175660ca6cd689bac1fb448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 22:41:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/479b115a318649279f4d12758411c633 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19a1648b5f2b434392735b9ec04d2750 : SUCCESS in 12m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0ead031fe764206abbb9cb2bfac9e28 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebd87baa1d4470fbda88a48e2812ca1 : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77094f6388424412b2e6aceaeffb6136 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12a2acc3b9b04e7c9904bc46ba371d06 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64463de33e094d938a7a14ea4bbadd22 : SUCCESS in 14m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a74512306b45a496d981792149549f : SUCCESS in 6m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/047102ae54b04767a202f5de2001b6d8 : SUCCESS in 7m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/861b490c997d40e5a4af3fe4bee8c87d : SUCCESS in 9m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/329cc9bdb85f4e3db30bab602418269c : SUCCESS in 9m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc938b724bc141f7a219a4520cf5a11f : SUCCESS in 49m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8622f58aff0d419c87cfe24723ea1977 : SUCCESS in 1h 58m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63c0489e1b6e440194f6ff0c1210aed8 : SUCCESS in 49m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e6f98d863e58415cb10f1a07ed7aebbf : SUCCESS in 54m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/deb7440513e946f8b467215c814a0978 : SUCCESS in 1h 01m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/839f160086dc480c93a6341d4ffaf7a9 : SUCCESS in 1h 48m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f97d55444e749c2bd0571853963c2c3 : SUCCESS in 2h 02m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c207abdbe2e241778d26627bbc79dfc0 : SUCCESS in 55m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84adb3cf807fb5888e47de2b65ff05c322ca7fbc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-15 15:00:58.000000000","message":"Abandoned\n\nThis won\u0027d work, need to cherry-pick https://review.opendev.org/#/c/726972/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a285e0f68122c160d3e712744626878ac7e9bcd1","revisions":{"4f3ec3e90be9403d4f85be6985f601e6d014fa73":{"kind":"REWORK","_number":1,"created":"2020-05-12 17:12:54.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/80/727280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/727280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/727280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/727280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/727280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/727280/1"}}},"commit":{"parents":[{"commit":"0e5372fb3a2e1a34e598da1991a41739547035f5","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e5372fb3a2e1a34e598da1991a41739547035f5"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 17:03:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 17:10:27.000000000","tz":-240},"subject":"Cap flake8 to be \u003c 3.8.0","message":"Cap flake8 to be \u003c 3.8.0\n\nThe 3.8.0 version introduces some checks that cause failures\nwith the current code. To avoid that, cap flake8 to a\nversion that had been tested without errors.\n\nChange-Id: I2a913b1ecea4247d2e33b8d68c7aba66c516fa06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f3ec3e90be9403d4f85be6985f601e6d014fa73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f3ec3e90be9403d4f85be6985f601e6d014fa73"}]},"branch":"refs/heads/stable/ussuri"},"a285e0f68122c160d3e712744626878ac7e9bcd1":{"kind":"REWORK","_number":2,"created":"2020-05-12 20:21:52.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/80/727280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/727280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/727280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/727280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/727280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/727280/2"}}},"commit":{"parents":[{"commit":"0e5372fb3a2e1a34e598da1991a41739547035f5","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e5372fb3a2e1a34e598da1991a41739547035f5"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 17:03:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-12 20:20:57.000000000","tz":-240},"subject":"Cap flake8 to be \u003c 3.8.0","message":"Cap flake8 to be \u003c 3.8.0\n\nThe 3.8.0 version introduces some checks that cause failures\nwith the current code. To avoid that, cap flake8 to a\nversion that had been tested without errors.\n\nAlso had to pin pycodestyle and astroid.\n\nChange-Id: I2a913b1ecea4247d2e33b8d68c7aba66c516fa06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a285e0f68122c160d3e712744626878ac7e9bcd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a285e0f68122c160d3e712744626878ac7e9bcd1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
