)]}'
{"id":"openstack%2Foctavia~727833","triplet_id":"openstack%2Foctavia~stable%2Fstein~I38990c686666c8501f78f003f1fa1e330cb3c33a","project":"openstack/octavia","branch":"stable/stein","topic":"pep8-test-stein","hashtags":[],"change_id":"I38990c686666c8501f78f003f1fa1e330cb3c33a","subject":"DNM - pep8 test for stable/stein","status":"ABANDONED","created":"2020-05-13 16:51:21.000000000","updated":"2020-07-02 17:49:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56f7add2fa064ca1fcf4b43e81fbad94b50a635a","_number":727833,"virtual_id_number":727833,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-13 19:02:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 19:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9563f7777969e72a4a7ccf6eb3d3edb80b05529f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-13 16:51:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2064228d901eb0edb32d75aaa63261cfeae02092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 19:02:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa21ea9cd25f4f3da577e965c1333514 : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53b20ed8b30842db976aad19661b419a : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85d300ff571f42208a242466483ef379 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49ae7b9cb34b42ff9550dbb8a724b484 : SUCCESS in 7m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9323e6445ed14c3dba3733165cc1de92 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5469eb83c1b044d7a712a19bbca95642 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373f7942872c4834be9e3135f97a791c : SUCCESS in 10m 02s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1dcc636467874ce3b534720a58ab50d6 : SUCCESS in 1h 33m 04s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/aeb7e67892184bbc98da8eaab8d608b8 : SUCCESS in 1h 31m 09s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f321ee1954174739976c47e49be53aa3 : SUCCESS in 57m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e35b7484a2ff4df6b54d99b79b2f6007 : SUCCESS in 57m 43s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f1e93ee88cc438f8efb06e3eed1a437 : SUCCESS in 8m 58s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b94ce685d6eb4485b1918da9f1d5d074 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97eb0d35525b453fac2d97d58491d417 : SUCCESS in 8m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f15e01ef3db432c94adb7fafff4ebe7 : SUCCESS in 1h 17m 20s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bc7917057cd04219aeb119c88ffe4391 : SUCCESS in 1h 07m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c712d4d16724516add0aaf71e8f861a : SUCCESS in 50m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a92718830b14410daa0892a1b4d2afb0 : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7cf6e40c366466587dddd22306805d7 : SUCCESS in 1h 52m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ada258573bf44e30b97687fb1b414831 : SUCCESS in 1h 54m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7a133c48af4b456897c35f6212a3b428 : POST_FAILURE in 20m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/341e170db9034bdc90baa0da35bf2d7d : FAILURE in 31m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c820765a3f6b4356be32e04943ddaa7b : FAILURE in 2h 06m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d682dd6e57443aeb07837347df54a9d : SUCCESS in 1h 48m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dca7bde5e8e5446381e08eb0eb3d711c : FAILURE in 44m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f65ff642e30a456f8e3112decd08eb0a : FAILURE in 1h 07m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbad4877872ebea775d4a3653516e7e58e44e353","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-02 17:49:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48c05c6aa85073650a74d5cbfb71d940b7de8acf","revisions":{"48c05c6aa85073650a74d5cbfb71d940b7de8acf":{"kind":"REWORK","_number":1,"created":"2020-05-13 16:51:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/727833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/727833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/727833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/727833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/727833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/727833/1"}}},"commit":{"parents":[{"commit":"922c4987fb4db019212c7db322a65c3998f0c1a9","subject":"Remove xenial based jobs from stein gates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/922c4987fb4db019212c7db322a65c3998f0c1a9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-13 16:50:56.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-13 16:50:56.000000000","tz":-240},"subject":"DNM - pep8 test for stable/stein","message":"DNM - pep8 test for stable/stein\n\nChange-Id: I38990c686666c8501f78f003f1fa1e330cb3c33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48c05c6aa85073650a74d5cbfb71d940b7de8acf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48c05c6aa85073650a74d5cbfb71d940b7de8acf"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
