)]}'
{"id":"openstack%2Foctavia~727871","triplet_id":"openstack%2Foctavia~master~I6c916c794526feac2d125045de4c80318d1dea29","project":"openstack/octavia","branch":"master","topic":"remove-provider-base","hashtags":[],"change_id":"I6c916c794526feac2d125045de4c80318d1dea29","subject":"Remove deprecated DriverLibrary class","status":"ABANDONED","created":"2020-05-13 20:43:02.000000000","updated":"2020-05-20 22:08:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d733f06ef112d940476bcdbe822eef48a6e7349","_number":727871,"virtual_id_number":727871,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-13 23:13:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 23:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55980df2475e577a52b2a4ea79675cf38d1a19e5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-13 20:43:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ed2a17326441ae0e4fd39f8136fcc0d3ae97b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 23:13:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f61bc237571461483a3c73ceb40c616 : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdf0adaab9a048778fdccddc903aa8c2 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ffc40e7362142d1857d1dafc6d21192 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f9fc767b224162b72771f286aea6c9 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de7043afca764394b2161a148d36dd83 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4cad3fa89045b1a913930ea5cd36a4 : SUCCESS in 11m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/759db1522c3b4c0ba866a2becf51d287 : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c2d46b0e664f4a038107903ce6a3e510 : SUCCESS in 10m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82011d0920d64197bb7ea3138b7c7b5a : SUCCESS in 8m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c5ab4a9b17cc46b2bc87ebef2447ff0c : SUCCESS in 54m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33bc0d6e5cdd499aa7f1dab40f1015e0 : SUCCESS in 1h 13m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/778054248b3343b49976ab8b50742b89 : SUCCESS in 1h 01m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e584288a217b446ca6c928802373dfd0 : SUCCESS in 59m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbf95f0a6d6c4aea8f0ba75629ca0d0f : SUCCESS in 59m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b085c0a1ac5540f09f297694ad445af5 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/471bf03dbbf44e3d95aa1c71e7d3c5a3 : SUCCESS in 1h 55m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36a9eab008d64c4a8b501891ee5b2cbd : SUCCESS in 2h 29m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9e415234b9c44c82bd69d94ee266d527 : SUCCESS in 1h 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c67dd875d9fa3acdbcd85400a1599f1c9c9bf67","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-20 22:08:43.000000000","message":"Abandoned\n\nSee https://review.opendev.org/#/c/701473/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dadc9fc021f7a7a0db13c88894de083d169aaa3b","revisions":{"dadc9fc021f7a7a0db13c88894de083d169aaa3b":{"kind":"REWORK","_number":1,"created":"2020-05-13 20:43:02.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/71/727871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/727871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/727871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/727871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/727871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/727871/1"}}},"commit":{"parents":[{"commit":"0153b5e0707497e4fed3cb2191d89631ec6e25a0","subject":"Remove deprecated data model classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0153b5e0707497e4fed3cb2191d89631ec6e25a0"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-13 20:31:58.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-13 20:39:29.000000000","tz":-240},"subject":"Remove deprecated DriverLibrary class","message":"Remove deprecated DriverLibrary class\n\nThe DriverLibary class in octavia.api.drivers.driver_lib\nwas moved to octavia_lib.api.drivers.driver_lib in Stein\nand marked as deprecated in Ussuri, let\u0027s remove it.\n\nChange-Id: I6c916c794526feac2d125045de4c80318d1dea29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dadc9fc021f7a7a0db13c88894de083d169aaa3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dadc9fc021f7a7a0db13c88894de083d169aaa3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
