)]}'
{"id":"openstack%2Foctavia~728025","triplet_id":"openstack%2Foctavia~master~Id52745f4751a62c53afbde94038b18b614f9b2dd","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Id52745f4751a62c53afbde94038b18b614f9b2dd","subject":"Remove translation sections from setup.cfg","status":"MERGED","created":"2020-05-14 10:55:50.000000000","updated":"2020-07-08 10:26:55.000000000","submitted":"2020-07-08 10:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728025-1594203733983-86bf99d6","meta_rev_id":"a6dca77426c17abbdbedf1f53302cc14367808cb","_number":728025,"virtual_id_number":728025,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-06-25 16:38:49.000000000","_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 10:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-07-08 08:24:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2020-05-15 18:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-07-07 17:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-29 14:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-07-08 08:24:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2020-07-08 08:24:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 18:16:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-29 14:25:12.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-07-07 17:08:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-08 08:24:11.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-08 10:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07945a7d3576fb838c995c55507312313340a474","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-05-14 10:55:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a696fbef527f790680e8ea54dde564301bd216e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 12:57:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3950049b83d54319965321dac7cc7acc : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86366345297542679f483cd6be1d32b2 : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36ac4a10fec643ad9a4b77caa6ba0b34 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd4595ef6244c5b92c5c4b8ba41cebf : FAILURE in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50e4cb138f2946ada5f635169683d909 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5ace26374074abda0ad1fc7c7239487 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e1a77a38684eedb85d2a67825c9d56 : FAILURE in 11m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/01fb862d677b45b6a12de93160040843 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8bdcdd2bc3ec47a0b9e45ce7abfb6f69 : SUCCESS in 9m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23a39299600f463c98ba13c0dd16c45d : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc35adeab43647a2bbc927dc34366f89 : SUCCESS in 44m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21d02d50b225421fb7d20696898a57c1 : SUCCESS in 1h 50m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/273c3f2e378748d68d102b3009126464 : SUCCESS in 1h 02m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1c71696e66f6421897fdb5bea154b664 : SUCCESS in 56m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fd266f05bde473c8db407c70cc7d4f3 : SUCCESS in 1h 02m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8e059a084f0d423db9242434075c6b70 : SUCCESS in 1h 46m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3829213ebb704899a3fec780bf8f13f5 : SUCCESS in 1h 46m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0bb128f1757244d583eb42e217d7e3f9 : SUCCESS in 1h 53m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fd885643ebfd438d8bc041097b0caf11 : SUCCESS in 1h 54m 50s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_amphora.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_quotas.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py","accounts_in_message":[],"_revision_number":1},{"id":"80e712d8a0d5ea884638c97c1a55532c31930195","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-15 18:16:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"12a812ddcebfd05c221e42b8206fc9ee12b09bee","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-15 18:16:59.000000000","message":"Patch Set 2: Code-Review+1\n\nJust re-based on-top of pep8/doc fix.","accounts_in_message":[],"_revision_number":2},{"id":"b3cb154ae94b27a28a28097800b9393a640987af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 20:35:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b899cb50ef9d44ef9a2d8a9d38ecee56 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9954a2d79eec460ba722451a56984369 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c02abcc655946eea8c3bc601f1f72b6 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cedfd3be73e4cc9915b6cd6ba5e6e9e : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3b942e8db34798bea31ff1cf48fabf : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a205e6b211042558299fd8f132caed7 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2640cfaa7c4860b90e4d0a3231cf71 : SUCCESS in 12m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/549e3efb75ef4b368951a00e53ecb5c2 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ee820081f82b4a3497b423ea387dc182 : SUCCESS in 10m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bce1dd2f26614bcf8bdaef5565f67213 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ce1360288c2463b9be5cbc27d861b0f : SUCCESS in 47m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b13c09140d2d40d09e000ca31318a777 : SUCCESS in 1h 42m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57034d4f5f1242d48af4c20339348b4c : SUCCESS in 56m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5bf8566ae81b4da4a4b9c45df4a51a79 : SUCCESS in 56m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2436c574d594eb5ac09b196352b1dac : SUCCESS in 1h 07m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f16931b8213a4139902e501c12dd9d80 : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f7b74b591a2f4aab8dd95ca302129239 : SUCCESS in 1h 35m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/adaff1a8abf1471988e1bdf6b9521ff9 : SUCCESS in 1h 55m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3115a52d6bb24c879549ed0476f547b3 : SUCCESS in 1h 03m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"092ffe4d6c82ae3e2facd491127980137e0c21b9","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-06-25 16:38:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"27807a891f30f4b87bcd346001d6089cfc3f822f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 19:04:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a096ad3c47344996b59ae27bfb672d12 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34e9d0002c5546d3a79f25feff46c28d : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2361c4aa44949ac85f14f8f1b8d56c7 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f9b8334b4b482db1c5dac3ce532a77 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ccde121565748a0b753632d164e8641 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4cb051a98414554a0de6178d8ed5665 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4119760aa7ef431da53ae3452a7845fc : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e3bb6771fe004eb59915aec276c714ea : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d5e193769a864883a90eee23ca7ef6bd : SUCCESS in 12m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff4a4fa5303d4ecc90d3c02a102758b8 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8306f20f239a43cb9cc259cd5f2ec654 : SUCCESS in 50m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee4423a098664beaaee0e8817ef1f447 : SUCCESS in 1h 55m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b6358225b2b41f884da1ec1767ff547 : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0a5f30b835ce48d3b1a720eb6d888c8c : SUCCESS in 55m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83b0b680a3ed415e8c0d36e277ab91ac : SUCCESS in 1h 12m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4caeb6a78f864faa91b908af3cc2c4cb : SUCCESS in 1h 34m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/62f223d50a6a4442a5e7bc56328e2e28 : SUCCESS in 49m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d93a5758e2942629c5ad892cbb68a8d : SUCCESS in 1h 45m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/143561a8e11244deba0fab8c296b048a : SUCCESS in 1h 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f5cac50264040e2a72969b35d5d66f84adda3c7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-06-29 14:25:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2481d4576fbb2baf0abecf71abd9d6b6b5150b06","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-07 17:08:01.000000000","message":"Patch Set 2: Code-Review+2\n\nYep, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"9ddde2ae38f6307b7ed46a97366166040772d84f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-08 08:24:11.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd51566c5e81cd26f260ef77ad39e698cb657377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 08:24:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b1b1f63cba15ea886d4e0325a4013e6234eb8c1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 10:22:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1bc7fe2eb9f9ca62efc450d505398f6df3bccf81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 10:22:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/87e141e520da461f9190b682cce1dd89 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05ed126874154a2d9651343348bc36e6 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e238973d774faf84c873d763d82bcb : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc4b4cc9cb1a4a21aad274026e53e6e8 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73d84642190a46faa61f431b40ff4e3f : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337c954a470243d28d260fb32c908e83 : SUCCESS in 11m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5cf34a9892c54aac882c6644bedce26c : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ca17a30092b4cceaaf10875df439045 : SUCCESS in 41m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6f32ef237bc4586ae34fb0df8612c98 : SUCCESS in 45m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f826f7d5c02f437d8c1c3f63e2d5d884 : SUCCESS in 53m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5771d31f6a894aeb83f3dce6a3050880 : SUCCESS in 1h 00m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4ee7957514b4f0488095420c586ce21 : SUCCESS in 1h 17m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8506fbf0eb4e48b38d3f259ddc9afb88 : SUCCESS in 1h 50m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a6dca77426c17abbdbedf1f53302cc14367808cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 10:26:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944aa2e6cc414363afc4ef7b23e1847c : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f610b953463a3d186ac1e86edd56e93b1115c98","revisions":{"d11aa9803c5ff6a59ecb014a47a9efa9edc6d481":{"kind":"REWORK","_number":1,"created":"2020-05-14 10:55:50.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/25/728025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/728025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/728025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/728025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/728025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/728025/1"}}},"commit":{"parents":[{"commit":"5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd","subject":"Merge \"Migrate grenade job to native Zuul v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 10:53:59.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 10:53:59.000000000","tz":480},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore, Babel can\ngenerate translation files without them.\n\nChange-Id: Id52745f4751a62c53afbde94038b18b614f9b2dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d11aa9803c5ff6a59ecb014a47a9efa9edc6d481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d11aa9803c5ff6a59ecb014a47a9efa9edc6d481"}]},"branch":"refs/heads/master"},"2f610b953463a3d186ac1e86edd56e93b1115c98":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-15 18:16:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/25/728025/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/728025/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/728025/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/728025/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/728025/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/728025/2"}}},"commit":{"parents":[{"commit":"9a1d6d3585070da570bcf6313e237b486bdac81b","subject":"Fix E741 pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a1d6d3585070da570bcf6313e237b486bdac81b"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 10:53:59.000000000","tz":480},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-05-15 18:16:25.000000000","tz":0},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore, Babel can\ngenerate translation files without them.\n\nChange-Id: Id52745f4751a62c53afbde94038b18b614f9b2dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f610b953463a3d186ac1e86edd56e93b1115c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f610b953463a3d186ac1e86edd56e93b1115c98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
