)]}'
{"id":"openstack%2Foctavia~729417","triplet_id":"openstack%2Foctavia~master~I3511bc2cf1714be89322fa483640ef607a9bf435","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I3511bc2cf1714be89322fa483640ef607a9bf435","subject":"DNM: Testing the impact of mitigations on the jobs","status":"ABANDONED","created":"2020-05-19 22:57:39.000000000","updated":"2020-08-18 18:45:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0842d50a5aa654783db206c372d9ef4d328370d5","_number":729417,"virtual_id_number":729417,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-20 02:09:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 02:09:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e01bfc56ae41fbd62f5c1fb7fc944d763b07f15","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-19 22:57:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d94ac5dc20bc0fbfdf9f1dea35ef772a600b61b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 02:09:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed2faad8d5724e2abdd63677d5362c5d : SUCCESS in 12m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c093e3c17304ad4a3d9865b18163bb0 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a34d61c5ecf42deb3c2f3637bdfd0bb : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b960a686a3134223a04418ba686dceff : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2192ce6e59f4a1f80bb2c14537fb243 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3226a10a20a04046a3781c413cd31482 : SUCCESS in 13m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b4cdcd8658224187b0260e7bea61e5ac : SUCCESS in 9m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/44665f9e63d747e8a8e40a2fd6ec4143 : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6c161291ae34f4c882d3a84b19e59e6 : SUCCESS in 8m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0f20eaa6fd745d59838e8b4f42f6970 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/adf40bc824374e1e80463d6303ef2537 : SUCCESS in 1h 34m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f20da4d11d03459a99e089d1389e195f : SUCCESS in 57m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f9c2c18dbbd54c999f70a6044e2cd233 : SUCCESS in 54m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/390cb9f0d3bb4283800d0407b8f5ea9f : SUCCESS in 1h 01m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bcdffb76385e41359d0052df2916fa41 : TIMED_OUT in 3h 03m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/04e3195323b846b4b4acc434dde3a55c : SUCCESS in 1h 25m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/449de49084894559ba7203884cce145c : SUCCESS in 1h 41m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ec37d6500854a989d8c22962b51ca48 : SUCCESS in 1h 59m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0842d50a5aa654783db206c372d9ef4d328370d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 18:45:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"811b8f800da534d9f623517471333bfa810afa34","revisions":{"811b8f800da534d9f623517471333bfa810afa34":{"kind":"REWORK","_number":1,"created":"2020-05-19 22:57:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/729417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/729417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/729417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/729417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/729417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/729417/1"}}},"commit":{"parents":[{"commit":"6b2b3694c3ebef811db8e96e2917b5bb94ae73a4","subject":"DNM: Checking on what CPUs we are ending up with","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b2b3694c3ebef811db8e96e2917b5bb94ae73a4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-19 22:52:39.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-19 22:52:39.000000000","tz":-420},"subject":"DNM: Testing the impact of mitigations on the jobs","message":"DNM: Testing the impact of mitigations on the jobs\n\nThis patch adds kernel parameters to disable the CPU mitigations\nto get an idea of the impact they are having on our test runs.\n\nChange-Id: I3511bc2cf1714be89322fa483640ef607a9bf435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/811b8f800da534d9f623517471333bfa810afa34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/811b8f800da534d9f623517471333bfa810afa34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
