)]}'
{"id":"openstack%2Foctavia~730135","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I05a541a77f254a77ad5036e1062b61c8ce93b754","project":"openstack/octavia","branch":"stable/train","topic":"fix-trust-token-stable/train","hashtags":[],"change_id":"I05a541a77f254a77ad5036e1062b61c8ce93b754","subject":"Fix getting user neutron client","status":"MERGED","created":"2020-05-21 21:51:20.000000000","updated":"2020-05-26 17:25:43.000000000","submitted":"2020-05-26 17:15:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730135-1590513324564-41f9ad85","meta_rev_id":"4d5fb1af19cbed1512ffbcb6ae56b08ded203702","_number":730135,"virtual_id_number":730135,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-05-22 02:15:27.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 17:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-05-25 11:37:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2020-05-26 15:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-26 09:15:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":1,"date":"2020-05-26 15:02:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 11:37:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-05-26 09:15:49.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-05-26 15:02:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-26 17:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"090ddf9c5ddd0841c7223ba35f17d388980bb4fd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-21 21:51:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1b3395208037805e661558c3a6a0c12b50694025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 00:11:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e543403c32d84a6d87eff1c0fb1827bc : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afffbe7ba5734c3bb04508f26a35796e : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e044eafb8ef4353baecf2d751fa62ee : FAILURE in 21m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b90cdbcfa3b49cb9044bc612ad4e5f0 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01e2394718dd4ff29305989c38b3df1f : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe4d7e408cca42d38ceb41cf77942506 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee811cb025d40f28e4106c888dade5b : SUCCESS in 11m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f99e62cd4708451db2441ccd922405d7 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0f11c3fdce214f6088e8d53ce83a5d5c : SUCCESS in 8m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4325df44968f4d1b8bc46ad2ab94a088 : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d47fb79f3e274f7d817e29fa3558ac8e : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7191013d4e01466da2c780444d0ff05f : SUCCESS in 41m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/56937d4af33141e7bc9a9482f65c740b : SUCCESS in 46m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/226f49e0d5424f1e908526992aba4eb7 : SUCCESS in 2h 03m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8ab9fe856d674172b30663c937f6bc1a : SUCCESS in 1h 51m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b5af2d36260e4eda829d9039b91bd6a4 : SUCCESS in 2h 19m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53c41509366f4defa156a867f56ee46e : SUCCESS in 1h 06m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c469f8a6c3c4a8f923ee9c97493c2a6 : SUCCESS in 1h 52m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/abc9ba8b97ec464c93b5fd08a04224d9 : SUCCESS in 1h 18m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c5575530cfe41e7b49da84ca79a898f : SUCCESS in 50m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5e9dcad040f641038fa4ebbf7744508b : SUCCESS in 58m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e1246fa44444414a01dcb038c21e6d0 : SUCCESS in 1h 04m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e3cc5f9ad5914bc38e38b4e9562d3032 : SUCCESS in 1h 24m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3459694a9bcd4053b1288e829017059a : FAILURE in 1h 04m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c50b8c232a73cf64967d89568c2891d19cdcf117","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-22 02:15:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be1a8a1f17e2ebb27f3e7368a0ce980b6f572655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 05:12:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d97ea67809143ecb094e11e64fc160c : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a2398a784ac4fa2b362c48e447826a6 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe9c0fdef554836bffadd7a77bc2155 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fabd37ea22bc411c8c7e8a5e4a36c435 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac9373fca78e4f758076a0126524f0eb : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2180fda9e0d740658ad69b3bbc7528d4 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c912221d63d648a6918a8de728b570f9 : SUCCESS in 10m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/41fa7d7d480a48bebdaea7aa709b6b2f : SUCCESS in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/215b2da0f1fd4b469abc965338048a8f : SUCCESS in 7m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e98ebe5597fb421d8a9ee6a1c285c026 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37f10d1c9e3e47a0a6d71d99a6d133af : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dec535dd270a4a4da7fa93dd864484ef : SUCCESS in 53m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d85a65d48220498dae173cacd138eb1a : SUCCESS in 49m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c2824a21dd94f2e94cd1782dcfc3aa9 : SUCCESS in 1h 51m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d7f967ce6c3a4824b9d04f5e92877d9b : SUCCESS in 1h 40m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1f9c78d2585d45d195fbbb81de5777f3 : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e35b1706fd94b8ca795a734e94c52e1 : SUCCESS in 1h 08m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/752e99a9bc9b45bc8b05b5faff83c564 : SUCCESS in 1h 58m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/194b6bb92f174df0b44137fe0017b12c : SUCCESS in 1h 18m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4cb0036e22bc447ab07d54e71dde9034 : FAILURE in 51m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/91210a3f9931431a9a0914fd39de41e2 : FAILURE in 1h 07m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75a0b77614af43f3b8bc85c6558dd3a2 : SUCCESS in 1h 01m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0419c68fb2fe4c9f95b4b4b81b16bc29 : SUCCESS in 1h 08m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ca6d75c40ca846cc8e60c6d52c1430d7 : SUCCESS in 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9f837cf0878f1267f247802cb74e5de0f0d5349","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-25 11:37:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"269c5df22983ffa38aff4f733ce37c8d0a13d122","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-05-26 09:15:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d2d344c8a4cb368824a5d76c5663c9b310c054f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-26 15:02:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"9091b7e443e137afe186cddbdf01a30576195fd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 15:02:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"299fefcb2b9d7e1e8f8574dfb5c26b4e24e47082","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:15:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d4123d792411c3334e1af911859ac896f7f375e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:15:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84ab63cb43e24bbaa956f565356e79bf : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf539a690db414387aac95b4c14c6ab : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/650bd88fc91d46f3b383a6d80cfc209e : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dca699def44c431496cc7319b5629cee : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55d9b404b07b42c9bfd8d6b3f782e25a : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6692e7589748a7b5b0b9009c61f3e6 : SUCCESS in 12m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fa022cf048b243bb9fb9841aec4ad647 : SUCCESS in 10m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3e41dffe68644a6979604513e926935 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/363eb74aa2d941dabfb12e99f25e33ea : SUCCESS in 53m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d38740f74c644728ad358964b38388c4 : SUCCESS in 47m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d910af2907044cf48d537db1e16e6a60 : SUCCESS in 1h 39m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/225cec6253494d73bfb8207ab0207521 : SUCCESS in 2h 09m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c5315f094a3c4691ac93519532102c3b : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a48bf222dde14f100880b186187caaae4f962a7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:25:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218ffd4be1f9438da4590c2dc8821f28 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09e863eb539e29c8950ddfce12d602c0e77c1b7e","revisions":{"09e863eb539e29c8950ddfce12d602c0e77c1b7e":{"kind":"REWORK","_number":1,"created":"2020-05-21 21:51:20.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/35/730135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/730135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/730135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/730135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/730135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/730135/1"}}},"commit":{"parents":[{"commit":"dcd80d3028e989fe5f2147b3ea8f074765eb953d","subject":"Merge \"Fix padding logic for UDP health daemon\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcd80d3028e989fe5f2147b3ea8f074765eb953d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 04:31:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-21 21:51:20.000000000","tz":0},"subject":"Fix getting user neutron client","message":"Fix getting user neutron client\n\nUse token and endpoint URL to initialize neutron client for the\nrequest user.\n\nStory: 2007619\nTask:  39641\n\nChange-Id: I05a541a77f254a77ad5036e1062b61c8ce93b754\n(cherry picked from commit b54f373cac62ec170562d7a2d84e66773b9a2deb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09e863eb539e29c8950ddfce12d602c0e77c1b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09e863eb539e29c8950ddfce12d602c0e77c1b7e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
