)]}'
{"id":"openstack%2Foctavia~730136","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I05a541a77f254a77ad5036e1062b61c8ce93b754","project":"openstack/octavia","branch":"stable/ussuri","topic":"fix-trust-token-stable/ussuri","hashtags":[],"change_id":"I05a541a77f254a77ad5036e1062b61c8ce93b754","subject":"Fix getting user neutron client","status":"MERGED","created":"2020-05-21 21:51:40.000000000","updated":"2020-05-26 16:49:30.000000000","submitted":"2020-05-26 16:46:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730136-1590511588840-7901bf70","meta_rev_id":"e0fa27c51c2cc4d95866cc355380a3ac5160ea0a","_number":730136,"virtual_id_number":730136,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 16:46:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-05-25 11:37:38.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-05-26 15:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-05-26 15:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 11:37:38.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-05-26 15:01:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-26 16:46:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6151370be21d3fe5aafbc7f1684c6517e3c5a511","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-21 21:51:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d69c2cacf9c1eb1e964f95b43dbf32a5aaf03d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 00:04:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8377bf5f383a4cc4b1a06add9b051700 : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f24a796b908b4ade90f588c40cd687ce : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/276648e84e0c4fc689a5be11fa46eccd : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0e1554fa51c4fae9009f71e8952c8b4 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/243dfd5410104344bbf622c2bab61425 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e8fba364d64e04adb995a73129afaf : SUCCESS in 11m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aea4e5b2830f4ab0b658273e53b09846 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ca0197ba70e84606b412d6759a141c64 : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70a79b6cf456445c81db3397635ca071 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf65a878bb074e5c9e8306e8108a4d6f : SUCCESS in 40m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b1e2044526b474e960c9b5791320950 : SUCCESS in 1h 49m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d95b038391340c8803a384a6e6f14d0 : SUCCESS in 57m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f12f4c65269f492d82a745d62f1f288e : SUCCESS in 53m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d24af8e0b0774860b9bd2694a4661ddd : SUCCESS in 1h 03m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/67f1c74c9d8a498194896b86b0374f3d : SUCCESS in 2h 07m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/469e6b9c5cd945b1b0de07329ae16623 : SUCCESS in 1h 09m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44696a25184549eb9f48db5ef2c6b310 : FAILURE in 1h 03m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c9c39506f25241b690d7247fea2dca5e : SUCCESS in 1h 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c54f0bde453341ae7b27333f37f20f4b162a2dd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-25 11:37:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d360db6271a882c06da3c322de5d030de35b5c2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-26 15:01:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"33acee710319f6661e39274cd2a3e99ab59e3a23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 15:02:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23dd04f638a81cfe5d059d5037035954b3709410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:46:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"75f1603a4a4ab8f3cdd9c13a200789ef68d8d60a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:46:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ccbeb63f4254e62902e2ccafc42eb4d : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdd6637467104adaab402701f5f461b0 : SUCCESS in 10m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6519e2cb2639490cb9530c8f44b14a49 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf807514203f4e4fb572d13411c8c679 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2555bcc89f641e8af89402db6f0da55 : SUCCESS in 12m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa0e4a7de79d42ad9f622c5db1b8fffe : SUCCESS in 12m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/466b4357756644a59afb04641c6dceab : SUCCESS in 53m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4d78fee86de49ef9a273998574bd1c3 : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3e1e1164c7bc4dad8b635ca588d1aa59 : SUCCESS in 59m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e63235405d9546169fd2bea4511b7491 : SUCCESS in 55m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65ef135d64064c28826cab4710b506ac : SUCCESS in 1h 37m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6c20abd2392f335129bc04cd5dc4f207d160af66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:49:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6bd1204b244591ba73fbbe19156f5d : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90defc9bf91baae3b1b5ad65d211c2c09ea6a3db","revisions":{"90defc9bf91baae3b1b5ad65d211c2c09ea6a3db":{"kind":"REWORK","_number":1,"created":"2020-05-21 21:51:40.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/36/730136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/730136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/730136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/730136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/730136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/730136/1"}}},"commit":{"parents":[{"commit":"0636cd1b43e5a2902ee51a7a8553061fcec9ae73","subject":"Remove deprecated exception classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0636cd1b43e5a2902ee51a7a8553061fcec9ae73"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-07 04:31:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-21 21:51:40.000000000","tz":0},"subject":"Fix getting user neutron client","message":"Fix getting user neutron client\n\nUse token and endpoint URL to initialize neutron client for the\nrequest user.\n\nStory: 2007619\nTask:  39641\n\nChange-Id: I05a541a77f254a77ad5036e1062b61c8ce93b754\n(cherry picked from commit b54f373cac62ec170562d7a2d84e66773b9a2deb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90defc9bf91baae3b1b5ad65d211c2c09ea6a3db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90defc9bf91baae3b1b5ad65d211c2c09ea6a3db"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
