)]}'
{"id":"openstack%2Foctavia~732078","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I6543fe5caf539c9b1167a7c58984b3474879e1de","project":"openstack/octavia","branch":"stable/ussuri","topic":"story/2007734-stable/ussuri","hashtags":[],"change_id":"I6543fe5caf539c9b1167a7c58984b3474879e1de","subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","status":"MERGED","created":"2020-06-01 01:28:17.000000000","updated":"2020-06-02 09:56:54.000000000","submitted":"2020-06-02 09:54:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732078-1591091697399-e3312ef1","meta_rev_id":"2c3df627635a81d41444167ab5bb398fc51ef0d3","_number":732078,"virtual_id_number":732078,"owner":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 09:54:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2020-06-02 08:46:01.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-06-01 15:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-06-01 20:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2020-06-02 08:46:01.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 15:19:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-01 20:42:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-02 08:46:01.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-02 09:54:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc7a8b0c1bd91f7f99d29a4ff5ea80cadaf37c51","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-01 01:28:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"187b88dfbaa096bccbe37522ccfc527115d9ff6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 03:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a339268bb684402b87e2af143513f5b : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34cd3f5647d0404695834e315b1d620f : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3e081cb25041d68dff8efe4e64f3ec : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5e31a8bd95c40ec98f2fc0fbd2a81c2 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f54ad0e352cb4baab8290efe419312b4 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6113ddbccd364f0d914a210943826076 : SUCCESS in 12m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/18bdf92a597e4f9a9c0eb13c94e1402a : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b606b06ed9af4d86a6fb2cc51138035c : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c92230ff8a54e54a808f343d3b77ca3 : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7867e482336e4572be9f0eafcdbce4a4 : SUCCESS in 43m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86ab8affd6524597abd9338e80511b37 : SUCCESS in 1h 47m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aed6fe5ce9424af995fe001a33e22680 : SUCCESS in 55m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e10e99857e9e4d2c8d2ea7ac88b27175 : SUCCESS in 53m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af361b0e99244b90a57db3cd11840a54 : SUCCESS in 1h 03m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aa7334f673a64eb89cb2271acf7e7a19 : SUCCESS in 1h 47m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1b720be78b146f7845000099eaae727 : SUCCESS in 1h 59m 19s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9621e17681bd4322ab980059164d3bbc : FAILURE in 1h 02m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/79b350f4ed3c4d5699a00c9db514f67d : SUCCESS in 1h 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a9a65e07fc6f78b69abe9d4172a53b0fe7259ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-01 15:19:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a82cb5987c037e66d97e6e029b8a5dc0196abe4a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 20:42:17.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"bdb335b21f70b3ba6155f73c00acc7982250be86","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-06-02 08:46:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"772ecc258d8f6bdb8a67f4553ffb03406b2254fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 08:46:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"713b48f9aa2de67ad2bab1596683b8ebb1aa2790","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 09:54:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"beae977bd740bf21c4525a2c1d893a06a1841655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 09:54:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ffdb9ed5c544df6bbe4964395488520 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40cfdc47f2c4bfc8b9ee22d88dccbe7 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00288845174e4399b702bb0bdd2446fd : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85334e5b2d1949a6a0eb16f62120d37c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c65a3b4782406483c23d441b82ff65 : SUCCESS in 12m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c82bcf060d04163b8075b9a1de2efc8 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d628b9511e44cfc9ec9fc72687a330d : SUCCESS in 44m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c126e1f0293490c92e1c8ff0ec18e5a : SUCCESS in 1h 01m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c083cccae0204729babd17ab7ba92ac9 : SUCCESS in 56m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/93095205b9394c599f659bb2c610e4d2 : SUCCESS in 43m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f72003486ba848548030c6065fc00247 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ab3a4c02efa5ee6bb4fbb29488b6cbd7142ff146","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 09:56:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba6c0e9d4214f5c91b65ce530faef36 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44df1ea060c57e3bc44c7b21206d7aae67fd7ec4","revisions":{"44df1ea060c57e3bc44c7b21206d7aae67fd7ec4":{"kind":"REWORK","_number":1,"created":"2020-06-01 01:28:17.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/78/732078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/732078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/732078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/732078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/732078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/732078/1"}}},"commit":{"parents":[{"commit":"90defc9bf91baae3b1b5ad65d211c2c09ea6a3db","subject":"Fix getting user neutron client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90defc9bf91baae3b1b5ad65d211c2c09ea6a3db"}]}],"author":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2020-05-29 08:46:54.000000000","tz":480},"committer":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2020-06-01 01:28:17.000000000","tz":0},"subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","message":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027\n\nstory: 2007734\nChange-Id: I6543fe5caf539c9b1167a7c58984b3474879e1de\n(cherry picked from commit 7c96e455a9f7818cb2c5c16ed12120e719f8d992)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44df1ea060c57e3bc44c7b21206d7aae67fd7ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44df1ea060c57e3bc44c7b21206d7aae67fd7ec4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
