)]}'
{"id":"openstack%2Foctavia~732079","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I6543fe5caf539c9b1167a7c58984b3474879e1de","project":"openstack/octavia","branch":"stable/train","topic":"story/2007734-stable/train","hashtags":[],"change_id":"I6543fe5caf539c9b1167a7c58984b3474879e1de","subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","status":"MERGED","created":"2020-06-01 01:29:06.000000000","updated":"2020-06-11 03:00:44.000000000","submitted":"2020-06-11 02:58:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732079-1591844317825-59896c92","meta_rev_id":"6c4b6c84721f974b03329bd5b3e9d0fb9073ebb6","_number":732079,"virtual_id_number":732079,"owner":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 02:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2020-06-08 11:48:04.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"value":2,"date":"2020-06-08 11:48:04.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":1,"date":"2020-06-08 11:48:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-06-08 11:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"value":1,"date":"2020-06-08 14:44:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 15:20:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-01 20:42:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-02 08:46:17.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-05 09:36:44.000000000","updated_by":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"reviewer":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"state":"REVIEWER"},{"updated":"2020-06-08 14:44:00.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-11 02:58:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd92b3fd84d40ad7d616ee68a96fe31d02dc8e36","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-01 01:29:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b978bb4caf3738d34cc517a5d44b18c4df072c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 04:02:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78bf8ea26ab84c1ea39e754131fe3887 : SUCCESS in 8m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0902472f893d4fa99db827664b25afec : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c08ec4f79fe14764995657b5156a8801 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc14ccb96a40446da81d39767f8400c0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1984feb19bb34d3e85be911be8dd66ba : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12d988180a2e499bae1d76b30d412f93 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7532d44eac1e4336a567980c9e009c0d : SUCCESS in 10m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dee78f6c179c4255a098ae9d3c6acae4 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e0c371be36f74db6b7fd3f1cddc5df6e : SUCCESS in 6m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8e38ab88af764fb49c2447d2feff9f9a : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3ec8e97bf924f66baa7d495a1671508 : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d818f9a977984083b45f1a9a1835d1f6 : SUCCESS in 43m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f06af9bc85f24616a02be245dc678960 : SUCCESS in 46m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bd2fa5cf1504342be807461eb8d1f74 : SUCCESS in 2h 01m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/77a88c6bfc3d4eb2a114451225bb30de : SUCCESS in 1h 03m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/21a11a1651254310a4e79845efe50e45 : SUCCESS in 1h 54m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9bdfb7be2d264cfa8c72392ee8ec8884 : SUCCESS in 1h 54m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/902a49fdf3274eecbffe64b100fceb08 : TIMED_OUT in 2h 32m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6e673660997b4d1da6c32d764a45a823 : SUCCESS in 1h 56m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0afb593302e2438ea145f77e881449a5 : SUCCESS in 1h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b2baf30372f640e796e19014d9e67309 : SUCCESS in 1h 20m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df10401d45d045f89befc9fbfdf7c204 : SUCCESS in 1h 04m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f17529eb68d94e2aaee5f3a393f98ff0 : SUCCESS in 1h 03m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/85c44fa2d08c4e0aae8b84e9903a6649 : FAILURE in 1h 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c51161b1df43d64b583a056ace1ea929e0354052","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-01 15:20:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ba29786460d0cde957a5fe4a0c2abde5838828bf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 20:42:39.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c0aab5b363e44fac7822aa016e199dcef567d62c","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-06-02 08:46:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f1de961bd2f5f39342069f5983b4499924052cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 08:46:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d40ec8bc9a60bafd8f05de84b132a8f4bc2f669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 09:54:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52ac6b4c551b4a30b2a2b4d98cb78607 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbd3ffb31f94a049a2c1bb1af0da8db : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7aeeb77447b34480a375751aaca2f36c : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b2f9807b60c44649548fb55834f4047 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfb4a91427c541219b217e43107a1460 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db5dfea261341cb979cd1134c3d2798 : SUCCESS in 11m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4a239e1e38704aeeb5e3058bde16b769 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a436e1c8d0a46ee90f53706b4d764f5 : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8fe5e9532e3d4514be0b4d2e2da17f57 : CANCELED in 25m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/94bd58e64fc64fcc813839e888e7d576 : CANCELED in 25m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08258199f80a447ab3fc64a153cf7a8b : CANCELED in 25m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/87ece0078c264895838a54e305dafada : FAILURE in 25m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78066e3064b54d18a4cb6ce68c4aae80 : CANCELED in 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c78f71e41f08fce12cb3330780088d5efdd77844","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-03 02:56:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99ade95a4d0bfec30a73b95f0b6bdf9c88f4eaec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 06:47:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55ccd77c3dc447a0899ac6da65cade0b : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc8a3d259d2a45fcb93323c17e639ae2 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1518e9426d345f7993a3d54b4aeeed3 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5134d6a3cd48464fa219bf322f6cc912 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac2439f4a18448d2acc8ccc011e823a0 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e045ad9db6a4f9c9dce1a0c3aaacce6 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65e0e0eb51e41e1adfb93479071351e : SUCCESS in 9m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/efd1f59ace284a5fbaa350ce9e77c09e : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/522def0df133460c9eccb14e2fd4e4aa : SUCCESS in 8m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/266bfda27d8e4ce2a1d8b59fe863d30c : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c06b66c273474b86a2ad3848acf8f0cd : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bba146f64ca442a386115d247f80bb6c : SUCCESS in 45m 37s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6e06f3c7690941458d9ae6948a9b2113 : SUCCESS in 52m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07709fed5fae4c78b8fc66e9839ef131 : SUCCESS in 2h 25m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dba0f5fef7644b1f92e220078d8f2968 : FAILURE in 31m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5adcf918d5084731a1569bed5c1aaa2b : FAILURE in 38m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e96bf02129e14574b8962177eed73755 : TIMED_OUT in 2h 35m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/57d6064f6b5f48a09065c2a85c1fe058 : FAILURE in 20m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4d8ca6853bc549cf8bc9d7e56a692957 : SUCCESS in 1h 21m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/edbae2c6d3f940ed86f565821032f99d : SUCCESS in 59m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a99024cf9f54ec78d961a18393cce14 : SUCCESS in 2h 02m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22cadefc90e34b4da71c0099d1980c31 : SUCCESS in 1h 10m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a2c4dd6fe2e14d3facdcf95910e6e0d2 : FAILURE in 31m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6aafd197c9d645ae887365ae3f5af0c5 : FAILURE in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e1218e53b028229b9f89a9f60981ee8da0522d4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-03 08:27:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27d2307c4f161ec17dd5c7e81d942919d8e5722c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 11:02:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8095069ea36b424a8f9ff4a971222ec2 : SUCCESS in 11m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e30cbfb681d84e5e9d06ec4472d578bd : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770d21116e9147d49c61fd6e426e66b2 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/701ac052d8f4447fa1ff27dba8ae3eb2 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e93b5578588459eb7e248cdb35b0b6c : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9f5d6ed8faa480d87a04a1fae7b9c0a : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889ca40faa7f4d5fb6fbcfba36013ae5 : SUCCESS in 16m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/27efcdf70e5946ac8ba2b206e8e73ae6 : SUCCESS in 6m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ee657397b73040de9a31882471a939a2 : SUCCESS in 8m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/849d1701989b471b8d9e11c3da79300a : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89ec44a1d251467eb20c348ab5244aa4 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/178ea026738249f88d0df5c8e7ac9eaa : SUCCESS in 42m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a4caa09c4fc14e628f64bba79b5221a1 : SUCCESS in 47m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/faacb4ebd2f845318f285c442dc9288a : SUCCESS in 1h 02m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5df292510125427e90cb0b4cc8be3383 : FAILURE in 49m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f7a4c06cfc704925b936605de6c9e581 : FAILURE in 52m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/168af1b892d0407fa9b501b5625d2c0c : FAILURE in 1h 45m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/77c73454c84c422989d3ca5da370905d : FAILURE in 29m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/591c3ba4bd7a4892909ae3e03f779fd2 : SUCCESS in 1h 51m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e094a73a19784be9a1737cce4bf6cbe4 : SUCCESS in 55m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8669279adf6d4386aa5ac63d76bf9592 : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b90ae27181af42799a4598feaa7eb5af : SUCCESS in 55m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ff262b89607c49cf90f4d9af0e78dfa3 : FAILURE in 22m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c7e42976d1a34d06963e4e1284a6276b : FAILURE in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca5779dcaa2a2a6223bb2a097a71191568de04ef","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-04 04:38:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28e6520a70b06596c4ba5df95e9bae78da13dccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 06:45:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bed6231be95441f7a03a846fd866db74 : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9e2bb250c164b64bee752ad2a94d315 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7c0b2be7244bb285761c4b2ac148c9 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44fccc6d5c6147c7b59e0023dd1bc727 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed16ec2196bf441aa6d922d57b32503e : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c290c4456dcf4a68939edaac04f91d12 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260619b273f04fe4aa810d0d933b1aca : SUCCESS in 9m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1d82a64a69f34f10b3bae743d321bfef : SUCCESS in 5m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e13ff2cf35514714a3a134d29d266764 : SUCCESS in 10m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/653ad4f2f87842e1ad7af289ebdd47a2 : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cd58e28d828445fbb30739efb60686a : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb01b6ccf411425d9a54356ccb4a0868 : SUCCESS in 47m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7ff7b95aa14c4a1f9aa5c65599544aae : SUCCESS in 42m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d0fca454c424c6d8cb46fd3f2d1aa79 : SUCCESS in 1h 43m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3514eb49d8ae4bc9a5f0d182b0d18418 : FAILURE in 28m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d67463365d1d460bbed1d75be4e41bd5 : FAILURE in 30m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/792ab04224334c3a9293d350933d9bdd : SUCCESS in 1h 12m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ec68295e34146eba49c29558fc774f0 : FAILURE in 35m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/caaeb2014aaf4427a2af7d02fc8ccf96 : SUCCESS in 1h 32m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0c10e81b84d4874b4b5e576cf587797 : SUCCESS in 1h 01m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/459fe7bdb9264e3dbe0bcb9c40356ed7 : SUCCESS in 2h 01m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46f265fb9f584a29a3514db2e1f7ff52 : SUCCESS in 31m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6a3d52e0c4854877a3dfcfc14a266244 : FAILURE in 21m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a59abab552f54cedb29416e8efe5bcf2 : FAILURE in 46m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9acd07207ac7d8a49001de38e47290f8f304cd67","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-05 06:12:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0ec93e8a8863ba252289878047cb50fa5960f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 08:00:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4054f5e436ba4be991c60daa455fc031 : SUCCESS in 14m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68be035fb37247a88f8c34cdd6be3657 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b234f0c55ab4bd590b93ac6664522f4 : SUCCESS in 9m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4ccf5e534684782b563651afb31bb48 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff9a031c76047db9e62ac2b73e2fdbb : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22e0e4a18ac8495191f0bc6a4709a78c : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f085e2f6e647c3a13cea28b35a0eac : SUCCESS in 13m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/95c6c195eff245b1b76d77631cfeab6e : SUCCESS in 9m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1b97d11f9c4a4a53bedf0a00944f8bbb : SUCCESS in 11m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/acf38144d9cd4a978ed02c0cd264eae4 : SUCCESS in 11m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7154cf00079d47a4b315bb6532f3d53b : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b209486e2c54057affc46a3f3b4a58e : SUCCESS in 46m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2a7fb8daa9a44b92bcc790c554a05165 : SUCCESS in 50m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/917540bd6c0e4a6aaaba4e345aa9a068 : SUCCESS in 59m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b90d22d0c0684805a5123f96990366c0 : FAILURE in 34m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/686b3394642d466b9b4b8a73472a34f5 : FAILURE in 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0109352ab7e14e2c8019e22d1adae127 : FAILURE in 1h 39m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/eadef8e3930c45df8080b1f1a385a233 : FAILURE in 34m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f6dc89c6ba64cbdb29a9111aef317fa : SUCCESS in 1h 16m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/114145f295c64685869b407d3a226564 : SUCCESS in 58m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e099e5067b3b4af795c9c4092706723a : SUCCESS in 1h 37m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85462f3830094a909d653a2e0239b427 : SUCCESS in 1h 10m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1bf7d555886d4480b3fe81a770d836e9 : FAILURE in 33m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2008a73db5d841f0873c517d483048dc : FAILURE in 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55bfb222714fd750fb90b5858dece399d4ea299a","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-06-05 09:36:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4576e798813ff2686a2f01f45cdec8ae11987c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:46:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e03dff0c73b493ea6e8b66f3bdb1f81 : SUCCESS in 9m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a1e8494a7a041e6a45d7ff7460251ec : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8414be4ace5648bf9b69c28864f1417c : SUCCESS in 7m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7990b702cec648e5b13ba3a19721adb7 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf816d48ce1a477c996c71091ee1e1ff : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0677e69d79c34f34829075c77dba024c : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d471a158a6240708958d6609817a3b5 : SUCCESS in 13m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b7ce0057b24e4f8fbdf2ee862563ed76 : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0dcdfc842109422aafa1fe362458cbd3 : SUCCESS in 9m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/53672db255064fc7ac6d636e1aafd43b : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1953f99b3ce3445ea68daf5d17d9505b : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc8d34044b4f45a183ca3cfe8df0072b : SUCCESS in 55m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0a20cc09b32c4244921142b0ef61d925 : SUCCESS in 51m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb3a4c06cdf540d7951ed8483b5091c9 : SUCCESS in 1h 55m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2870e8a2beb24e15a5291f86495b7148 : FAILURE in 26m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/490ec6f0cfc749dc8b79c007c5f7e914 : FAILURE in 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/929cfc29158a42fa8321c00294535e74 : SUCCESS in 2h 02m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b397c43ea9241ec8d46981b01af8b5e : FAILURE in 47m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf4a7ff4021a4c41bd3ba6c0fdd094cf : SUCCESS in 1h 29m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0ef17b8af7f4ad599b87403a381bd38 : SUCCESS in 48m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc3b9a8896ad4ea4b4066cc240a8d9c7 : SUCCESS in 59m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b922ff9b06a40d7a8c9c0aa3d20495b : SUCCESS in 34m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d321f5325ddc4fa99e167fb4b7ddf7aa : FAILURE in 31m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cb830ed4c3c348d3949cebb464e3399b : FAILURE in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26ff9f5c713542ac08dc4a1386bb2e568f191c57","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-05 11:47:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f0c6a54d0705d6a7a9297cb5a3ae27a0703c366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:10:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce70b857f87e4c8f8679f0d5fc9208a6 : SUCCESS in 11m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/165643d646144ea4bdaa74dd0608f10f : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c87065d48b09454e93e392d905b66d09 : SUCCESS in 7m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/648e8abeb2e843caa8c66c46c5cfa361 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52ed2fc4a357420787c7ae9e18ebbe9b : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95978265d0cf48d4bb10a6283762cb46 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f0454764e34c6ca401611bf414c98d : SUCCESS in 12m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6e9cef765bdd46889b36c3f27d662cd2 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/82ebf0713fbf4bb4b16e1ffb358756df : SUCCESS in 10m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/047078774e1b435bab3511967448036b : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b36e96c7ad34c3e9eb2edbaa18e6710 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0dd19b7312946488a8736a12219604f : SUCCESS in 53m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2d9c33a71ba346d2b245cd0ae6fc2cdb : SUCCESS in 51m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae7e33b3c2cf4bdf851ffe0aa1609f1b : SUCCESS in 2h 16m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0e7cd68e3a204e529779e079a8f355c4 : FAILURE in 28m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/aaa8c285b02f40378c28258f4dc5559c : FAILURE in 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6f5829cb15b4253a698f2e5b6a02d0e : SUCCESS in 2h 07m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c625e30139d3416cb44f7d3129ac0a04 : FAILURE in 28m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb824b9115484969abdddded611bc087 : SUCCESS in 1h 16m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/231747b96ed24d9fa766b32ad087029b : SUCCESS in 1h 11m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5f6d0dba630479aae44222fb638b7fe : SUCCESS in 1h 42m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36f05cba7bda4172bba02a6b81d69cc2 : SUCCESS in 1h 09m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8409b29bc203489a8462a4f09d89e7ea : FAILURE in 35m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/52b99916b04c4e6c90372fc89ecbdedb : FAILURE in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04d074985b83a4a997a79219f0f2cbfedf343f81","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-06 01:49:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6fe89e2d8996ef2ce0422022ecf6d63cd0f1ebe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 03:37:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/507a5882b62840f1a2f35dcdf5df3ff9 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4e3ac2aee6f45f6bc57eb8591058541 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c2f8c7e663480a8ae88d29a080b20d : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd2d19858e5d4ebaa77c280325044a96 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c4ad0c442d44a7fbabee5548a4e2fb7 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/955f48be5a29463286b621fbebedc1a4 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400d65998fb0419e875427acab7d7f97 : SUCCESS in 13m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a1de4d7d50b542d7820ca27b9034d5e2 : SUCCESS in 6m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/14f3490888d441c883824d6c89d7ede5 : SUCCESS in 7m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/55b7f05552854f1ea6abdfed4c484b63 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3799e0c1d73547328973bfef66d5f8a1 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30d8effb7a8a4cd0aa3514cbbb3847b6 : SUCCESS in 44m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1beec8f19bc245c69fe2ed5255f66d9f : SUCCESS in 40m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72f28076a2f144169740cb5eff955fe4 : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2df0f41e4ada44b094d1ab366f15d5a2 : FAILURE in 48m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/46a7fd503d1b4bd68bac49a44f1f4d1d : FAILURE in 35m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9ed884b78e94178b1efb6ceb2d2a43b : SUCCESS in 1h 08m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/29e86e202a67489fa359f77042a4eaf1 : FAILURE in 26m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c8ad43436ea0475c9a3871d72cfabd61 : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75cbb993225942bda4f5460c604869ca : SUCCESS in 43m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eaf49852304042619de68d26dc383623 : SUCCESS in 1h 43m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/263b01977dbe4f6cb081e3c1c0943bc2 : SUCCESS in 1h 04m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cfd0de7615654e269ff05c9e02c31ff6 : FAILURE in 25m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0a57f1e6feb84fdbaa06034b538e00a6 : FAILURE in 28m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d8ca63926d12958d8f3e6497f76602acfdc0738","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-06 08:29:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0f5f98442bdd8cdd0411704d874c995d62b3453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 09:51:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b4cb76705c1427b9d8b23b192bd46c9 : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b05d7e7b38645c09aa24ef2057fd8eb : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640532f59e6f492795df67754054a04b : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/964516917086488897f76564773f5bed : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46f06cd2053143e481600564ae9f843b : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75abe21e6223400e9a0e7328a9a9759d : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2e49ece5d242dba79580396279616d : SUCCESS in 20m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7640860166e845b7bcd357810201a7c8 : SUCCESS in 8m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e0307a83aec14f12a3e466a14d539efc : SUCCESS in 11m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2633d6d9ff8241a59b661c5c12cf3257 : SUCCESS in 10m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dffabe51cb74693a8e754b72f1b6ef6 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23b2ba95d6634454a63e7f0bce425722 : SUCCESS in 1h 06m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8790e5326f7b48f1ae1fbb7fabeae5d4 : SUCCESS in 1h 08m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa205335d92d415fa24e90778d4bcda2 : SUCCESS in 1h 01m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/26444dd94dfe422488d01f835bb136ce : FAILURE in 50m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/31094148e087458d9963b2ee7be0b0c5 : FAILURE in 51m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/03d4c414d3a24a759197c51ca5ce00e9 : SUCCESS in 1h 16m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d45f0d0131874d91b939a7d4932fb734 : FAILURE in 33m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa82d7a1b74047c6bd4462eafc368f91 : SUCCESS in 1h 16m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4614fbb0b5634f25a986051072fb6395 : SUCCESS in 1h 09m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ed3b1e668374b6f8aeb6f207c2d5ab6 : SUCCESS in 1h 04m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d1bab89e0974f539b73ff1aaa7bb1b5 : SUCCESS in 59m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a75396806b7645388da07291aa756163 : FAILURE in 22m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bc74f719d05b4ef08f299fa15e52acf5 : FAILURE in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc20b4b4d0999456dac9df5ab80f9cb6959c5ff3","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-08 01:37:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11548e638cad896f0b6602d9832621c0ea25db96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 03:53:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bc013b87ec642b9b923040c93423f7f : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29fc4d76daf44e388aafc2dd39c88057 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9030fcafd23244a9b3ec972f96994fa2 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/085bd4ef2cf14b12ab234e1ea96466df : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12dc0ad597ef43f080806e4c2dfe7c0c : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a3098b8412945f5b8108883321efd64 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fc9f47218e4c4383b50d72b5a3ecc9 : SUCCESS in 9m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aa232b1802324f53b5852b18b6764f46 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ad76a6ce023f42d2b29bc5e83880bc97 : SUCCESS in 7m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6e0f4aff33894b94abc3b5b9cb9d9002 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef4964fed2c247d0817b3d3b95f3566b : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/805727a350cd4e95a9c322653425bbf3 : SUCCESS in 47m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/92ccf093aa9840f7a45c80636020939f : SUCCESS in 45m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/712ca85290144eb8826394c31c00bcfe : SUCCESS in 1h 49m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4cffd386e8f747d78b78d981aa6f2b06 : FAILURE in 24m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5b33bdd14ea3410caf4d9d8b53b96956 : FAILURE in 36m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0467bf7caf1e45c7b91b5327769d8900 : SUCCESS in 2h 10m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/37cf9cfbca164526afd5da1b1def016a : FAILURE in 26m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/071918cf95fb44fcb6ddf467a98cc96f : SUCCESS in 1h 14m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82426831bece48718df52e8bd1121136 : SUCCESS in 1h 00m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c7e7b52c870d49079e4ff165fe3453ef : SUCCESS in 50m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b206c85d8824a259f843f71766e08bc : SUCCESS in 1h 02m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/91827f659e6b4586bebff8859dabf301 : FAILURE in 33m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0426759c1c39492aa0b71d6db25cf042 : FAILURE in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc705e90a54aa44419b768d1742a2685260129ba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-08 05:57:21.000000000","message":"Patch Set 1: Code-Review+2\n\nPlease rebase on top of https://review.opendev.org/#/c/733681/","accounts_in_message":[],"_revision_number":1},{"id":"c4251ba0f38e99c406737713cb11b5441176f465","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-08 11:48:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dc1074e5e6f34a17ecacba89d048fb94ed09f517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:28:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c21d34d27c1a43e593e2d1cfb3182d50 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a593146d471d4dc1b04b0760a10e7500 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d61a2545b5749738fcd86a4f4a9c393 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2246b5850a4943c1a66ed1d989ca3bdd : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb85ecbce2040f086f8fe247e5bebab : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/718fc780937440e18e67b9a27afeed7c : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c218fa4232ea40338efff89c96885673 : SUCCESS in 11m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bc36ee7b35924c4c89922d1986006bc1 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d80db09e6f9d4739abbe90661589bdd6 : SUCCESS in 8m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7170b10d758646c29d3f0a0a40a4b57d : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1edfb7f66262484c9e3b2479a4e32daa : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ce9447ed2874ee9935ee9dcac3136d8 : SUCCESS in 43m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d83db0eefeb6401c9aea31e6b59dada1 : SUCCESS in 48m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d428a66e7874e818ba710411faf1441 : SUCCESS in 1h 51m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bfb4bcf38e434bb0a55416d01269ac13 : SUCCESS in 1h 01m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9ac5838fe9284799bc12e788805e9480 : TIMED_OUT in 2h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cfe42dbe8d3b41ef9c1137f7089f43e7 : SUCCESS in 1h 55m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9373d5c59d74450f88179a249aa9797c : SUCCESS in 1h 29m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98741c997bc04e4eb63993be8875f560 : SUCCESS in 1h 22m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe1d4f03710c41618757b2ba657bd58e : SUCCESS in 56m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/02b64d6057eb42ec8b2d0509d72a433a : SUCCESS in 1h 59m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/580a0bb8e69440c1b7863d16e6063cac : SUCCESS in 1h 06m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/12b8a43adbd84128bcbb1ba192662851 : SUCCESS in 1h 22m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4597cdcec9124ac6b7bc6803b4389d4a : FAILURE in 1h 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3177414dadf5a5294f69a42eada78fa2f843590d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-08 14:44:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5683b4008f31d1540116c0a3db73d55a713b46be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:44:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14b5418eac475d6c27206427f52a8ccd9d40a908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 02:58:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55314be8030341efbf6553e5f7f9dc79 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05dfb138b69146d4a02395371f190de8 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc2c5a5bf1e74affbb55f576f281cef0 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3622fcefb8544a7c9fef2982e11ab47f : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5c66e389a6147a787dd1c3eae7c7c49 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcba20b879b743169f287a8853d179c5 : SUCCESS in 11m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dc664d08e1314c10a64a699dc37ce0da : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/360a70136b064237a9953987fccc3ffc : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc61402ae21c473492818deb6dce6041 : SUCCESS in 52m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/241c72397bd847a4947b2628f584d25c : SUCCESS in 46m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c10803b64e4efb9c31b0e59a4f5bb5 : SUCCESS in 1h 00m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e5da5004a983409eaaf03a4e5c85a321 : SUCCESS in 1h 51m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/758e820eb2ed45f291417997e78436db : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":2},{"id":"165ef72f14a48ec3c3e241544378d5b7bb81cd72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 02:58:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a38f702abb02366dabc1c4ce6a4ebb2e793cf224","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 03:00:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69593a5d31e0453a9e78a532bcee7b5b : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e358637a10e6b3c2df336d8dad42cecbd97e3dd5","revisions":{"e2f3a77d60e467427be066f9519757c5c4e96e75":{"kind":"REWORK","_number":1,"created":"2020-06-01 01:29:06.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/79/732079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/732079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/732079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/732079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/732079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/732079/1"}}},"commit":{"parents":[{"commit":"09e863eb539e29c8950ddfce12d602c0e77c1b7e","subject":"Fix getting user neutron client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09e863eb539e29c8950ddfce12d602c0e77c1b7e"}]}],"author":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2020-05-29 08:46:54.000000000","tz":480},"committer":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2020-06-01 01:29:06.000000000","tz":0},"subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","message":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027\n\nstory: 2007734\nChange-Id: I6543fe5caf539c9b1167a7c58984b3474879e1de\n(cherry picked from commit 7c96e455a9f7818cb2c5c16ed12120e719f8d992)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2f3a77d60e467427be066f9519757c5c4e96e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2f3a77d60e467427be066f9519757c5c4e96e75"}]},"branch":"refs/heads/stable/train"},"e358637a10e6b3c2df336d8dad42cecbd97e3dd5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-08 11:48:04.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/79/732079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/732079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/732079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/732079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/732079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/732079/2"}}},"commit":{"parents":[{"commit":"4a95b354901c81f135b61d39754e1419dd134f0e","subject":"Do not install diskimage-builder from Git","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a95b354901c81f135b61d39754e1419dd134f0e"}]}],"author":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2020-05-29 08:46:54.000000000","tz":480},"committer":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2020-06-08 11:48:04.000000000","tz":0},"subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","message":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027\n\nstory: 2007734\nChange-Id: I6543fe5caf539c9b1167a7c58984b3474879e1de\n(cherry picked from commit 7c96e455a9f7818cb2c5c16ed12120e719f8d992)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e358637a10e6b3c2df336d8dad42cecbd97e3dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e358637a10e6b3c2df336d8dad42cecbd97e3dd5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
