)]}'
{"id":"openstack%2Foctavia~732375","triplet_id":"openstack%2Foctavia~master~I59d94bc9743b04674caca3ffcb8fc91ba803298c","project":"openstack/octavia","branch":"master","topic":"trivial-fix","hashtags":[],"change_id":"I59d94bc9743b04674caca3ffcb8fc91ba803298c","subject":"trivial fix","status":"MERGED","created":"2020-06-01 14:07:42.000000000","updated":"2020-09-02 02:08:06.000000000","submitted":"2020-09-02 02:06:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732375-1599012401050-436c11b4","meta_rev_id":"ca4fcd5d17b4b9577b0754cba00563e9f9af2ec5","_number":732375,"virtual_id_number":732375,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 02:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-28 05:56:49.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-08-27 12:07:12.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-08-26 21:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-08-27 12:07:12.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2020-08-27 12:07:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 21:21:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-27 12:07:12.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-02 02:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0ea874570a6a595efa4a80e329d5a711298d098","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-06-01 14:07:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e6c710c307eb48052c7c98caa879f425b934c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:09:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffbd24fe0b9f4e9094b305711f448bb7 : SUCCESS in 12m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e97326e20f146b591a145a507045555 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eed25e4b3df47ccb7aef4e8c38d6def : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ce9d2d284494abb95bd607e0382874e : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdc8011b2b024573b954c3aee1b307d0 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18cf319596e4a4397d7f5c82d248a35 : SUCCESS in 10m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c51857ebb90a442f9038751399a20924 : SUCCESS in 9m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1a16216d783d48af841556774ea26dfe : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8a9710ab786464394952dae31a6d92d : SUCCESS in 9m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00e8d2a2fbea4330b83daec5b25dfc01 : SUCCESS in 41m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11d1b6bfb6184ab8975f2d7a3f5dd2e2 : SUCCESS in 1h 46m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db591b20299c4c1d90e74af3f39d5aa4 : SUCCESS in 35m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6b0a06cc5ac44b0f96cc347f6268ab24 : SUCCESS in 52m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9e7084d6e0c46ba8a811af837b43c4b : SUCCESS in 33m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/47517ff3679f4b52aea0e86340b776aa : SUCCESS in 1h 31m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5fc5de3c85a44ca687ae58fb6d03d8e5 : SUCCESS in 1h 48m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e691ce848a15476a8fe64d1a9e133158 : SUCCESS in 1h 55m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7fd27914c1054ebab2b7cefdbd8bfb82 : SUCCESS in 1h 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9fb98fe8acf88647f860cf91926ce902dd3fd8b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-26 21:21:43.000000000","message":"Patch Set 1: Code-Review+2\n\nWhis the commit message was better, but it\u0027s just a typo in comments, so LGTM","accounts_in_message":[],"_revision_number":1},{"id":"8ee57058dbe653f1c9705832afa72eedd772253d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-27 12:07:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d00c0a3a6b5a71a3209605bfdde6e8a21e6d60e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 12:07:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b676b719e059e16dbf680eb34ddbea5ea257639","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-27 12:07:50.000000000","message":"Patch Set 1: Backport-Candidate-2\n\nMinor thing (typo in comment). Not worth backporting.","accounts_in_message":[],"_revision_number":1},{"id":"f69416013a269cde206525b28843c4d1e73ae87c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 12:24:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27a3c66c03644b5caef8aa9b4554d481 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af0f18369774de8960861d743ed3136 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/685db2ae3eaf46dbbbd898b66e7f9723 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7d8171f2874435cbbd599b327b3c7a6 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d067c8c0bd2f4dc2a4fdf408df9de850 : SUCCESS in 12m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c5d57ca1d524068b4774342c3fb34c0 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c1aed12ece074002a063b6aeae621436 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ac09a03c9c2426b8c671d635c63e23d : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c5d9d28f76bb432fb2181ec3ac4ae02b : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/85ec13edb39c4133b508a2b7ff734c00 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4cddace3fac47409f3506f5165bf0dc : FAILURE in 10m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/23ea8e183c6f4b66a0bdd219ff3590ed : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"1061b6beaa50d3a95915e9067deeb041f06185d4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-28 05:56:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5613e7668eb07845d0938526b0775774c4a36475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 07:50:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8885cdd363644a4a961a8c5cce7da212 : FAILURE in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f22c00ae45f641558127019c3ca87d21 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b1baf180574fefaf73b9c332e1fe7e : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c01c0fdcef842c5b01b4f64ab5384d1 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afee7b01cc0c49d7b3aa7b4c8695b06b : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b949347bd34c26ac246629bbb56c5c : SUCCESS in 9m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2ba3db547bd14310a56d4bdaab5976b5 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1cda264fd95b470e87844436f480cc78 : FAILURE in 9m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b301a29db62d4e688f9a0e27450a9226 : FAILURE in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53987a651de8467ea3e4d039c8776001 : SUCCESS in 34m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f55758a90b946ddb9b976ecf103ba27 : SUCCESS in 1h 15m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75e9f63f328d496a8f31f153a7c3ff8c : SUCCESS in 56m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/caefc3ff163048d2bf7d4d9059a1891d : SUCCESS in 52m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1aee5111dcb24ab297a3785bd4521f3a : SUCCESS in 53m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/80e1a052622d4837ab16ead2428be9d8 : SUCCESS in 1h 50m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bbc5783f519249b99bd326c9435f09bd : SUCCESS in 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ebfc0228f9844541bfe38d838b0c5359 : SUCCESS in 1h 11m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5416cdc7d2ee4107bec8772f9ac99bab : SUCCESS in 51m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c132d9deef349929cdd9528f8f15159 : SUCCESS in 58m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/519688b688ae44478a4bf6d6f19ce4ce : SUCCESS in 46m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/607c50b1ccba4d74abc925b8b36beec6 : SUCCESS in 1h 03m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1a577decd432e414f14f6b0b98ea5846d9cef97","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 05:12:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"686e2de42ba1571cf80eceade70b6a044c4c1781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 07:14:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52383d0ab4514fe68dd7f25c0dccaa04 : FAILURE in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c85a14083614d59b4cfc84b5c4ca666 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1325525d81624dd5bc4bb55737641d61 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/206e4da10afa498d86f9e59811035c34 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/907f78a73e1c460ba533431af204cb52 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f2a8c37f3447e996d971752a9cd917 : SUCCESS in 11m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9cf0fbb4022044d58a5151c21b4b3ca7 : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9d949e0408c44d42a2b3d9c24c0d2b4c : FAILURE in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f8dc16f7ca24bda852d34e94d3d11e0 : FAILURE in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d7c61f765cf4ddba3e83e14110ea3ad : SUCCESS in 43m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0efe7357a93c42be8453ad5bddf1e559 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9b1354c394f647da9a755f21f0e9462b : SUCCESS in 41m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9aa82197d2164eceba6a1b6dc61884ab : SUCCESS in 1h 03m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b27fe697cd44e4898a0d64dbc21283b : SUCCESS in 39m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2371450befec4f2d83e7857bf188bc44 : SUCCESS in 1h 58m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/301f38b3f5664da0bf8fbe7c52c55c51 : SUCCESS in 1h 25m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fafd36ce602843d1806b793aa9d7d53d : SUCCESS in 1h 12m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae09a6c266d344858150febd17a32166 : SUCCESS in 1h 18m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/363f7b1caff244148c2fe26bfc5dfb3d : SUCCESS in 54m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e9a60864081d4c06bf1b0a10b75b5f37 : SUCCESS in 50m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2d242442b0b246618c0d5db9b1092fe7 : SUCCESS in 1h 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4958a52e1e7acd78e5bcac711b5fb85537462223","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 23:38:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd94e4c7a86e252ef42f799e885e252e09d653f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 00:57:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95903d6224fe4389bf58c17a84ca8bb2 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68eea9f170e043feb08851ee3dca0a66 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2ffe97bbaa94519a4e6f7351d2742c4 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ebf3cdc889b4300859869f5b0f0ff6b : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cbc1af82098460496dd06fbc4cc21b9 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f485b6d970c740a699b07c0071cef0cc : SUCCESS in 11m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f3c6a40e233449f7a8b23c663de24fbd : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ceec44b80d84481fb8b74b34ec5abfaa : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7115e7a987e441f9a44bc516d3b07afa : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59dde48d5d85426f808270d2bb4a1f9a : SUCCESS in 48m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fff932ce3734f8bae8b188f4f2fa74e : SUCCESS in 47m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eda837dcdf9046f0a1eadb68eeadd9d9 : SUCCESS in 32m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e319a2435d1c4e8292a25223f2efad2d : SUCCESS in 31m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12a6557b489a48719f76c24c1e711556 : SUCCESS in 31m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1fd33bdd05e491398a3efaf11123c3d : SUCCESS in 1h 03m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ed52590de624e05818a856decd6523a : SUCCESS in 46m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27cb922c9c64458892fd42288957a4c8 : SUCCESS in 49m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db7df8e9f477482588942e41aed2ae72 : SUCCESS in 1h 00m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/36b422f7647b41c9be46e6bd7a7ed613 : SUCCESS in 1h 17m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9c6d970c15fe4798ab1106686c221a74 : SUCCESS in 1h 18m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3dce36f1ce3a4effae591a8d5d271095 : FAILURE in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9a7b49294ac90ea418cd9c475391128591c9982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 00:58:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dea37954d85aa608735f676323a6c91cce07640e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 02:06:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c31e6093d6b04fe28ec36f7b5a713e27 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/084c31210f7441a7a59feba771efc199 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/599ae11167284c69ae51ff19a6af61e3 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e4da43543254c448f390d897d1e2cd9 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528cf27ca5824a8d8d0d488f03019ea7 : SUCCESS in 10m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb67f68c30cb435c9d37e1464c3b445b : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90b69585ffbc434b9016432e9c3c0da3 : SUCCESS in 48m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b573bdb4ec048f58c6f4f1b238de577 : SUCCESS in 48m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fb9e4265b32243d4b6b7e5b7bb28e0ce : SUCCESS in 1h 06m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cfa231fcbe064634a8875af33ef4f0b4 : SUCCESS in 30m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce506a39a5ca42f9817d6ad0c514a9dc : SUCCESS in 38m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/edd5b69a69fc4ae088a39b2befc29bd4 : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bcba047d8d1c1b128d54e9c4c6dfdfe08e6e5a4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 02:06:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4f8896ff3e5044f256b59f2612cad0fa8b5098ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 02:08:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3f1344aa234cefad43c704d44b8f30 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bee93b20189136f5d8d4de7d0ed38f66d776e4e","revisions":{"9bee93b20189136f5d8d4de7d0ed38f66d776e4e":{"kind":"REWORK","_number":1,"created":"2020-06-01 14:07:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/732375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/732375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/732375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/732375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/732375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/732375/1"}}},"commit":{"parents":[{"commit":"2c7c7747b74e866e5ab41139e2f3f01d9dce15a5","subject":"Merge \"Workaround peer name starting with hyphen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c7c7747b74e866e5ab41139e2f3f01d9dce15a5"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2020-06-01 14:05:12.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2020-06-01 14:07:38.000000000","tz":480},"subject":"trivial fix","message":"trivial fix\n\nChange-Id: I59d94bc9743b04674caca3ffcb8fc91ba803298c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bee93b20189136f5d8d4de7d0ed38f66d776e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bee93b20189136f5d8d4de7d0ed38f66d776e4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
