)]}'
{"id":"openstack%2Foctavia~732430","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ib0fc26877710dea423a5ebcf1f71077665404377","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ib0fc26877710dea423a5ebcf1f71077665404377","subject":"Workaround peer name starting with hyphen","status":"MERGED","created":"2020-06-01 18:04:01.000000000","updated":"2020-06-26 10:16:46.000000000","submitted":"2020-06-26 10:15:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732430-1593166518304-0147cf3e","meta_rev_id":"b91bbbcdd934ed9857334629e9d308733dd6f2ab","_number":732430,"virtual_id_number":732430,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2020-06-26 06:23:54.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 10:15:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-06-08 15:54:15.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-08 16:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-08 16:13:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 15:54:15.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-06-08 16:13:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-26 10:15:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"786cf47fcdee396599c04561a8d1bea6366a47b1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-01 18:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2227b80a792324b3d61910e04499586b4f6b9a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 20:35:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69fd15b92bdd44cf985d53de8feb5817 : FAILURE in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30e5e913fec84a8d9263219c978cf838 : FAILURE in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8e8bad7903443cbb17863a6af0d981 : FAILURE in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d4318cbdc054b6c9514cbc44cc7faad : FAILURE in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05406dac2a4a4d52959e48e75413e4d2 : FAILURE in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b3705bf27e649158818f639e53310da : FAILURE in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98cb9b069a049ef87720ff8a360ac0d : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdc2b1c5c368455a97b45ae006a2cc0d : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4f20acb0d4284fac944951edb23bb267 : FAILURE in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/60d4a9b293874a2493699de0a2cbff79 : FAILURE in 9m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/016c109b19e14a229b97e700f99440b0 : FAILURE in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71523874b0794ce39da0eb396e2c3f68 : FAILURE in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0242dae59b440d7a8ca42934b313d2e : SUCCESS in 38m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/da432982cda64f14a5a03a7bb6052a22 : SUCCESS in 45m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0e75ac2110748d7ac684bbc841880c8 : FAILURE in 1h 00m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7308c45ce543422fba96b1e8c5dd068c : FAILURE in 52m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/760ba468dfd545c3beba60e86fddaec3 : FAILURE in 1h 15m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/656edff39b66491b8fb06b833caf9c27 : FAILURE in 1h 07m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/65e22aeabd4e4a659f3e34cf0354fa54 : FAILURE in 2h 00m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af00f8c3d0614ec6ac011da2d607a962 : TIMED_OUT in 2h 11m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d39802aa65e459cba86254960c71e15 : FAILURE in 1h 43m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/db902d96a0a4420d999e2b462b4fc347 : FAILURE in 1h 47m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/402e23982a9f4a3fa3d0b3176ce6c993 : FAILURE in 54m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7b7cba69c96a4168a3c529b7375d16be : FAILURE in 1h 37m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fc0d9759a285438aa8bda8ac9297b892 : FAILURE in 1h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c36794cd17367edcf49888698828239bf5ca922d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-02 08:59:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33a481b7e9ab74b3870a748557663f32dc31e9c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:49:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4a6aea635c045b3b0a85c602921fa7a : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/599cabb205954472853b21c94800b654 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4fb9ef67db4abd9081e5a5aa04a0af : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa5b4ee783a54041a8463144c91c93e5 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1154fec4cf7e4db59f41539b3a692216 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/828c7e56c4d54147b2d141887c96fcd4 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/056a8a578d744a7585751883b27e852c : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a36505a03774b348cbdf7c6faa78c41 : SUCCESS in 4m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4b1f07e0091c483daed2bb239d820683 : SUCCESS in 7m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bcb8d99d52654a3e9e9b147e4299adb2 : SUCCESS in 9m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/774a9153df754b94b85afd5caab67c0f : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb332d79c96549b69141243667d5a771 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1800c76d448548fa8e6fd171e5cd641b : SUCCESS in 38m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/647c0e3a598a4f048b98f302d20ebd2e : SUCCESS in 46m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f713bfc64a14124bb49f4db65c0b71a : SUCCESS in 54m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6feeb74d3129499ebdbebeb948f89f6e : FAILURE in 24m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a147c08d466f408a9b0430b58423df44 : FAILURE in 36m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a1df3a974ec494b8d5b353d5dc855ef : SUCCESS in 1h 16m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ba5f67929144d59bb4f7f6e16d67f1f : FAILURE in 32m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd65d5ecaa104b839f4a988415508043 : SUCCESS in 1h 15m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71fff4ef28ce41f897ceefbdde76cf40 : SUCCESS in 57m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6c2a175dbc1549beb06cac9a1de4cbd2 : SUCCESS in 1h 44m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c812b3cecc3743d09c72472985be4248 : SUCCESS in 1h 03m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/83e3ec66e6594f61bbe7affafca720a7 : FAILURE in 28m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/becad21c42cc4bc6bde8ee57c1cd4567 : FAILURE in 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"215e9d166bb0f3d2d917b33ab7554df16e477b4b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-08 06:03:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"096164ac29ef436cf9ae2225fe207aff6d124b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 08:47:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9e01de42b2a45e0a352426803c5c5d3 : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ab5e0cd1c404f5bb772d687b712ecd7 : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bab2e5f0a8d481c90d4c942cd6b1377 : SUCCESS in 8m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91715c505e7c40e08905a342ea25735f : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f725a058982c465fa9ce23561020a3fe : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9907ab6111e24a42a31a5438c0bf472e : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0677c8f46e442c1812826c46b9589ad : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26abdcb189e84155977456b78ae43fd1 : SUCCESS in 8m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db2f18d64abe438fa0605c640a405c38 : SUCCESS in 9m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ae4e0cfb76bd4146aec99a4e754afd77 : SUCCESS in 11m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5ea7aa244c51488a922704a20a8fc3f0 : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b30f051f1b4b4beba71940e6104b8a8a : SUCCESS in 10m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/057537104be04fb985c221ee8b537986 : SUCCESS in 50m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d950aff6d7d74e23bfffbb532f375f18 : SUCCESS in 46m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78064f97fee440878c443287847d3169 : SUCCESS in 1h 59m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6d958957cad74c179cf1fb2446f71a9c : SUCCESS in 2h 24m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1f6b22a308b14c3eb4f5eae115021831 : SUCCESS in 2h 17m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe744a577b66419884ba9001e12ad4eb : SUCCESS in 2h 14m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3074e84869f44a7ea4680999c3ac2d73 : SUCCESS in 2h 13m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b02f16928e874523926edfa09754d902 : SUCCESS in 1h 17m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ff6a94f13184ef3bbf4efd38f396a59 : SUCCESS in 1h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/52db47cb8edf4ee0ae85216e39c817ec : SUCCESS in 1h 42m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b43c1b21e50f42c0ae44a380aa73af12 : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ac5a50b29d0f41e2a326edea08968aa6 : SUCCESS in 1h 22m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d0af50cf8c3e428dbbde92ad7fed59f1 : FAILURE in 1h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"216e45d807730e72132719ab77f44dfe646db366","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-06-08 15:54:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0d4ac178cf036170784c25d21d55ec61378bc612","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-08 16:13:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"6ad68801954b262f43ab5aadbf89aa5411556202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:44:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"afb06ab4a6ddf5b421951cbfb9c8dc3bfd30b824","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 01:22:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/169a1a4895204170b341aafe9ce220b4 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0355a688ff64bb59b46a9ecb50df2b7 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fe7f3e95ea94b27a71c142f36edd518 : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e248223aa6a44448f69c574200bb827 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/928ebbca6d1845cfb28391a76dc3bc50 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d58c7c1ca546f7b3146dbc98e338d3 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e77a24ff5ab54e99acd2f85f495396a4 : SUCCESS in 4m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/70bd3edce5ff43eca55259188708bfbd : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc9cd995dc5e412c84d987b603ec1817 : SUCCESS in 14m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a8cb2e2a4054aadb1ba8be94078023d : SUCCESS in 41m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/350098a466d64a0ca4952dd46100f460 : SUCCESS in 46m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/447b206e8a6d4b67a50a6717c97d06ba : CANCELED in 1h 20m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e07ff8bb1e6e4d37a9a6af2b3611231c : CANCELED in 1h 20m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6d61cf9e140413aa1b6b42554fbd8f5 : FAILURE in 1h 20m 09s","accounts_in_message":[],"_revision_number":3},{"id":"9481c98a9d6e14a24f40c5c4600ffedf3c155bc5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-11 09:21:05.000000000","message":"Patch Set 3:\n\nrecheck\n\nComputeWaitTimeoutException: Waiting for compute id 0aa1ce69-d1fc-4c8b-9605-e4b8f1066989 to go active timeout.","accounts_in_message":[],"_revision_number":3},{"id":"fc36336cc1257adfab5cefdfd972e5c713b5b027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 11:50:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0c329bc9b0e4cd8a718124aff0f1324 : SUCCESS in 12m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6cfb4c124f44ec4a2de7efdeb234e4f : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c9054762734bd69ae1a8b1727253f9 : SUCCESS in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dddcb200b829462abb6d4467e629deca : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/374e48ef1c5047c3a5624b331c27b727 : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17026e7b18f64fb385e51664f61002c3 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda7447d1bbb4e339ac2f1e40d09dd56 : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6f01cd09ef247dba7746e795c7bbb1b : SUCCESS in 7m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d1cf442f10f04699a1272845d7aa6c69 : SUCCESS in 7m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/82acd00e76a143ebb0159d203c0782ed : SUCCESS in 9m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f4d3fb9538c24c78bf8bf47029fddfe4 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bfbaaaf92c164e1dacd08cf640618d1f : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1d9f42e916b4b95822cea45998bc4ed : SUCCESS in 44m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/38a66c396276449ca16040ded4ce414a : SUCCESS in 41m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4727879c1fd749d18c14e384cf64407e : SUCCESS in 1h 38m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/66c84336b1354f9fb04918adf40d9ff5 : SUCCESS in 2h 23m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/66d5d810a027473f84550e6668b30a82 : SUCCESS in 2h 16m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f69f0124fb544ab8aaf8f629ec0bad2b : SUCCESS in 2h 09m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/efa130f006c64540bec0a82b4572cf2b : SUCCESS in 1h 46m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0a1b686d41b4eafae59aa8cdccfefb6 : FAILURE in 34m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf4ff56f03674cf9b22465fc5ca11d60 : SUCCESS in 59m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/684d569cf96a40e98184170faa3c8f50 : SUCCESS in 1h 45m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/949883762c8e402bb3ed6ad0e10e31c2 : SUCCESS in 1h 05m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6bb6c30e04f044d6be733e661d090027 : SUCCESS in 1h 15m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/73ddf45a5ae94fc7b21a493b0cc1eefd : SUCCESS in 1h 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54862a8f3672bb6ee1b84e7e1acea6d519a4edec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 16:32:46.000000000","message":"Patch Set 3:\n\nrecheck\nnova failure, instance did not go to ACTIVE","accounts_in_message":[],"_revision_number":3},{"id":"763e4e22b5fd6085d076df83d12e31f344210800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:11:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d2d78991d244d31a42580a575f9c938 : SUCCESS in 11m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba404fb8e7fc47f8a43b0c2064e2593d : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa07ea914154cac93e04f93a6d3dd81 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4d75f5754744746bd6aa976647bec71 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da8dc3309d2348b5ab507076881d2783 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8697e61460344638b5fba5ee5e28ca2e : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d46183c975b4aca8ee69caf71ce9041 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80cda4a6a0ec466bab191d1917eff591 : SUCCESS in 4m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/864677c9c1334eb5a3ca1ef9611df4f3 : SUCCESS in 7m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4dfac5d139ad480aaf44505de5388012 : SUCCESS in 11m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3541c8b8637c47bdbf134b2a7ce3c6bd : SUCCESS in 8m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dbd83582d1ac499baa4e94b31f025138 : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55ed5a3a54b84b8885742533d2130e19 : SUCCESS in 44m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1367f1aedf5f4a359b05b8008be7fa83 : SUCCESS in 49m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67f335a7db624182aac1a4383d7e4405 : SUCCESS in 2h 08m 30s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f2c17d438f9b4493b9236c7ea624fac3 : SUCCESS in 2h 31m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/30ac3c06b0424f2094ec94e78a9baa1f : SUCCESS in 1h 28m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c03e68ae5ae457c9a53f088601219b2 : SUCCESS in 1h 47m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/390df3e721d1497b800522c8ad1abc88 : SUCCESS in 2h 18m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/710cfc5393fa4da3b39cadd8f2007da9 : FAILURE in 33m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/73faf8985e444e99a3dc81f5e50982a5 : SUCCESS in 1h 01m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4a6a65fcd49e4aecb4125c33e6ca94c7 : SUCCESS in 1h 04m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1d45b34b75d49c79f6c2836c598c354 : SUCCESS in 52m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8175f85deaad48bab0f89a35cdfd00d9 : SUCCESS in 45m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/800b9206e5d84c7f93e790701a127808 : SUCCESS in 40m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c1e70ab7fdde7c887eb15d6c584a7ad97813d908","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 16:44:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b277260ccf271b7fb5a6b36c36e941af1cd2a9c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 19:48:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdb89a61862b45e5bbd1e112a8bbbdce : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7da7a2bad52459e9a26120980a3063e : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3158b65beada4deeb8a7bdb3b71c2137 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c836977e28784f19a3b24957147e8e3e : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ff46aa36de24c6eb3097763e9bfb774 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40a25eb865d94a5ea5e93978b29d6971 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db556fc523649a4a27a761c820257b2 : SUCCESS in 14m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1027281fc2054d1cb602ac8121c683fe : SUCCESS in 5m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ca8f05371aee48ab91a9c1053b3caf64 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/877a0679014647c8a46bd38be77f5601 : SUCCESS in 7m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0c648a0ffee341e38415fe369db899b0 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff93aeee722b4148a54e289019c1dc48 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c11296f8d3b49b6954aae550030f7e1 : SUCCESS in 46m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/19cdefba84b64622ad058047cd168b44 : SUCCESS in 49m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/488de0581dfc4a80b8383ed2104b1707 : SUCCESS in 1h 55m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/adad31109203482a9e4cc4f7edbb6204 : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ae59434ae9ed49b6b12f3a59e8519fd4 : FAILURE in 12m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3303bc9a2e0426d80f7b79704714cd6 : FAILURE in 56m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/af59a08566a74ca9b716551fb83d7be2 : FAILURE in 56m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ff89fe1d8c4468a989fd8c2a648e776 : SUCCESS in 1h 13m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/880213a8336c4de3949ce0c810c8cbe3 : SUCCESS in 59m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b7e6ba955e7d4333bdd258ae0b9d5424 : SUCCESS in 1h 58m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9788702f99c417cae91317b6d8edaaa : SUCCESS in 1h 16m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a558a2fbfaef43928aa3714427a1a418 : FAILURE in 34m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0bd046d0acc0432ca3b9438e8b773c96 : FAILURE in 13m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b6cd09da9c14a9f29e8792ec7424263e1b57610d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 19:48:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a05bf088dcbf5b54fc69fc381aaa376d458392f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 21:44:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57d88a423c794d8c9ceef4a37b6bc9f2 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c669fda4b7da4e4d8874c8cd97d7e67e : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14e8cef4c1c3499cbb7a4d0f9c3e1160 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed00e1048827403792f54c0959692635 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9995edc849cb4c2697833a1b53182fcc : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc5c61d11d84bbf9ad6da7fd40b384d : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87f288e8053049a384ac37cca861d0e1 : SUCCESS in 5m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aa861f2e4f4f41b8a141b689988b4f9b : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b629faa1e74a49ae984f3c43e4ebfab8 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e6b10a7b3fb467fbe36356569f64a26 : CANCELED in 36m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/18c36d8ea20a4bd8bb206fa3aaf7dc79 : CANCELED in 37m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2a8074486d64a1ebcbf051d07edb568 : CANCELED in 36m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/24edb46e7e964b7fb7f98066d4838ce0 : FAILURE in 36m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0dc94134c4124aea97f7891d322f1edd : CANCELED in 36m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c551342fc93b4104753bc8ad55f1be3273e5d67e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 06:23:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a91bf776fb3ce94ace932e8a82f112dec6604cd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:32:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ed84ed5c961428e80dd7df3973890fc : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d938205fe6cb490e8d10798d7843bb13 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/035ce3ec570240139cc10be04cd844cd : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d67cf8f5a0c142d399bf12512d50ef5b : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aef86abeb15346c2b21feda1b85e5768 : SUCCESS in 8m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/494a3367f2284cf788ba3928957e56e3 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f3eb5f2aac4dae8152a177f7a7b934 : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfab135d4abb432fa36a702388d11643 : SUCCESS in 6m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef05f6639b2b40198d1a8fdfdee0544a : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5fe8757f943d4c3b989a50c467da2d4c : SUCCESS in 6m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d595a0742d7141f482dfb1928673ac50 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e34dedbe48af4f76803c3e38510cbea8 : SUCCESS in 10m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53274c4b6af24c01a492f08645e6bce9 : SUCCESS in 49m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d238424d324e495191f92a3109b6d82a : SUCCESS in 54m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b38b81e0c7734876b8144354c9a22154 : SUCCESS in 2h 01m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/665d145be4e449d3a37761605dd5b0db : SUCCESS in 52m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/13df1ffbdfbc46ff8b24a7211e9f031e : FAILURE in 15m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/52cfaefccc3a433f943885722fcc4c5f : FAILURE in 57m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5febe48373e4030b6ae48e7d36f61d8 : FAILURE in 49m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aaa65284be8840ad9f3dd2baeddd262e : SUCCESS in 1h 19m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e42c5ad7db44abd9132543abc3150a1 : FAILURE in 49m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3ef1e6da49cc41918ff5b968a823166a : SUCCESS in 1h 06m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/faf16fea67b042c2abfe81f25f3a8acf : SUCCESS in 1h 16m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/881ccb2336e345f9a79045df960b4c77 : SUCCESS in 43m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1f6e843f0c8546e3817d130bd7bf73cf : FAILURE in 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b0f59ff15b80b870958001cb9865dca5d02fcfba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:32:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"266587c6482b55ce62b9b6b50f4eb03cac05fdd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 10:15:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c7c165e14544fc69ccaf1dc8da3bd36 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c1468615b0405591894e173518d051 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c127866843484ab3b0fb15f4576e3421 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7390a7ecad944c3961dec0c1de0d42e : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cab9b5ae1cc04f3f9bcac68eafea7ba6 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9969d9d136f747b796e3d00860c449ef : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f708f6a6c7f9495ea8ed229f7a77de5a : SUCCESS in 5m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6eae1306809642b49198570bff0e8534 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f207a33041348c29023991cc3651ca8 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8714f812ae3046d499b6ebd23b47ac4a : SUCCESS in 41m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/72cd0412c4c247ffa8f0a3410c742677 : SUCCESS in 43m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5b8c0f5f5044573812a29f1ced1c16e : SUCCESS in 1h 31m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f0d015a7501646f8b5a62288968f5198 : SUCCESS in 1h 34m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82b5ddaa17584b738520d6eabc26dd1c : SUCCESS in 1h 35m 46s","accounts_in_message":[],"_revision_number":3},{"id":"f5db6d2e669ac37f6cc8924b52234460bb4d962b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 10:15:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b91bbbcdd934ed9857334629e9d308733dd6f2ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 10:16:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/539afc216bc54d308c714d33a50249aa : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea7f0a422d3945b1813990d4ecd04801 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b35f77d3b42997e882036e4dc9359b1afb78ffac","revisions":{"1975bd1d2926a1e30fe2e21dbf242b8d26e19d50":{"kind":"REWORK","_number":1,"created":"2020-06-01 18:04:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/30/732430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/732430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/732430/1"}}},"commit":{"parents":[{"commit":"09e863eb539e29c8950ddfce12d602c0e77c1b7e","subject":"Fix getting user neutron client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09e863eb539e29c8950ddfce12d602c0e77c1b7e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-25 18:47:34.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-01 18:03:45.000000000","tz":120},"subject":"Workaround peer name starting with hyphen","message":"Workaround peer name starting with hyphen\n\nThe base64_sha_string method is used to set a base64-encoded peer name\nin HAProxy. There are cases where the peer name can start with\nan hypen which is troublesome when used in HAProxy CLI. Specifically,\nHAProxy fails to reload when local peer name starts with \u0027-x\u0027 [1]. When\nthis is the case, an amphora goes to provisioning status ERROR and later\nis scheduled for failover by the Octavia Health Manager service. A new\namphora UUUID is assigned and base64 encoded, hopefully not starting\nwith \u0027-x\u0027 again. However, this is far from being ideal -- we incur in a\ndataplane disruption (single topology) or reduce HA capabilities\n(active-standby topology) for some time.\n\nFour possible options:\n\na) add prefix to peer name\nb) change b64encode altchars\nc) quote peer name in haproxy CLI command\nd) substitute first character if hyphen\n\nOption a) and b) are not backward compatible with running amphorae. Peer\nnames of existing amphorae that do not start with hypen but contain\nhyphen at any other position would get different peer names.\n\nOption c) would nonetheless still require an amphora image update to add\nquotes in the HAProxy init service file. Continuing to generate peer\nnames with hyphens at begininng of the string is avoidable and\nrecommended.\n\nOption d), while also requiring an amphora image update, it would get\nrid of hyphens in begining of the peer names. It is also backward\ncompatible with all running amphorae, except for those starting with\nhyphen but are broken anyways.\n\nThis patch takes option d). It substitutes hyphen with \u0027x\u0027 character.\n\n[1] https://github.com/haproxy/haproxy/issues/644\n\nTask: 39850\nStory: 2007714\n\nChange-Id: Ib0fc26877710dea423a5ebcf1f71077665404377\n(cherry picked from commit acc38391dea12a7f70142077250d15a4eb53cb87)\n(cherry picked from commit df36c2c8ca54e9b8f63bb3c0231e85f1b87b35b3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1975bd1d2926a1e30fe2e21dbf242b8d26e19d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1975bd1d2926a1e30fe2e21dbf242b8d26e19d50"}]},"branch":"refs/heads/stable/train"},"e0b53b2cc7a977d903328e43bac68c0b378439f5":{"kind":"REWORK","_number":2,"created":"2020-06-02 08:59:24.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/30/732430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/732430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/732430/2"}}},"commit":{"parents":[{"commit":"09e863eb539e29c8950ddfce12d602c0e77c1b7e","subject":"Fix getting user neutron client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09e863eb539e29c8950ddfce12d602c0e77c1b7e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-25 18:47:34.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-02 08:59:05.000000000","tz":120},"subject":"Workaround peer name starting with hyphen","message":"Workaround peer name starting with hyphen\n\nThe base64_sha_string method is used to set a base64-encoded peer name\nin HAProxy. There are cases where the peer name can start with\nan hypen which is troublesome when used in HAProxy CLI. Specifically,\nHAProxy fails to reload when local peer name starts with \u0027-x\u0027 [1]. When\nthis is the case, an amphora goes to provisioning status ERROR and later\nis scheduled for failover by the Octavia Health Manager service. A new\namphora UUUID is assigned and base64 encoded, hopefully not starting\nwith \u0027-x\u0027 again. However, this is far from being ideal -- we incur in a\ndataplane disruption (single topology) or reduce HA capabilities\n(active-standby topology) for some time.\n\nFour possible options:\n\na) add prefix to peer name\nb) change b64encode altchars\nc) quote peer name in haproxy CLI command\nd) substitute first character if hyphen\n\nOption a) and b) are not backward compatible with running amphorae. Peer\nnames of existing amphorae that do not start with hypen but contain\nhyphen at any other position would get different peer names.\n\nOption c) would nonetheless still require an amphora image update to add\nquotes in the HAProxy init service file. Continuing to generate peer\nnames with hyphens at begininng of the string is avoidable and\nrecommended.\n\nOption d), while also requiring an amphora image update, it would get\nrid of hyphens in begining of the peer names. It is also backward\ncompatible with all running amphorae, except for those starting with\nhyphen but are broken anyways.\n\nThis patch takes option d). It substitutes hyphen with \u0027x\u0027 character.\n\n[1] https://github.com/haproxy/haproxy/issues/644\n\nTask: 39850\nStory: 2007714\n\nChange-Id: Ib0fc26877710dea423a5ebcf1f71077665404377\n(cherry picked from commit acc38391dea12a7f70142077250d15a4eb53cb87)\n(cherry picked from commit df36c2c8ca54e9b8f63bb3c0231e85f1b87b35b3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0b53b2cc7a977d903328e43bac68c0b378439f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0b53b2cc7a977d903328e43bac68c0b378439f5"}]},"branch":"refs/heads/stable/train"},"b35f77d3b42997e882036e4dc9359b1afb78ffac":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-08 06:03:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/30/732430/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/732430/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/732430/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/732430/3"}}},"commit":{"parents":[{"commit":"4a95b354901c81f135b61d39754e1419dd134f0e","subject":"Do not install diskimage-builder from Git","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a95b354901c81f135b61d39754e1419dd134f0e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-25 18:47:34.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-08 06:03:56.000000000","tz":0},"subject":"Workaround peer name starting with hyphen","message":"Workaround peer name starting with hyphen\n\nThe base64_sha_string method is used to set a base64-encoded peer name\nin HAProxy. There are cases where the peer name can start with\nan hypen which is troublesome when used in HAProxy CLI. Specifically,\nHAProxy fails to reload when local peer name starts with \u0027-x\u0027 [1]. When\nthis is the case, an amphora goes to provisioning status ERROR and later\nis scheduled for failover by the Octavia Health Manager service. A new\namphora UUUID is assigned and base64 encoded, hopefully not starting\nwith \u0027-x\u0027 again. However, this is far from being ideal -- we incur in a\ndataplane disruption (single topology) or reduce HA capabilities\n(active-standby topology) for some time.\n\nFour possible options:\n\na) add prefix to peer name\nb) change b64encode altchars\nc) quote peer name in haproxy CLI command\nd) substitute first character if hyphen\n\nOption a) and b) are not backward compatible with running amphorae. Peer\nnames of existing amphorae that do not start with hypen but contain\nhyphen at any other position would get different peer names.\n\nOption c) would nonetheless still require an amphora image update to add\nquotes in the HAProxy init service file. Continuing to generate peer\nnames with hyphens at begininng of the string is avoidable and\nrecommended.\n\nOption d), while also requiring an amphora image update, it would get\nrid of hyphens in begining of the peer names. It is also backward\ncompatible with all running amphorae, except for those starting with\nhyphen but are broken anyways.\n\nThis patch takes option d). It substitutes hyphen with \u0027x\u0027 character.\n\n[1] https://github.com/haproxy/haproxy/issues/644\n\nTask: 39850\nStory: 2007714\n\nChange-Id: Ib0fc26877710dea423a5ebcf1f71077665404377\n(cherry picked from commit acc38391dea12a7f70142077250d15a4eb53cb87)\n(cherry picked from commit df36c2c8ca54e9b8f63bb3c0231e85f1b87b35b3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b35f77d3b42997e882036e4dc9359b1afb78ffac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b35f77d3b42997e882036e4dc9359b1afb78ffac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
