)]}'
{"id":"openstack%2Foctavia~732433","triplet_id":"openstack%2Foctavia~stable%2Frocky~Ib0fc26877710dea423a5ebcf1f71077665404377","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Ib0fc26877710dea423a5ebcf1f71077665404377","subject":"Workaround peer name starting with hyphen","status":"ABANDONED","created":"2020-06-01 18:05:23.000000000","updated":"2020-07-24 11:41:59.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f731410db73648d95714c4a6ca3f17dd5ead8555","_number":732433,"virtual_id_number":732433,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-02 10:44:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 10:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37186ca801c0ee7effd0c90e50301075848d8a20","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-01 18:05:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb6f786ea04000d740f537ff06cf9523de0b6db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 20:05:03.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/595678714a7a4044a976f0ee3faf7404 : FAILURE in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080c161c55f64128bff837179703dc54 : FAILURE in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9f17429800c4795aab4e6ff56181d4e : FAILURE in 7m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/650bbb6cbd4b4219acd050ea8bcbe5dd : FAILURE in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cbea605ab364df4853c9ed57ef0196a : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a21a9c2eb35c42c1b28c26401413911d : SUCCESS in 5m 27s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6e51315823324c95b555944e18315ed7 : SUCCESS in 1h 16m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/05b3514c1848486d81ef40bd828f2b8b : SUCCESS in 1h 09m 31s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f465560bb9648b482805bfb92139f01 : FAILURE in 1h 23m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/574b7294f4fe49b49dfa979b672a941f : FAILURE in 1h 00m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cfbc9bc8cdd249b3b59e1b8c85ce3256 : FAILURE in 8m 39s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/45c4a3ac993f47a6934072d74315a805 : FAILURE in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d872fbd555924b709efab74618e6a52e : FAILURE in 7m 32s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5af55b6915849c89ebec680d80f4d29 : FAILURE in 53m 23s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a920cb77a43a47d889176f6f6bdb852b : FAILURE in 59m 55s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f34e0ee341254db4b15543c5d77afeac : FAILURE in 1h 11m 31s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/53505d8e4faa48239e54597a6e562479 : FAILURE in 1h 22m 33s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/3b0b7a6d57cb4deca669d8ae5e649d0a : FAILURE in 29m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2909c4e5d1e54e67a0a59d9139b720ce : SUCCESS in 1h 07m 47s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/6d8ed1b6fbad4222b5ffbaac88336322 : SUCCESS in 1h 07m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80a544dce1f6424392a6fc7a955a46ab : FAILURE in 1h 53m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/507fc412c7354f459adebd8e45e30459 : FAILURE in 1h 09m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8d0c04a7f6d44ee7b210f3004689375b : SUCCESS in 1h 08m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77d5bccbb7c646639b5d37d4a00ace37 : FAILURE in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"43bac8ee73e3d5e8097ebe83868edf88324587fe","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-02 09:01:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7cedfcda6ce8ea06951d119510e44d5da6f214df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:44:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ffe3821c6904adfb14d38a4e6b34d34 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82dd1d202e640489a0edc629d30aca4 : SUCCESS in 8m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/636f6fbd3af3428988f7cdac7f32bd03 : SUCCESS in 6m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/472a2360391d4f9092f622b614937a4f : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8a80a98a7a4a979a0b9ca1db4aec5d : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/793b1a93e5954ad1a5240d49d104ed05 : SUCCESS in 5m 08s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ef0f6f3526bb499cbeb5ae61e4a90622 : SUCCESS in 1h 17m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/57d7de3339724e48af04a894b2f7189c : SUCCESS in 1h 17m 46s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8c34910e2bc4cb09172f1842868c3fa : SUCCESS in 1h 20m 10s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6a786ca8b57e40afa57dc26539c8c8d0 : SUCCESS in 1h 36m 37s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fdc27d3cd3e94408baf28345001a5cea : SUCCESS in 7m 33s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fa644f820d5f49cea0403e40d01a6fb6 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/31b236d1ddd84988aa731c3f4f4036aa : SUCCESS in 7m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9568941752e245b9b188ecba3d542aa9 : FAILURE in 30m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/90dba15760a14853be723df241142906 : FAILURE in 32m 36s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/31c9ed2a9a434cda99b756f65d0b89d8 : FAILURE in 33m 02s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/641b932c8b4e4afaad9440c3b51eb457 : FAILURE in 33m 56s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8dd9e0f59a5e4dca8946501087bb968c : FAILURE in 30m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0909534f9538490ba9319d4aee4d3585 : SUCCESS in 32m 40s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/986c016a57bc4ee6a85308d74fe62c16 : SUCCESS in 50m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2268e6f821aa47c79ab86093fee1341d : FAILURE in 31m 26s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3f6519aad779430ba3bc034c76b77627 : FAILURE in 18m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4354342086df4456a7b649907853ccb2 : SUCCESS in 1h 22m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/597dd132fe344cef97681de63020f93e : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f731410db73648d95714c4a6ca3f17dd5ead8555","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:41:59.000000000","message":"Abandoned\n\nRocky Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1eb5de99c624dfd9e2900b192b74b816bd17d1e","revisions":{"779588b1a57a59dfb08f02ab29622df53b05ebd8":{"kind":"REWORK","_number":1,"created":"2020-06-01 18:05:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/33/732433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/732433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/732433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/732433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/732433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/732433/1"}}},"commit":{"parents":[{"commit":"126ad303ac998f2fa36de572bb37cb36f366913a","subject":"Merge \"Fix the amphora noop driver\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/126ad303ac998f2fa36de572bb37cb36f366913a"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-25 18:47:34.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-01 18:05:14.000000000","tz":120},"subject":"Workaround peer name starting with hyphen","message":"Workaround peer name starting with hyphen\n\nThe base64_sha_string method is used to set a base64-encoded peer name\nin HAProxy. There are cases where the peer name can start with\nan hypen which is troublesome when used in HAProxy CLI. Specifically,\nHAProxy fails to reload when local peer name starts with \u0027-x\u0027 [1]. When\nthis is the case, an amphora goes to provisioning status ERROR and later\nis scheduled for failover by the Octavia Health Manager service. A new\namphora UUUID is assigned and base64 encoded, hopefully not starting\nwith \u0027-x\u0027 again. However, this is far from being ideal -- we incur in a\ndataplane disruption (single topology) or reduce HA capabilities\n(active-standby topology) for some time.\n\nFour possible options:\n\na) add prefix to peer name\nb) change b64encode altchars\nc) quote peer name in haproxy CLI command\nd) substitute first character if hyphen\n\nOption a) and b) are not backward compatible with running amphorae. Peer\nnames of existing amphorae that do not start with hypen but contain\nhyphen at any other position would get different peer names.\n\nOption c) would nonetheless still require an amphora image update to add\nquotes in the HAProxy init service file. Continuing to generate peer\nnames with hyphens at begininng of the string is avoidable and\nrecommended.\n\nOption d), while also requiring an amphora image update, it would get\nrid of hyphens in begining of the peer names. It is also backward\ncompatible with all running amphorae, except for those starting with\nhyphen but are broken anyways.\n\nThis patch takes option d). It substitutes hyphen with \u0027x\u0027 character.\n\n[1] https://github.com/haproxy/haproxy/issues/644\n\nTask: 39850\nStory: 2007714\n\nChange-Id: Ib0fc26877710dea423a5ebcf1f71077665404377\n(cherry picked from commit acc38391dea12a7f70142077250d15a4eb53cb87)\n(cherry picked from commit df36c2c8ca54e9b8f63bb3c0231e85f1b87b35b3)\n(cherry picked from commit 1975bd1d2926a1e30fe2e21dbf242b8d26e19d50)\n(cherry picked from commit 331393a167c829016fb31df971b06961b499914a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/779588b1a57a59dfb08f02ab29622df53b05ebd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/779588b1a57a59dfb08f02ab29622df53b05ebd8"}]},"branch":"refs/heads/stable/rocky"},"b1eb5de99c624dfd9e2900b192b74b816bd17d1e":{"kind":"REWORK","_number":2,"created":"2020-06-02 09:01:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/33/732433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/732433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/732433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/732433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/732433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/732433/2"}}},"commit":{"parents":[{"commit":"126ad303ac998f2fa36de572bb37cb36f366913a","subject":"Merge \"Fix the amphora noop driver\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/126ad303ac998f2fa36de572bb37cb36f366913a"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-25 18:47:34.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-02 09:01:14.000000000","tz":120},"subject":"Workaround peer name starting with hyphen","message":"Workaround peer name starting with hyphen\n\nThe base64_sha_string method is used to set a base64-encoded peer name\nin HAProxy. There are cases where the peer name can start with\nan hypen which is troublesome when used in HAProxy CLI. Specifically,\nHAProxy fails to reload when local peer name starts with \u0027-x\u0027 [1]. When\nthis is the case, an amphora goes to provisioning status ERROR and later\nis scheduled for failover by the Octavia Health Manager service. A new\namphora UUUID is assigned and base64 encoded, hopefully not starting\nwith \u0027-x\u0027 again. However, this is far from being ideal -- we incur in a\ndataplane disruption (single topology) or reduce HA capabilities\n(active-standby topology) for some time.\n\nFour possible options:\n\na) add prefix to peer name\nb) change b64encode altchars\nc) quote peer name in haproxy CLI command\nd) substitute first character if hyphen\n\nOption a) and b) are not backward compatible with running amphorae. Peer\nnames of existing amphorae that do not start with hypen but contain\nhyphen at any other position would get different peer names.\n\nOption c) would nonetheless still require an amphora image update to add\nquotes in the HAProxy init service file. Continuing to generate peer\nnames with hyphens at begininng of the string is avoidable and\nrecommended.\n\nOption d), while also requiring an amphora image update, it would get\nrid of hyphens in begining of the peer names. It is also backward\ncompatible with all running amphorae, except for those starting with\nhyphen but are broken anyways.\n\nThis patch takes option d). It substitutes hyphen with \u0027x\u0027 character.\n\n[1] https://github.com/haproxy/haproxy/issues/644\n\nTask: 39850\nStory: 2007714\n\nChange-Id: Ib0fc26877710dea423a5ebcf1f71077665404377\n(cherry picked from commit acc38391dea12a7f70142077250d15a4eb53cb87)\n(cherry picked from commit df36c2c8ca54e9b8f63bb3c0231e85f1b87b35b3)\n(cherry picked from commit e0b53b2cc7a977d903328e43bac68c0b378439f5)\n(cherry picked from commit b64dcdc38fd76b0bbad23b7b9663425ec6ac0891)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1eb5de99c624dfd9e2900b192b74b816bd17d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1eb5de99c624dfd9e2900b192b74b816bd17d1e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
