)]}'
{"id":"openstack%2Foctavia~732434","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Ib0fc26877710dea423a5ebcf1f71077665404377","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"Ib0fc26877710dea423a5ebcf1f71077665404377","subject":"Workaround peer name starting with hyphen","status":"ABANDONED","created":"2020-06-01 18:06:45.000000000","updated":"2020-07-24 11:41:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6616f347ffa9fa3d4a9124f6755ab16a8774cf2e","_number":732434,"virtual_id_number":732434,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-02 10:46:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 10:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed8b6ca37e15e72c5e0e79f67239a7d16cd50db2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-01 18:06:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a67a183270a20e3541c80d9b011c3ddc8e9a38a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 19:44:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/253d0a36d76f4689ad7402e499b373ca : FAILURE in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f13fb9c2c12245c0b805b6a02e9f67a4 : FAILURE in 5m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b7db254c67a343959d51d30b824d2c40 : FAILURE in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f7ef3ba6fad4ac8bf09220e54cbc36a : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b36013a968094f3ba1ee2b2d79c352dd : SUCCESS in 5m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/496798423ddf4ebeb26b1efacef12753 : SUCCESS in 1h 19m 23s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2b82e6439d594c9298a264561767f17a : SUCCESS in 1h 18m 23s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/637d2fd1c8e742d6bbef33253e056ec3 : FAILURE in 1h 31m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/480ed550a86e4ec3af2000e0794d0a1e : FAILURE in 1h 27m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1084a3de24bf4ff29971e04224d10bfa : FAILURE in 5m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/124df7882e1242e3a1821a728c6dc493 : FAILURE in 6m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/063ac1f0629949cf94abfe739f14a6ba : FAILURE in 6m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd81fc1cccd84c13b99ccb2f3a349812 : FAILURE in 57m 41s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a493e729dd9a4107beb1e7c5d140c2bc : FAILURE in 1h 05m 46s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/c946534c14ba4bb2a41763da459e08a1 : FAILURE in 1h 05m 24s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5be4f1ed29a34e3dba585545c77138b2 : FAILURE in 1h 18m 37s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/2b6eba4ea0634c2d9098c7c39b3444d5 : FAILURE in 37m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c005112d7f5543d88c528d2167f0a3a1 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/faceb845b6c54c52b32cd08249c1b728 : SUCCESS in 41m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b32116e574d4500a4ae94c0bf67802d : FAILURE in 1h 12m 04s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/383ef53e454e46a4815e98240b177742 : FAILURE in 1h 17m 36s","accounts_in_message":[],"_revision_number":1},{"id":"feed907024930b5f2aafe7aaf0a407555b91200c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-02 09:02:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6030eaba7257c5fd53051ded85d37a096c464cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:46:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445b1636a887480d90e172a17e92d316 : SUCCESS in 8m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2123fcca9c5f48c599b965e1581ad013 : SUCCESS in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b4f46f0bc38548e3beca534bc22eafd9 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4c5743594b4a459aa43edc0b3b8898 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d915cb570b8942be93a093468d52cfa7 : SUCCESS in 6m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1923995e6bd44a07976e0557ad32b325 : SUCCESS in 1h 01m 28s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d3c1eb8e1f49403e950c55f2d21f8fe9 : SUCCESS in 1h 01m 18s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27f1a1d9a5974a048042d8cea4031e1a : SUCCESS in 1h 17m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3ac2327b377440acaf4baabdb7525b85 : FAILURE in 1h 18m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0db81601b394e88a29b07f504f259d2 : SUCCESS in 10m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4d610d77f18d441a8d4155412d0939aa : SUCCESS in 7m 00s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/094a1131028c4d46a39a7b2172bb69fb : SUCCESS in 7m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44836b0a21844821a8fa475b8382dc9f : SUCCESS in 54m 15s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ad734874d7fc4d3a8ac55f1566c7859f : SUCCESS in 54m 24s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/50641d8c7033428684d3b35888ac5265 : SUCCESS in 1h 17m 04s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5f0b31d7272847b38c4b29e7adebb70a : FAILURE in 44m 08s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/10f067d5f2b24c689d18c513490e9617 : FAILURE in 29m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/131a744b4a434e289f04c75226fe924b : SUCCESS in 45m 47s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/af8277e2f0d5405783b990e7e297a4f6 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5aea0360e5434d798c117de7999e2b92 : FAILURE in 1h 38m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/8506342e428440d2a90c6533529b6942 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":2},{"id":"6616f347ffa9fa3d4a9124f6755ab16a8774cf2e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:41:11.000000000","message":"Abandoned\n\nQueens Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec6ee78067fb6d6b6c2863e9b71cb9afaa8dc8f2","revisions":{"9f2a011f7c7ad0a074aedfab04006cc231a5f83d":{"kind":"REWORK","_number":1,"created":"2020-06-01 18:06:45.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/34/732434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/732434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/732434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/732434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/732434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/732434/1"}}},"commit":{"parents":[{"commit":"299c54a5dee5898eb0712123b98670f4043bdec6","subject":"Merge \"Fix the amphora noop driver\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/299c54a5dee5898eb0712123b98670f4043bdec6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-25 18:47:34.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-01 18:06:31.000000000","tz":120},"subject":"Workaround peer name starting with hyphen","message":"Workaround peer name starting with hyphen\n\nThe base64_sha_string method is used to set a base64-encoded peer name\nin HAProxy. There are cases where the peer name can start with\nan hypen which is troublesome when used in HAProxy CLI. Specifically,\nHAProxy fails to reload when local peer name starts with \u0027-x\u0027 [1]. When\nthis is the case, an amphora goes to provisioning status ERROR and later\nis scheduled for failover by the Octavia Health Manager service. A new\namphora UUUID is assigned and base64 encoded, hopefully not starting\nwith \u0027-x\u0027 again. However, this is far from being ideal -- we incur in a\ndataplane disruption (single topology) or reduce HA capabilities\n(active-standby topology) for some time.\n\nFour possible options:\n\na) add prefix to peer name\nb) change b64encode altchars\nc) quote peer name in haproxy CLI command\nd) substitute first character if hyphen\n\nOption a) and b) are not backward compatible with running amphorae. Peer\nnames of existing amphorae that do not start with hypen but contain\nhyphen at any other position would get different peer names.\n\nOption c) would nonetheless still require an amphora image update to add\nquotes in the HAProxy init service file. Continuing to generate peer\nnames with hyphens at begininng of the string is avoidable and\nrecommended.\n\nOption d), while also requiring an amphora image update, it would get\nrid of hyphens in begining of the peer names. It is also backward\ncompatible with all running amphorae, except for those starting with\nhyphen but are broken anyways.\n\nThis patch takes option d). It substitutes hyphen with \u0027x\u0027 character.\n\n[1] https://github.com/haproxy/haproxy/issues/644\n\nTask: 39850\nStory: 2007714\n\nChange-Id: Ib0fc26877710dea423a5ebcf1f71077665404377\n(cherry picked from commit acc38391dea12a7f70142077250d15a4eb53cb87)\n(cherry picked from commit df36c2c8ca54e9b8f63bb3c0231e85f1b87b35b3)\n(cherry picked from commit 1975bd1d2926a1e30fe2e21dbf242b8d26e19d50)\n(cherry picked from commit 331393a167c829016fb31df971b06961b499914a)\n(cherry picked from commit 779588b1a57a59dfb08f02ab29622df53b05ebd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f2a011f7c7ad0a074aedfab04006cc231a5f83d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f2a011f7c7ad0a074aedfab04006cc231a5f83d"}]},"branch":"refs/heads/stable/queens"},"ec6ee78067fb6d6b6c2863e9b71cb9afaa8dc8f2":{"kind":"REWORK","_number":2,"created":"2020-06-02 09:02:02.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/34/732434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/732434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/732434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/732434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/732434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/732434/2"}}},"commit":{"parents":[{"commit":"299c54a5dee5898eb0712123b98670f4043bdec6","subject":"Merge \"Fix the amphora noop driver\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/299c54a5dee5898eb0712123b98670f4043bdec6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-25 18:47:34.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-02 09:01:52.000000000","tz":120},"subject":"Workaround peer name starting with hyphen","message":"Workaround peer name starting with hyphen\n\nThe base64_sha_string method is used to set a base64-encoded peer name\nin HAProxy. There are cases where the peer name can start with\nan hypen which is troublesome when used in HAProxy CLI. Specifically,\nHAProxy fails to reload when local peer name starts with \u0027-x\u0027 [1]. When\nthis is the case, an amphora goes to provisioning status ERROR and later\nis scheduled for failover by the Octavia Health Manager service. A new\namphora UUUID is assigned and base64 encoded, hopefully not starting\nwith \u0027-x\u0027 again. However, this is far from being ideal -- we incur in a\ndataplane disruption (single topology) or reduce HA capabilities\n(active-standby topology) for some time.\n\nFour possible options:\n\na) add prefix to peer name\nb) change b64encode altchars\nc) quote peer name in haproxy CLI command\nd) substitute first character if hyphen\n\nOption a) and b) are not backward compatible with running amphorae. Peer\nnames of existing amphorae that do not start with hypen but contain\nhyphen at any other position would get different peer names.\n\nOption c) would nonetheless still require an amphora image update to add\nquotes in the HAProxy init service file. Continuing to generate peer\nnames with hyphens at begininng of the string is avoidable and\nrecommended.\n\nOption d), while also requiring an amphora image update, it would get\nrid of hyphens in begining of the peer names. It is also backward\ncompatible with all running amphorae, except for those starting with\nhyphen but are broken anyways.\n\nThis patch takes option d). It substitutes hyphen with \u0027x\u0027 character.\n\n[1] https://github.com/haproxy/haproxy/issues/644\n\nTask: 39850\nStory: 2007714\n\nChange-Id: Ib0fc26877710dea423a5ebcf1f71077665404377\n(cherry picked from commit acc38391dea12a7f70142077250d15a4eb53cb87)\n(cherry picked from commit df36c2c8ca54e9b8f63bb3c0231e85f1b87b35b3)\n(cherry picked from commit e0b53b2cc7a977d903328e43bac68c0b378439f5)\n(cherry picked from commit b64dcdc38fd76b0bbad23b7b9663425ec6ac0891)\n(cherry picked from commit b1eb5de99c624dfd9e2900b192b74b816bd17d1e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec6ee78067fb6d6b6c2863e9b71cb9afaa8dc8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec6ee78067fb6d6b6c2863e9b71cb9afaa8dc8f2"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
