)]}'
{"id":"openstack%2Foctavia~732453","triplet_id":"openstack%2Foctavia~master~I925cc05981e677c0552b18f845fdbc512d2af22c","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I925cc05981e677c0552b18f845fdbc512d2af22c","subject":"Switch oslo.policy over to yaml","status":"MERGED","created":"2020-06-01 21:32:22.000000000","updated":"2020-06-11 22:13:03.000000000","submitted":"2020-06-11 22:11:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732453-1591913471193-3023ec62","meta_rev_id":"3daa0426d3ba029e2c322ce93ea5fcda902f8f84","_number":732453,"virtual_id_number":732453,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 22:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-11 19:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-08 15:54:46.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-08 18:12:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-09 01:09:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-06-11 19:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2020-06-08 15:54:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 15:54:46.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-08 18:12:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-09 01:09:01.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-06-11 19:09:51.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-11 22:11:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94c5ab651d26d1c6138c155431f5fa51b55863f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 21:32:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"553f62b901b06f6b8ade572bb594cf90ea5e7917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 23:29:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b33deec03ad74f808176d06a7031c800 : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/652a5d49193147f28afbd8c0d79fc5d1 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d02d36f4f7a49e7bdba19b1649c5b06 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7ba6f2789745c8abb15f300e253ea9 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba49fb0ac0bd4e8d95b1c3c6c24684a4 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6e1c2488a6400098ee79bdad2ec746 : SUCCESS in 13m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cae096106b39406c95330dd6aeafa218 : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6f84e1c9fb6c49a1832b3ea501aefb8e : SUCCESS in 9m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e049d33a7f384a8dbc3550b0b0764f1e : SUCCESS in 39m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/802bf0a0e25040678280355b934ddc73 : SUCCESS in 51m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39ab3fc3145941afb4946f98fbf36796 : SUCCESS in 32m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c17714d63f8d4e359fe0bb7510de1c35 : SUCCESS in 56m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/808ae6bc7cb842af833924eea1962b64 : SUCCESS in 59m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/90c275967bce4bc19eb4ddfbfcaa58b2 : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/753b1a3de66f401380e6df8c0a3589f2 : SUCCESS in 1h 19m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a55a663a20304d1797579284ed2a8771 : SUCCESS in 1h 52m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f172e9f142494596a52b74233779e8ae : SUCCESS in 1h 07m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7525d6ab4ea9cd79f6afee5add86de49d7150450","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 19:10:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4cbf74c5b05928c852a92154b3dcdb37d39db77e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-06 18:39:20.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review-1\n\nThere are a few more references to file policy.json that should be taken care of. One in particular is in devstack/plugin.sh where devstack would fail to copy policy.json when OCTAVIA_USE_LEGACY_RBAC\u003dTrue.\n\ndevstack/plugin.sh:        cp $OCTAVIA_DIR/etc/policy/admin_or_owner-policy.json $OCTAVIA_CONF_DIR/policy.json\nspecs/template.rst:  policy.json file.\ntools/misc-sanity-checks.sh:    # For unit tests, we maintain their own policy.json file to make test suite\ntools/misc-sanity-checks.sh:    diff etc/policy.json octavia/tests/etc/policy.json 2\u003e\u00261 \u003e /dev/null\ntools/misc-sanity-checks.sh:        echo \"policy.json files must be identical!\" \u003e\u003e$FAILURES","accounts_in_message":[],"_revision_number":1},{"id":"bfb67a67cf6a6c1dcda9764c0898036db27b9477","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-08 15:54:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c76c0d0e30a6affc30ca186e8cb99af6e33226d1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-08 15:54:46.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"33cdf8bc9483ca05532e9b717d895a762f823d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:07:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b56a2dd495d048c6a31b22793ca54196 : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a6a1d22488841d98301e50ea15aec54 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f8d357730b449309a9ebb0337c3728c : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d423ebbaa6f44769987bd1c1b098efb1 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8882568613ed41c7b0bbfb0803b501c0 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b829934760d24370b02201286aab70ea : SUCCESS in 14m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0e154306d9f9436cb867b1823d962258 : SUCCESS in 8m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b9ebac7c58ba458b8543b9b960e9d4ef : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a6061331f0145fe83479930f508ef5e : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4149dee4d2342938134405434cb513a : SUCCESS in 48m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2318db9edb2247c08fe061ca0d38bb9d : SUCCESS in 1h 53m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/00e6f70f4b79482a926cba984d24dc65 : SUCCESS in 1h 02m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/34880a42997a418c90c92c190899c6cf : SUCCESS in 55m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01994481f3b44150bd99f5ead690b2cb : SUCCESS in 1h 07m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0a38b9e439945689bbb8196f6363108 : SUCCESS in 1h 18m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/057a4ad352d846c79d64cc9e0d803d27 : SUCCESS in 54m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/336b68ab3e7c4826ba2797c2f37d5921 : SUCCESS in 1h 49m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ed271cd724d14c9f824d0f7b8673bd8e : SUCCESS in 2h 07m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4af9d868be927ba8cfb0af46a1a2c9a18dc67a04","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-08 18:12:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5d6bb3e7c5ad46e47626c872a9a5ea73f342e2b4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-06-09 01:09:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7bfb5c6ac209c7a563924170a2fa2a404532c077","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-11 19:09:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nObligatory \"JSON *is* YAML\". :D","accounts_in_message":[],"_revision_number":2},{"id":"a5523219fe47c157351b5a222be4c4999049854c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:10:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c0cc6671fe1fd94be1ed33e02368ac90594a4a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:11:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/166236c013a74188b1e992a3d27f4309 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89357bce07a34cc6959fe75ce21be5a8 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67ba39b546f643df903d009e2462d3fe : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b613e0c74443abb54e47a27aabd8df : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f8782f4c4d4149b8e7ec3e21bbba12 : SUCCESS in 14m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e853fe10a69d422e92e19d4bbf1ecede : SUCCESS in 9m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0aba233d8c94a3ab97b77584b571f53 : SUCCESS in 46m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7529026384be462fb179de2ba99c2c18 : SUCCESS in 1h 40m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bfbf862703db44ad953e4a44183f1b43 : SUCCESS in 57m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2b0c4cd1adcd4ce9b510ed322960dce2 : SUCCESS in 52m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0776e6758c594b1f9cdd7f01d7066008 : SUCCESS in 1h 03m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a079b0a0eb714e088b1ba7f22ff44118 : SUCCESS in 1h 32m 51s","accounts_in_message":[],"_revision_number":2},{"id":"198a689197b2d11e6529b994ec6e637372bc6bb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:11:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3daa0426d3ba029e2c322ce93ea5fcda902f8f84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:13:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce7d7bfbcd243eeb351549de67163b0 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fafabad0422462e713da84b2467ba791b6af83dc","revisions":{"e19d8d2b82bed5db51030796f839f837da6ff4f2":{"kind":"REWORK","_number":1,"created":"2020-06-01 21:32:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/732453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/732453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/732453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/732453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/732453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/732453/1"}}},"commit":{"parents":[{"commit":"2c7c7747b74e866e5ab41139e2f3f01d9dce15a5","subject":"Merge \"Workaround peer name starting with hyphen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c7c7747b74e866e5ab41139e2f3f01d9dce15a5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-01 21:28:38.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-01 21:32:21.000000000","tz":-420},"subject":"Switch oslo.policy over to yaml","message":"Switch oslo.policy over to yaml\n\nOslo.policy is moving away from using json format policy files[1].\n\nThis patch updates the Octavia documentation, policy configuration file, and\nlegacy admin-or-owner policy file to be in yaml format.\n\nOctavia will continue to honor and support the json format file as long\nas oslo.policy does, but this patch will encourage new deployments\nto use the yaml format.\n\n[1] https://docs.openstack.org/oslo.policy/latest/admin/policy-json-file.html\n\nChange-Id: I925cc05981e677c0552b18f845fdbc512d2af22c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19d8d2b82bed5db51030796f839f837da6ff4f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19d8d2b82bed5db51030796f839f837da6ff4f2"}]},"branch":"refs/heads/master"},"fafabad0422462e713da84b2467ba791b6af83dc":{"kind":"REWORK","_number":2,"created":"2020-06-08 15:54:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/732453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/732453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/732453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/732453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/732453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/732453/2"}}},"commit":{"parents":[{"commit":"2c7c7747b74e866e5ab41139e2f3f01d9dce15a5","subject":"Merge \"Workaround peer name starting with hyphen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c7c7747b74e866e5ab41139e2f3f01d9dce15a5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-01 21:28:38.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-08 15:54:07.000000000","tz":-420},"subject":"Switch oslo.policy over to yaml","message":"Switch oslo.policy over to yaml\n\nOslo.policy is moving away from using json format policy files[1].\n\nThis patch updates the Octavia documentation, policy configuration file, and\nlegacy admin-or-owner policy file to be in yaml format.\n\nOctavia will continue to honor and support the json format file as long\nas oslo.policy does, but this patch will encourage new deployments\nto use the yaml format.\n\n[1] https://docs.openstack.org/oslo.policy/latest/admin/policy-json-file.html\n\nChange-Id: I925cc05981e677c0552b18f845fdbc512d2af22c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fafabad0422462e713da84b2467ba791b6af83dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fafabad0422462e713da84b2467ba791b6af83dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
