)]}'
{"id":"openstack%2Foctavia~733032","triplet_id":"openstack%2Foctavia~master~Ia7e8ee8ab5aed94352657becb3c3ce3d6d5f1941","project":"openstack/octavia","branch":"master","topic":"misspelled-quota","hashtags":[],"change_id":"Ia7e8ee8ab5aed94352657becb3c3ce3d6d5f1941","subject":"Remove deprecated misspelled quota fields","status":"ABANDONED","created":"2020-06-03 01:13:47.000000000","updated":"2021-07-28 21:04:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"68c1cb5d0489104312bc5688a16c6cc0e0aa25a5","_number":733032,"virtual_id_number":733032,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-14 21:31:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 21:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 07:55:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"b2187e6b9465dab208cd9e27437d0e5cfc4be49c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-03 01:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f0272de2b702145396657b068e0ef3b9b25a4cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 02:47:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4024a5480cd145babb43319c237f1943 : FAILURE in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ae8cf5b4343449987103a757a0b3f89 : FAILURE in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718ae08d85eb42acbdbcfcc604eac106 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cf705f7abdf4fa9b433849bb1a0354a : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b54f64c2eb204037a8af701bd2e23bee : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efbb5bc470548c188cf8766c219bd9f : SUCCESS in 11m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/08a899eafd8d439595fbc791c09e5632 : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/31b25408c769492db95177db890e1bb6 : FAILURE in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/803fab38f0f444428c1a8fab5cb6b999 : FAILURE in 9m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/107cbb863bd14232a68e595de7b9e978 : SUCCESS in 39m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d16a81fd13de438ea4677626d8f2ddbe : SUCCESS in 1h 15m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bdcf5cc5cebf4d3099db9973fa8d7ab4 : SUCCESS in 40m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/34355cf4ccd14c85afadb6c1ec376e52 : SUCCESS in 49m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9cdd0aba75e4161a32dcdd1029fbc31 : SUCCESS in 1h 05m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d1b84e8dfcd496d94a65e68054b2506 : SUCCESS in 1h 33m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb39bf8c8d5c4c648c33a16890033e24 : SUCCESS in 51m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aebc9638fe204c63ba8ea1124496ceb7 : SUCCESS in 1h 26m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a14b87958564dc7815d7c9ee03e8230 : SUCCESS in 1h 29m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"309e3a22d9f7a98467a9427627ae2785d8f3c187","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 19:08:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5cf65f7ba2871f5295c9dfb206cae8075109197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:37:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef1d48ce5f774c13aef7c66de6a9c530 : FAILURE in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86ecbf7432ce442d96dee2d7bd4d2019 : FAILURE in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82dc477b5284c39b7718aa7750ba400 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a7ac947939e48eeb420f8b791e83337 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cddb04155b4c429ea2017943bbc3be34 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61e4ca2127044e58aa482efc414c588e : SUCCESS in 12m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f471733ee4bf4927bdf17c634ba54ce1 : SUCCESS in 6m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3617bfdfa7ec4e91bc2f5cb66571a49b : FAILURE in 9m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7458e016237d4b82bfa9f9acce6c1736 : FAILURE in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dd683db5d7634902a09ad96a54af04a6 : SUCCESS in 48m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a43d51fc9b84016a96aeba5243c159d : SUCCESS in 1h 46m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eda3252faaee4bec84a40489a183961b : SUCCESS in 1h 01m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f769ac0ff8d540c4a2b176b11c1c70d6 : SUCCESS in 50m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5b35fe8254a44e58870cb81d8a3c494 : SUCCESS in 1h 01m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a00a76d457824206b2886e4c051ab693 : SUCCESS in 1h 34m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8fc246ef763f4f248b256cc1284ea929 : SUCCESS in 1h 55m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f280f0fcdfc1446e9777d895e72177e4 : SUCCESS in 2h 24m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/182733b03af14fb8a657d3d9c98dea30 : FAILURE in 1h 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7766b235396ab5dca2bb6a0bc48d1be1ae7d331e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-10 19:31:59.000000000","message":"Patch Set 2:\n\nFor my own sanity, I found part of the problem here.  openstacksdk also defines what the Octavia quota dict should be, and it also uses the old names.  I\u0027m going to propose a patch to that to fix things there, since the tests in this patch will fail without it.","accounts_in_message":[],"_revision_number":2},{"id":"06059faf3c1be00a8ca5d687ab18a3af64e1c344","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 18:35:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e130f56d458a2e8038b7b720c66c391abda201ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 21:31:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6311ef27a1094333b92b91a4da2db172 : FAILURE in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5586c5d77664e6a99825fe3ce574873 : FAILURE in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5e191cf2c34449864fdb01857988db : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/252037d937b14829b55abd46339eb9f5 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e12321b10c44ac6855c05ddd0a89bd2 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4334fcfd6340a986f31ff7ecfeffe3 : SUCCESS in 11m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6f3e9c24a7594f68879c2a136bca5d68 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ac39f7234b6f4cc3a560f3d2d6482f19 : FAILURE in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a7eb8b607b0340b2956570256945dfa1 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2488fc25141c42a7b46e2e41ba417d0a : FAILURE in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1ac4c75425f4dbf8499324c4bc5dfdf : SUCCESS in 1h 35m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0dfad7780ed4b87b0d11e3c524c8e18 : SUCCESS in 1h 44m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/859956cfe3084179b2b4f4312121a337 : SUCCESS in 48m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a8ff100bbfb34694896a7b934f00440f : SUCCESS in 1h 31m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74d927add1314cdb846c10f7d31a8b84 : SUCCESS in 46m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/efe6f05b54ed48c996b79d08e3502e44 : SUCCESS in 1h 05m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6d711c5ec924a88b3fbf9fdb91998cd : SUCCESS in 1h 56m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e10e009be3a64e30805b8d4acb523919 : FAILURE in 2h 20m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1cc6dded27464a77a91ba8a442081e33 : SUCCESS in 1h 54m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/feeeb143c2094227b5e080d0cf3593b1 : SUCCESS in 2h 20m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a33bb400c6cc4de5aa5302722396708f : TIMED_OUT in 2h 35m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ae46caf5eaf7480198571fbf755dfe16 : SUCCESS in 2h 09m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6545d9f92941237db38e65e13e4fff4751ed34d0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-18 07:55:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"899828f19088cc51fb176c6d9730054c364acb1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 09:58:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81dae91e7822485cbc9c45745b462f1a : FAILURE in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3364dd7c12514cd3a199687bfe86e3c9 : FAILURE in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb3eb037a4a4d3a9c2f5947f6ae9d77 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/023bd6507c65468ebe0ca15898836e55 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fca861c8d6d64a52b2f47fc5b9ae40d2 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90c2347363c14352ab409e8950bc5ddd : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc999228ec84dcb805650014b1362c6 : SUCCESS in 12m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/965baed2867b453e8ef9fd866bc5c96f : SUCCESS in 6m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c5fb5693145b40c383839e877ceb36a2 : FAILURE in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dd0edc07e9524bc5a994c006c242df07 : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c4ceca053c442c7bade82cd6d24d28f : FAILURE in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c5f518e64f449939b53f2f10944c40e : SUCCESS in 1h 15m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68c1cb5d0489104312bc5688a16c6cc0e0aa25a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-28 21:04:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f96135e1d9eac008bb52300f89415ca49c2664a3","revisions":{"fede020d6c7934f03fa050c070738d8e846932c7":{"kind":"REWORK","_number":1,"created":"2020-06-03 01:13:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/733032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/733032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/733032/1"}}},"commit":{"parents":[{"commit":"2c7c7747b74e866e5ab41139e2f3f01d9dce15a5","subject":"Merge \"Workaround peer name starting with hyphen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c7c7747b74e866e5ab41139e2f3f01d9dce15a5"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-03 01:11:30.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-03 01:11:30.000000000","tz":-240},"subject":"Remove deprecated misspelled quota fields","message":"Remove deprecated misspelled quota fields\n\nQuota fields load_balancer and health_monitor were misspelled\nand replaced with loadbalancer and healthmonitor, respectively.\nAs they were marked for deprecation in the Rocky to be removed\nin Train, they can be safely be removed now.\n\nChange-Id: Ia7e8ee8ab5aed94352657becb3c3ce3d6d5f1941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fede020d6c7934f03fa050c070738d8e846932c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fede020d6c7934f03fa050c070738d8e846932c7"}]},"branch":"refs/heads/master"},"8cbc1d5491fd0897ecb05172f1f681b1ef1a7e25":{"kind":"REWORK","_number":2,"created":"2020-06-05 19:08:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/733032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/733032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/733032/2"}}},"commit":{"parents":[{"commit":"2c7c7747b74e866e5ab41139e2f3f01d9dce15a5","subject":"Merge \"Workaround peer name starting with hyphen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c7c7747b74e866e5ab41139e2f3f01d9dce15a5"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-03 01:11:30.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-05 19:08:14.000000000","tz":-240},"subject":"Remove deprecated misspelled quota fields","message":"Remove deprecated misspelled quota fields\n\nQuota fields load_balancer and health_monitor were misspelled\nand replaced with loadbalancer and healthmonitor, respectively.\nAs they were marked for deprecation in the Rocky to be removed\nin Train, they can be safely be removed now.\n\nChange-Id: Ia7e8ee8ab5aed94352657becb3c3ce3d6d5f1941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cbc1d5491fd0897ecb05172f1f681b1ef1a7e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cbc1d5491fd0897ecb05172f1f681b1ef1a7e25"}]},"branch":"refs/heads/master"},"f96135e1d9eac008bb52300f89415ca49c2664a3":{"kind":"REWORK","_number":3,"created":"2020-10-14 18:35:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/733032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/733032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/733032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/733032/3"}}},"commit":{"parents":[{"commit":"8ea4d40281f4da7e5695128bc5a55994db0a3166","subject":"Merge \"Fix pool delete race on load balancer cascade delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ea4d40281f4da7e5695128bc5a55994db0a3166"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-03 01:11:30.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-14 18:25:20.000000000","tz":-240},"subject":"Remove deprecated misspelled quota fields","message":"Remove deprecated misspelled quota fields\n\nQuota fields load_balancer and health_monitor were misspelled\nand replaced with loadbalancer and healthmonitor, respectively.\nAs they were marked for deprecation in the Rocky to be removed\nin Train, they can be safely be removed now.\n\nChange-Id: Ia7e8ee8ab5aed94352657becb3c3ce3d6d5f1941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f96135e1d9eac008bb52300f89415ca49c2664a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f96135e1d9eac008bb52300f89415ca49c2664a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
