)]}'
{"id":"openstack%2Foctavia~733262","triplet_id":"openstack%2Foctavia~master~I699f95c89506de84074788b04cea32352d9d31b4","project":"openstack/octavia","branch":"master","topic":"supports-zero-downtime-upgrade","hashtags":[],"change_id":"I699f95c89506de84074788b04cea32352d9d31b4","subject":"WIP: zero downtime upgrade","status":"NEW","created":"2020-06-03 17:02:36.000000000","updated":"2020-06-05 20:28:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e7996fc2bd4cc104e71a66f079e610d6446c326","_number":733262,"virtual_id_number":733262,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-05 20:28:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 20:28:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69735a97bf92b7a603d28dae892b84912ee52fed","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-03 17:02:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a15eba4c744122e9ac0c058deaa1b1107fa0265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:02:52.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Job octavia-grenade-zero does not specify a run playbook","accounts_in_message":[],"_revision_number":1},{"id":"700b2c5bda060bd7e075fa1e79cfbea6a6dcae22","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-03 17:08:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e71d23ab423e09d7c21a2da9e33c94db5b2b6f8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:40:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2bc77817352495c8351c6df6b6935a8 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f89a24b1ad84356a562812bac35caee : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2bb7bfdc840466c9b569b2027574c0f : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38d838c189342b18a33e8d2e8d566ee : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdf662f9f2cb4efe9dc975adb99f21bc : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f5e9f40bda4ddca4725470f4697981 : SUCCESS in 13m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a179e4cd0d5e42a987e8c87cd4c2c5bc : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a850f73768f64295b8aa5c009d7d9fb9 : SUCCESS in 7m 34s\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/413c430a5cb449f5803b28fb2983682a : FAILURE in 1h 25m 59s","accounts_in_message":[],"_revision_number":2},{"id":"35ec7043977d06bde5a8521682250c5333e8f36b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-04 10:05:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"977d468e2354acd5b1c746901dca1379e4300db1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-04 10:31:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"61653dfcebbb2ed0c09ff5da8f833b7446f244ea","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-04 11:00:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dbf831178e57edc1b6b1a3c02596f308e3f35cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:56:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2686a69004dd45d986fda0cae0e21f21 : SUCCESS in 12m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c01a4724c50545e18499aacb2c7082af : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7498c5b5ce004e569ed8dd5d6a6c979c : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60f81d5d50a642f2bb9e050f7ab80613 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd310f5d90904367b6704bdb5be6072a : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0295f4571c46d9bddcb3a100d23915 : SUCCESS in 17m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd4667def3894087b9205b6803dce2d7 : SUCCESS in 6m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3e1fcf91dbe0489aad892b400b998864 : SUCCESS in 12m 32s\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/a72284e4e3334da9ab544596a85b59fa : RETRY_LIMIT in 14m 22s","accounts_in_message":[],"_revision_number":5},{"id":"a8c37570ac021fe074ad0f530f709752dba1c2e8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-04 12:48:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"950cff001081e4972ca99038bf15525456befe07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:48:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/24b983c27bd2444daef393bb90b3c30a : RETRY_LIMIT in 17m 50s","accounts_in_message":[],"_revision_number":6},{"id":"fe506b5f8167680b20cc061b07c218ec2d7cc1bc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-05 10:04:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bc127f49060fe155f461fc865492177f02c75d58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:04:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/eea21b5f9ed748ae9d9bb1674923c960 : FAILURE in 54m 51s","accounts_in_message":[],"_revision_number":7},{"id":"40ad4e8fda0ae125557d78f390cfff015cbd4d94","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-05 11:33:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c2beb1e4afb3841d228afd3431b0d808f0cf9d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 12:23:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/a7bdfeb62aee4ed9b610e23d96e629b8 : FAILURE in 47m 22s","accounts_in_message":[],"_revision_number":8},{"id":"b94b83a8991a2669a110203355f677aba62b7cdf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-05 12:41:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d3730f4875a92ce3b3c961561aabe67064a5b339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:49:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/8c4a7aea88da4e9ba877b021a941b39e : FAILURE in 54m 15s","accounts_in_message":[],"_revision_number":9},{"id":"dba84519176a0f0ab50cdb824209fd16cdc0d539","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-05 14:21:46.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"791eda6ccec2d59d44108e10527441b032a3d35e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 16:28:43.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/e2ebfa6052a041c1b24c532483b54015 : FAILURE in 1h 46m 32s","accounts_in_message":[],"_revision_number":10},{"id":"da0462ea5e743271c3a9796602a5260370467a66","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-05 17:15:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"aa8947b0d308fba0b8635b3ae56612a6536f9328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:28:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-grenade-zero https://zuul.opendev.org/t/openstack/build/60b7dd2a0a5648ada7bad619491bb2b5 : TIMED_OUT in 3h 03m 29s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d2fc8b2b03da8cbf10226cd621a6fde29e9e968b","revisions":{"83b754aa5c86559dbc4894ed9cc2858beb5179b5":{"kind":"REWORK","_number":1,"created":"2020-06-03 17:02:36.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/1"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83b754aa5c86559dbc4894ed9cc2858beb5179b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83b754aa5c86559dbc4894ed9cc2858beb5179b5"}]},"branch":"refs/heads/master"},"0dc0c49b30aedc06da8ccfaad436dda468e813d2":{"kind":"REWORK","_number":2,"created":"2020-06-03 17:08:31.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/2"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:08:21.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dc0c49b30aedc06da8ccfaad436dda468e813d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dc0c49b30aedc06da8ccfaad436dda468e813d2"}]},"branch":"refs/heads/master"},"d9ac2b1e71ff981b8494745599ac0986c306a29a":{"kind":"REWORK","_number":3,"created":"2020-06-04 10:05:02.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/3"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 10:04:46.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9ac2b1e71ff981b8494745599ac0986c306a29a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9ac2b1e71ff981b8494745599ac0986c306a29a"}]},"branch":"refs/heads/master"},"8d03ffcaf6a3875477217c8a59d019a72028bd52":{"kind":"REWORK","_number":4,"created":"2020-06-04 10:31:36.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/4"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 10:31:25.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d03ffcaf6a3875477217c8a59d019a72028bd52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d03ffcaf6a3875477217c8a59d019a72028bd52"}]},"branch":"refs/heads/master"},"d28bfb0a9fcdda0ff94dac339ccd7a8087020f21":{"kind":"REWORK","_number":5,"created":"2020-06-04 11:00:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/5"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 11:00:31.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d28bfb0a9fcdda0ff94dac339ccd7a8087020f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d28bfb0a9fcdda0ff94dac339ccd7a8087020f21"}]},"branch":"refs/heads/master"},"a16238e6526e31468bfb2a8c7bf32c26f144fb73":{"kind":"REWORK","_number":6,"created":"2020-06-04 12:48:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/6"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 12:48:21.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a16238e6526e31468bfb2a8c7bf32c26f144fb73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a16238e6526e31468bfb2a8c7bf32c26f144fb73"}]},"branch":"refs/heads/master"},"b252c059c1dd10795af3bbdb6d90076c8140707f":{"kind":"REWORK","_number":7,"created":"2020-06-05 10:04:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/7"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-05 10:03:53.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b252c059c1dd10795af3bbdb6d90076c8140707f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b252c059c1dd10795af3bbdb6d90076c8140707f"}]},"branch":"refs/heads/master"},"5f54d307798d87ad3e551e1f4a80ce2e904f9579":{"kind":"REWORK","_number":8,"created":"2020-06-05 11:33:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/8"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-05 11:33:28.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f54d307798d87ad3e551e1f4a80ce2e904f9579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f54d307798d87ad3e551e1f4a80ce2e904f9579"}]},"branch":"refs/heads/master"},"b66b019e556b21ef85e981b045042c0d2c11f57e":{"kind":"REWORK","_number":9,"created":"2020-06-05 12:41:14.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/9"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-05 12:41:00.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b66b019e556b21ef85e981b045042c0d2c11f57e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b66b019e556b21ef85e981b045042c0d2c11f57e"}]},"branch":"refs/heads/master"},"d001e3b439bc72cd66a2f5914e4be7c784e28409":{"kind":"REWORK","_number":10,"created":"2020-06-05 14:21:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/10"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-05 14:21:34.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d001e3b439bc72cd66a2f5914e4be7c784e28409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d001e3b439bc72cd66a2f5914e4be7c784e28409"}]},"branch":"refs/heads/master"},"d2fc8b2b03da8cbf10226cd621a6fde29e9e968b":{"kind":"REWORK","_number":11,"created":"2020-06-05 17:15:32.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/733262/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/733262/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/733262/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/733262/11"}}},"commit":{"parents":[{"commit":"8ef7d60c9141de46388197a647f81dbd23be735c","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef7d60c9141de46388197a647f81dbd23be735c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-03 17:02:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-05 17:15:18.000000000","tz":120},"subject":"WIP: zero downtime upgrade","message":"WIP: zero downtime upgrade\n\nChange-Id: I699f95c89506de84074788b04cea32352d9d31b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2fc8b2b03da8cbf10226cd621a6fde29e9e968b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2fc8b2b03da8cbf10226cd621a6fde29e9e968b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
