)]}'
{"id":"openstack%2Foctavia~733670","triplet_id":"openstack%2Foctavia~master~Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","subject":"Fix batch member create for v1 amphora driver","status":"MERGED","created":"2020-06-04 17:50:12.000000000","updated":"2020-06-20 05:43:20.000000000","submitted":"2020-06-11 22:06:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733670-1591913164904-2acfc7a0","meta_rev_id":"1996d3262a5ba3f09e5bdd539ff34f89b301ba44","_number":733670,"virtual_id_number":733670,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 22:06:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-11 19:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-08 06:41:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-09 20:33:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-06-08 03:12:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-06-11 19:08:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-08 06:41:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 03:12:39.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2020-06-08 06:41:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-09 20:33:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-11 19:08:33.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-11 22:06:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5d5de7727197216ae06c2a7325f6fb5b9aa897c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-04 17:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6194fe03d7e8781fc8532e08d975d2095ec28b47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-04 17:53:43.000000000","message":"Patch Set 1: Backport-Candidate+2\n\nShould go back as far as queens IMO","accounts_in_message":[],"_revision_number":1},{"id":"2388a8dc0aa4057b7d73b1460725bba2bbe4701b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 19:38:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2560bea5072f46d5a5b62d39d99a655a : SUCCESS in 8m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbc830625e524c00abe1ea0c368f6907 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad51d672837417b8c6f6547af1846c9 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a0196d35eda4237a17df7f69752037a : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bae88b677574fd4b87081f0cfddbe06 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf67c879da942fd8c6fe93beca0a4ae : SUCCESS in 11m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1fc8f71d6cff497b96e8b76637926548 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8b35cd39d13e4228a65ce5e74fe5b582 : SUCCESS in 8m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04d79e12c1924da48a51d25b56aa550f : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60069525dbbf4a71a9aa6829603448b3 : SUCCESS in 1h 01m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08901c20037b4a91b000258d26acb54f : SUCCESS in 1h 26m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a32d3aaa852467c893ff88a2afb5ea8 : SUCCESS in 55m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/51b36c9b92184ed08ec1dc5de866155f : SUCCESS in 52m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce6f76aff9c34a2dab3b0f0315e4a560 : SUCCESS in 1h 08m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6827ebddb04a4e7f952e99092051ace1 : SUCCESS in 1h 44m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/135fbc7a2d10425fb836642d92ddb105 : SUCCESS in 1h 21m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ef59bc18ea64a679bddd6632827c37b : SUCCESS in 1h 22m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3c44d8cdc60a4617949fdcc2e4e3f7bf : SUCCESS in 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e64f16b21ecd50cb295a3984196afd95394f756","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-05 06:37:34.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for the quick fix","accounts_in_message":[],"_revision_number":1},{"id":"17134cab8606bac876a02ff770359d25419936c8","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-05 06:59:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08879d9eb928eaaf84ca3717d66c2c453ce4186e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-05 16:00:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0adbcfb7280049fdfa4c8a867e920c4d544d0a21","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-05 16:08:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c2f009295ae7b0fecaf1ae34a60115e9cc34e48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 18:13:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4532ca1416e54ef5a50c886ca8280226 : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0cf15b846394403b2e6439503d0d9ac : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c177e5ef8c4f6abd6905128b6a8818 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a042f63e9ed54190a536fcf3245b0aed : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/712c8dbc1a894303b3a9878c1323b5fb : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4311bb9abe1f41c59ae13213987e382f : SUCCESS in 11m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/da19a98b02cd4c4a8c718b591d2e2e0c : SUCCESS in 7m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bc0d13da7e504c96941574e197526c52 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d293723e3d6f41c3b0e4fa7effe34f62 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fab5f808e854fdbbce20eb5bae7cc18 : SUCCESS in 59m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90fc436184fd4ec28c2336342ddbe972 : SUCCESS in 57m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/718f6969fd244adfa1d2428e7304af50 : SUCCESS in 48m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8d2fe6d109554efcbb4c9aa0b405a20e : SUCCESS in 51m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9dc151d3590e4d34af56cdff73827f66 : SUCCESS in 58m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29d643152d114fd38121a830927ec346 : SUCCESS in 1h 31m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6183863b0a094f33862697dfb5b42045 : SUCCESS in 1h 38m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c725b623e1034f58a27123cd26b5217a : SUCCESS in 2h 01m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7e8ae9ab09204365860314a77741b82c : SUCCESS in 1h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8585e2f8cab3ee05e9e9062c6d144bade45ed5e","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-08 03:12:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8f4765faf18428062241943f6374dcff648a124b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-08 06:41:24.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"b09bba1859aad4240bf2dd26f38d7a9651f4a27f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-09 20:33:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e2727add1b812f49d166030e8e0a40af8741be7e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-11 19:08:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9855fdc97b1d190fe652069598af04cf9a046061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:08:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd61a07f99ee7c4185fe6c19efb03eab6d90b84f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:06:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7af915c9a53940dbb25d8d034c9c88fa : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba8366ca4b0476190b83e97de331381 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2ba9e2f361949a697b8fdaf80b62672 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cc16960c50c49f9968273eef80a6d1d : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cabff8b8851454796d6ba84fbbce337 : SUCCESS in 15m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e96fe4cb569145c68705f032fe0d587a : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f0c1a5764c64977bfbd97d6307b915b : SUCCESS in 47m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e49babfd1e6641f7b6876878a77c0e19 : SUCCESS in 1h 02m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31298ea271764e62925fec8ad97a1518 : SUCCESS in 1h 09m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/045959e6e2aa4515aa2eba88fdbbd0d6 : SUCCESS in 50m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e69fe087a1664f7f9bdf2b5a11941cae : SUCCESS in 51m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33cbbfc118414e7e8bc96f61e362a74c : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e154437ddfbd7b766f35322a4160b4a9cd7ecba6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:06:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ca18875c5063715637c21b95b9acf615ca0a714e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:07:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a9825453714c00b46cf4ec456df419 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1996d3262a5ba3f09e5bdd539ff34f89b301ba44","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-20 05:43:20.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit bd2987ba234494aacff6268a14dd2e9ececa0108","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"630a4e6a3cb60a8edeb4d8ad4a9cd7b31b42469a","revisions":{"438989975bd8536260876f153be7112406728492":{"kind":"REWORK","_number":1,"created":"2020-06-04 17:50:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/70/733670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/733670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/733670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/733670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/733670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/733670/1"}}},"commit":{"parents":[{"commit":"0321c28588acf4e425dcfcc9f63353808e53247d","subject":"Merge \"Add TLS version configuration for pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0321c28588acf4e425dcfcc9f63353808e53247d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-04 17:45:35.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-04 17:50:10.000000000","tz":-420},"subject":"Fix batch member create for v1 amphora driver","message":"Fix batch member create for v1 amphora driver\n\nA previous patch[1] missed batch_member_update when adding database\nrepository \"get\" method retries for new object creation actions.\nThis patch fixes batch member create to retry the database get call\nwhen new members are being created via batch member update.\nThis issue only impacts the v1 amphora driver as the v2 driver\ndoes not need to get these objects from the database.\n\nStory: 2007581\nTask: 39503\n\n[1] https://github.com/openstack/octavia/commit/48e85569f7e43e3dd5f09fd1ef4cb165526a92cd\n\nChange-Id: Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/438989975bd8536260876f153be7112406728492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/438989975bd8536260876f153be7112406728492"}]},"branch":"refs/heads/master"},"630a4e6a3cb60a8edeb4d8ad4a9cd7b31b42469a":{"kind":"REWORK","_number":2,"created":"2020-06-05 16:08:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/70/733670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/733670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/733670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/733670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/733670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/733670/2"}}},"commit":{"parents":[{"commit":"0321c28588acf4e425dcfcc9f63353808e53247d","subject":"Merge \"Add TLS version configuration for pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0321c28588acf4e425dcfcc9f63353808e53247d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-04 17:45:35.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-05 16:08:48.000000000","tz":-420},"subject":"Fix batch member create for v1 amphora driver","message":"Fix batch member create for v1 amphora driver\n\nA previous patch[1] missed batch_member_update when adding database\nrepository \"get\" method retries for new object creation actions.\nThis patch fixes batch member create to retry the database get call\nwhen new members are being created via batch member update.\nThis issue only impacts the v1 amphora driver as the v2 driver\ndoes not need to get these objects from the database.\n\nStory: 2007581\nTask: 39503\n\n[1] https://github.com/openstack/octavia/commit/48e85569f7e43e3dd5f09fd1ef4cb165526a92cd\n\nChange-Id: Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/630a4e6a3cb60a8edeb4d8ad4a9cd7b31b42469a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/630a4e6a3cb60a8edeb4d8ad4a9cd7b31b42469a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
