)]}'
{"id":"openstack%2Foctavia~733684","triplet_id":"openstack%2Foctavia~stable%2Fstein~I057f9a8353de471d605a5a5c2860d7caa9d3bd28","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I057f9a8353de471d605a5a5c2860d7caa9d3bd28","subject":"Do not install diskimage-builder from Git","status":"MERGED","created":"2020-06-04 19:14:53.000000000","updated":"2020-06-11 04:19:13.000000000","submitted":"2020-06-11 04:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733684-1591849000719-d9d06e61","meta_rev_id":"390f99b9a8e83ebbb72f5ad150c50108d51e1376","_number":733684,"virtual_id_number":733684,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 04:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-11 02:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-06-10 17:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-06-11 02:23:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":-2,"date":"2020-06-10 17:34:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 19:00:46.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-06-10 17:34:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-11 02:23:35.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-11 04:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"178cfe4409de06bc440ef7a6c1df9f18c47392ac","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-04 19:14:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaaf8edd22430b8ba9319e238995dfd71806b148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 20:53:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94ea2a09ec31453c9ba1206c86d3f7d9 : SUCCESS in 9m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cf41393cb0a49f6ae7d7766571a99d3 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f381cee2754699ae6dffea0060832f : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/376a93044599420685b9d5327092f161 : SUCCESS in 12m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7a62b76a41344ceb8738d9303756c7af : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bae63011a7b4f07a9af9ab8d5d8c614 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2020f31e11a744ae96be9e8485ea1dd6 : SUCCESS in 6m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/fc83c6d374c54ac6a115c88e9eab4b65 : SUCCESS in 1h 17m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6916d9aac4e04214b2c2990dfd38a417 : SUCCESS in 1h 30m 37s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/985edb90f9fd4c488d0dff56aa222f5e : SUCCESS in 1h 20m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f1fabdbd63a74293ae093ace5ffafd18 : SUCCESS in 1h 17m 58s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e5ae613deb44257ab0b5f377b7a5cbf : SUCCESS in 8m 16s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/40e2318861694d85b086f3c602027583 : SUCCESS in 8m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59488b2da2e147cbaa13414e8acf367d : SUCCESS in 9m 01s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5da4fcf23f4448a5a573b77bf0c5024f : FAILURE in 21m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b1da93c480944e61a86840945ebcf8c8 : FAILURE in 28m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67f1c3959d8146a98898c5d0624f1f27 : SUCCESS in 52m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ae1e5fca658e472597af8b15c9eaf6ac : SUCCESS in 40m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6f78173f53b472a97170b3d82d4cf06 : FAILURE in 27m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ad15f1115f82485da96c2964c6ade4ae : FAILURE in 25m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/55ef7f4ceeff4c7d98fa5c4378f0f85e : FAILURE in 21m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a67dd2af86448d591ed4a0bfa4eb3f5 : FAILURE in 28m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd018be3399a447e911a097fc98c7f00 : FAILURE in 31m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7be5dc460f3d4cf19913f2a92739b8ac : FAILURE in 28m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a82efb461f2241348b8468ae13d364f5 : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf1ba09dcf74412da761ae4ff19c226d : FAILURE in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b7b13d2fd23d6da51a5d9397f542a1d504c27dd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-05 18:31:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4654d4aa3ac84bb3d7f279187934edeb1d655373","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-06-05 19:00:46.000000000","message":"Patch Set 2: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"2b38be457e916d320792e89ce7ddc247418bc4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:29:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58368a6d508d4b5389234c6cfd53b6b3 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/331d226b5ed84b179d4e49d7ac9b190b : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f5c14adf9549c8a5bd2d1e56acb595 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4313594af43f497bb5a6ab8f92732237 : SUCCESS in 6m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67d0c489fcd0491c8d37b08cf4df9513 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/848ede3303c04ffcb3172c5ef8e4c8e5 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617747f6f43a4dcca234e51a4f6525c2 : SUCCESS in 7m 03s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1f7af2d790fb411d961cb058d2b833ed : SUCCESS in 1h 17m 19s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f0cd36b505844f0b85828011023c6112 : SUCCESS in 1h 15m 16s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33cea31f07754c0b8ebfaed9c61bdc02 : FAILURE in 26m 31s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c95f3b12982245f69d22c3e9661f7820 : SUCCESS in 1h 19m 57s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d0515891d434d72bdbc94e33faf3247 : SUCCESS in 7m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e8b80145feac4f6da0fa1fc5d5aba266 : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f2a8b906415945838b7c419447511acc : SUCCESS in 7m 13s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbbd89bbc300444ba0735e32e63a8a27 : FAILURE in 29m 57s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8d5e541fd54d4d1cafd66071e70e7a97 : FAILURE in 30m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a29c81c13fc74facab98ea2ca4e808a7 : SUCCESS in 43m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/21b9b406da61471387369033932d7831 : SUCCESS in 48m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c8faaa555a84d569e1a19421ef0d030 : FAILURE in 24m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d070fa1ab2d445e0a509f85f7e256e80 : SUCCESS in 1h 57m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/39b74719320c4e24b9c55c881c0f9a2d : FAILURE in 22m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45567ebb2b67429e898f08b50bfe5f2d : FAILURE in 25m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3cc34c3792eb4627b5a5f8ad39353fed : FAILURE in 27m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/97506b625b3e46c096d093609966b713 : SUCCESS in 1h 38m 01s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5359980bda5247b1ae31a74f529c37d1 : SUCCESS in 1h 07m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7353aebbb13744deb1fe2d3195385486 : FAILURE in 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa5b82d0c379b7c2563f065a4d693f1234b3fe2c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-06 12:55:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f7cb9042758495a339eb4a620086e4725680822","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-06 14:39:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"afe643a566499c80aa113e056ce7149ee9a47231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 16:23:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7209dae22b0c4d619033b37761d4282e : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66695a7954334fc7a2e38f77079dc539 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261dd6966d2a406a8c2f9e89ae6fd47e : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8256dc836bea42e1b335064ad3981142 : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/253975aa04284bc2915d48eee260a4c9 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c94ea7f5bcb848019954260030b68b16 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6403c236b64d7b9d7b1a572e47fc55 : SUCCESS in 6m 04s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f680d788d7604c2ca96a35e688837038 : SUCCESS in 1h 18m 09s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f2895df9893b439d8754bdc8e7a9da84 : SUCCESS in 1h 13m 22s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/548f7ea5dc5c4a2993fc4eb9ae159495 : FAILURE in 30m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2b36afd966ce42078f2914fc738282d4 : SUCCESS in 1h 22m 21s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ce4c7e833e6749268d0280090fdf0222 : SUCCESS in 6m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/bf208729d47945eabc35efb9f2e97257 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05741416142f4253a38cc937570e5bf5 : SUCCESS in 7m 32s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a3ef5e669ec43b69acd87895a1414e2 : SUCCESS in 56m 46s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/855dd98d8a674e8587a7590bcb528caf : SUCCESS in 55m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f1121deb3245490180504fb1754754e1 : SUCCESS in 42m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/18b9774907ee47c89730c17cbcfcf1ae : SUCCESS in 42m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a434d7a0461b4e399f2bebf77ded385d : SUCCESS in 1h 25m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/57102ec467c24f97ad3238aa52f2eb98 : SUCCESS in 1h 04m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6b2601c702c046fe9d3ad4644b6d73f3 : FAILURE in 24m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d52b9cee4844e20bdbb3187cda7b72f : FAILURE in 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/52cdd4459e764c30b0c9912cad4ebfb5 : SUCCESS in 1h 10m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c62370ebe0764604ba58d6c3b49658ae : SUCCESS in 1h 38m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d1b7868435014dc2a99579d234a56187 : SUCCESS in 1h 11m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c6bf4a659694e62afee331358ffb71a : FAILURE in 56m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"235770a206f384a337581b1221c46c3f8bbccb5b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-08 16:11:59.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"107045b3be274e62de3c9d39cda41883431650f8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-08 23:53:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bd19acbcb79a9e1a30542b2d1a1f840388dfb9fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:30:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e50f2a2bc890904241b98585cc85cfd709dd9a5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:39:57.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"31d55477316f04d85a029a04835b667ea8f19cc3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-10 14:01:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e6b1a15bba34dd7afd48082193d3d6b378767c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 16:19:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c32dbd285caa4e8ea0c71ab4374441f8 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12cd7e0320d7457e97098c6f2ce99391 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65a0021841147efb0628975cbaaeecc : SUCCESS in 8m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02ed9b03e8b54db3bd0b8f37ccedddaf : SUCCESS in 6m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7208bba0495e4f67a9ed8a2ed187ec0e : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7ad62aba89640ac9f2b0f6295858f75 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1931d99c3e42a2846d5b22bfeeb1d0 : SUCCESS in 8m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0df6569c6e0b45f482f9112a230ba170 : SUCCESS in 1h 25m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7f0706f3dd0c40baaf9471158a2c7ed1 : SUCCESS in 1h 22m 55s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09031caf615d4de6b5063a055667ca87 : FAILURE in 34m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b31019dc072847ad96f3ad06a67af29f : SUCCESS in 49m 27s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/756f1161c2ab4366885b6573a8265b5b : SUCCESS in 8m 29s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0ec908708d9e4097a05ae585efbfadc9 : SUCCESS in 9m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f72a2681bf854c40a9a50cf63bbcb094 : SUCCESS in 9m 00s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a907f2cdc9b34751b1222117b49d4668 : SUCCESS in 59m 08s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0109274e7e4244749acecc5e6675f6dd : SUCCESS in 57m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55b9507eace34d2da28b313f26221d4e : SUCCESS in 52m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c170463b34ed4700baa39b4252ab2c5b : SUCCESS in 49m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19a134e040294852a942e9a12681b0ac : SUCCESS in 1h 16m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4862d66fa35040ecb43bd39eb092bc5f : SUCCESS in 2h 04m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a783feaa404644af8cabcb124c2caedd : FAILURE in 29m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a742d2e62d242369c2ca63b3b358c13 : FAILURE in 40m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8591f57a6c7d487ca5beaa117b3828b7 : SUCCESS in 2h 10m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/08f6078cdcac4e819ea86a157333cfb3 : SUCCESS in 2h 07m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/477aa9f3db93444ca44a8cea7000f9b7 : SUCCESS in 54m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8b2cbe06a604e0ebdf9ee7312807300 : FAILURE in 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0249233d518e7134346ef2af7fdf3401d6483974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 16:19:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"00562a517e8d0825d240ac04152363fae3a8b67c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-10 17:33:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75a322c9ea1497d039d6aff29aea13d3aa5e49dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-10 17:34:13.000000000","message":"Patch Set 5: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"7c2255168f2fb2ec9373cfbd5036eadc68013551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 20:10:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8925a6a99e4b42a68d232ae676e6489b : SUCCESS in 11m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64257ac29e084ce19bd5778d0eb44d66 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc2f93ede4c4bd1915a71b1ffcfcd11 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0cd55c26dba14ac78c5cdb507b329459 : SUCCESS in 7m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9018ec8190c94e1ab586b66418771b79 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be56532234f49288fc1423a5599090e : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d08f90f99064d32927488d417b0350f : SUCCESS in 7m 22s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/92b25cf11d56460dafbccc5583a88e4f : SUCCESS in 1h 19m 22s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6090b528c3ff4e7aade95778b26eb83c : SUCCESS in 1h 22m 17s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88118101b1114f4eb3c26a708ba683b9 : FAILURE in 29m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/237749e2524f4ceda8a269b5dedec0f6 : SUCCESS in 1h 31m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/24da7beed84f42ae8da40cab4a00313b : SUCCESS in 10m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7db1dfe33bf941d4b49b6a78304a0dd0 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/830734d071fa4dc290492ddbc1ed6318 : SUCCESS in 10m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/819862af23294e3cbfc767e072383b0e : SUCCESS in 1h 06m 40s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a56f53126a164262b6bcb10c51815e55 : SUCCESS in 55m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54e6a92a4088419e9e392fd15ea8400f : SUCCESS in 43m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/de49e684aedb4d4cb9017d8bfcd34c01 : SUCCESS in 47m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d9baa7c7b42450ba98b81f7e5609adc : SUCCESS in 1h 39m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/38da2829773f411089df7165699f4945 : SUCCESS in 1h 35m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0f49d657550b4b7e8f01311f1f737663 : FAILURE in 29m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a92f4baefa994eca8b3736ff31a457a0 : FAILURE in 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e513e1bf6c1b4cac8f719bd0bc344f90 : TIMED_OUT in 2h 32m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/86a6d5ea1d864d2389f066497ee01b7b : SUCCESS in 1h 22m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/afefd5e9b4b040c08c84d8275b3443dc : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2094b5612de484b9406b67b84acaa1b : FAILURE in 52m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b94d2a014472674b4ff693e8351117f4cde31a0d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-11 02:23:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c3ec13356569d693e7b223201d4a0ee5fd370f66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 02:23:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f44acdacc4f744cfe192999f3bc9a3f3bd3396a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:16:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37db0222c19547e89c97e4396c314247 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/622e4c40a0c6440ab07f1b66e0ea2cd3 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e10d0461f7de47d0993ea833ca2be31a : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/77e2396672d0405bb5a8e33eee5d237e : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f48e342c754bcbaf9ee4d9a34a9af7 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938e6d76cb634a3d9114b572590e8f91 : SUCCESS in 8m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d9bc29c848348dd93d1480362867a90 : SUCCESS in 7m 15s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3f5a95824b964eec8a24342591c54679 : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/488c5ba439c24aae9ac07d5c4cdac863 : SUCCESS in 6m 14s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b6cc7edb87e4110a6bf999ba9ab3ca4 : SUCCESS in 57m 37s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fca860844cd84e2eaf76f0a5d4ca4c6f : SUCCESS in 39m 28s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c8a925516c60489f9299efaed5663008 : SUCCESS in 1h 16m 45s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/36f8ec3b9f9b4535ba2eafaa4732cdd7 : SUCCESS in 1h 51m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b2f8a0f433246bf9f223304e6275cc7 : SUCCESS in 48m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2a60f077544f4c60b33bb408183d63ed : SUCCESS in 1h 08m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1979914ddb7e419dbd0fa79bbe0f40a7 : SUCCESS in 1h 27m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc03bb2e7f864b3ab7fddef5f79db211 : SUCCESS in 1h 34m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/089e81526d0e482ba77909829b61e143 : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":5},{"id":"6d452c3a81a7c4907f5f6deb68e3cf5901ad9efe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:16:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"390f99b9a8e83ebbb72f5ad150c50108d51e1376","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:19:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a6ff135adb49cd8ba928e2100e7b6a : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aefbef124586d624d8978342942585b6da2049af","revisions":{"b954b86c0f5b5f5be49cf08b2bb31f477fa78c3f":{"kind":"REWORK","_number":1,"created":"2020-06-04 19:14:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/84/733684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/733684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/733684/1"}}},"commit":{"parents":[{"commit":"5dd672c58c322e142b0e187da82267ffde606450","subject":"Do not run Tempest in octavia-grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 19:07:03.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 19:07:03.000000000","tz":120},"subject":"Pin diskimage-builder on git installs","message":"Pin diskimage-builder on git installs\n\nDiskimage-builder recently dropped Python 2 support. Even though there\u0027s\nan upper constraint version for DIB in python 2 installs, that\nconstraint is ignored when installing DIB from git rather than with pip.\nFor this reason, we need to pin DIB to a Python 2 compatible version. We\nchoose 2.30.0 as per the upper-constraints.txt in stable/stein\nrequirements [1]. This should take care of both user installs and CI\n(dsvm live base and child jobs).\n\nNote that Python 3 devstack installs will also have to be pinned to same\nversion because stable/stein devstack ensures that all libs being used\nfor development are installed under both versions of Python [2].\n\n[1] https://opendev.org/openstack/requirements/src/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323/upper-constraints.txt#L511\n[2] https://opendev.org/openstack/devstack/src/commit/ab86ced64cb6c053cf3340ab4d8a1d988a6b76c4/inc/python#L379-L388\n\nChange-Id: I057f9a8353de471d605a5a5c2860d7caa9d3bd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b954b86c0f5b5f5be49cf08b2bb31f477fa78c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b954b86c0f5b5f5be49cf08b2bb31f477fa78c3f"}]},"branch":"refs/heads/stable/stein"},"cafb484d1d7299b4ca720c192cde9b7b14b9e7f8":{"kind":"REWORK","_number":2,"created":"2020-06-05 18:31:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/84/733684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/733684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/733684/2"}}},"commit":{"parents":[{"commit":"5dd672c58c322e142b0e187da82267ffde606450","subject":"Do not run Tempest in octavia-grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 17:44:23.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-05 18:30:27.000000000","tz":120},"subject":"Do not install diskimage-builder from Git","message":"Do not install diskimage-builder from Git\n\nDiskimage-builder recently dropped Python 2 support (\u003e\u003d 3.0.0).\nInstalling DIB from Git pulls Python 3-only DIB which breaks Python 2\ndevstack environments.\n\nThis patch drops the option to install DIB from Git. DIB will now\ninstall what is set in stable/stein upper-constraints.txt [1].\n\n[1] https://opendev.org/openstack/requirements/src/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323/upper-constraints.txt#L511\n\nDepends-On: https://review.opendev.org/#/c/733778/\nChange-Id: I057f9a8353de471d605a5a5c2860d7caa9d3bd28\n(cherry picked from commit 4a95b354901c81f135b61d39754e1419dd134f0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cafb484d1d7299b4ca720c192cde9b7b14b9e7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cafb484d1d7299b4ca720c192cde9b7b14b9e7f8"}]},"branch":"refs/heads/stable/stein"},"84d645ca7e26117bbcac4adebf4a75cb7ab6a125":{"kind":"REWORK","_number":3,"created":"2020-06-06 12:55:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/84/733684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/733684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/733684/3"}}},"commit":{"parents":[{"commit":"5dd672c58c322e142b0e187da82267ffde606450","subject":"Do not run Tempest in octavia-grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 17:44:23.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-06 12:55:14.000000000","tz":120},"subject":"Do not install diskimage-builder from Git","message":"Do not install diskimage-builder from Git\n\nDiskimage-builder recently dropped Python 2 support (\u003e\u003d 3.0.0).\nInstalling DIB from Git pulls Python 3-only DIB which breaks Python 2\ndevstack environments.\n\nThis patch drops the option to install DIB from Git. DIB will now\ninstall what is set in stable/stein upper-constraints.txt [1].\n\n[1] https://opendev.org/openstack/requirements/src/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323/upper-constraints.txt#L511\n\nDepends-On: https://review.opendev.org/#/c/733778/\nChange-Id: I057f9a8353de471d605a5a5c2860d7caa9d3bd28\n(cherry picked from commit 4a95b354901c81f135b61d39754e1419dd134f0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84d645ca7e26117bbcac4adebf4a75cb7ab6a125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84d645ca7e26117bbcac4adebf4a75cb7ab6a125"}]},"branch":"refs/heads/stable/stein"},"91b8e97b05aa77ff3496319760f6946ec16e2e72":{"kind":"REWORK","_number":4,"created":"2020-06-06 14:39:57.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/84/733684/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/733684/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/733684/4"}}},"commit":{"parents":[{"commit":"5dd672c58c322e142b0e187da82267ffde606450","subject":"Do not run Tempest in octavia-grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 17:44:23.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-06 14:39:45.000000000","tz":120},"subject":"Do not install diskimage-builder from Git","message":"Do not install diskimage-builder from Git\n\nDiskimage-builder recently dropped Python 2 support (\u003e\u003d 3.0.0).\nInstalling DIB from Git pulls Python 3-only DIB which breaks Python 2\ndevstack environments.\n\nThis patch drops the option to install DIB from Git. DIB will now\ninstall what is set in stable/stein upper-constraints.txt [1].\n\n[1] https://opendev.org/openstack/requirements/src/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323/upper-constraints.txt#L511\n\nDepends-On: https://review.opendev.org/#/c/733778/\nChange-Id: I057f9a8353de471d605a5a5c2860d7caa9d3bd28\n(cherry picked from commit 4a95b354901c81f135b61d39754e1419dd134f0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91b8e97b05aa77ff3496319760f6946ec16e2e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91b8e97b05aa77ff3496319760f6946ec16e2e72"}]},"branch":"refs/heads/stable/stein"},"aefbef124586d624d8978342942585b6da2049af":{"kind":"REWORK","_number":5,"created":"2020-06-10 17:33:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/733684/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/733684/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/733684/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/733684/5"}}},"commit":{"parents":[{"commit":"5dd672c58c322e142b0e187da82267ffde606450","subject":"Do not run Tempest in octavia-grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd672c58c322e142b0e187da82267ffde606450"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-04 17:44:23.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-10 17:33:52.000000000","tz":-420},"subject":"Do not install diskimage-builder from Git","message":"Do not install diskimage-builder from Git\n\nDiskimage-builder recently dropped Python 2 support (\u003e\u003d 3.0.0).\nInstalling DIB from Git pulls Python 3-only DIB which breaks Python 2\ndevstack environments.\n\nThis patch drops the option to install DIB from Git. DIB will now\ninstall what is set in stable/stein upper-constraints.txt [1].\n\n[1] https://opendev.org/openstack/requirements/src/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323/upper-constraints.txt#L511\n\nDepends-On: https://review.opendev.org/#/c/733778/\nChange-Id: I057f9a8353de471d605a5a5c2860d7caa9d3bd28\n(cherry picked from commit 4a95b354901c81f135b61d39754e1419dd134f0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aefbef124586d624d8978342942585b6da2049af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aefbef124586d624d8978342942585b6da2049af"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
