)]}'
{"id":"openstack%2Foctavia~734002","triplet_id":"openstack%2Foctavia~master~I9348cfc8e87351d94c225f07533b7676dcbe50e3","project":"openstack/octavia","branch":"master","topic":"story/2007762","hashtags":[],"change_id":"I9348cfc8e87351d94c225f07533b7676dcbe50e3","subject":"Also handle network deltas when deleting members","status":"NEW","created":"2020-06-08 03:03:52.000000000","updated":"2021-07-29 22:21:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7c67d3dbdad929835d465ce9847bc892f4cc187","_number":734002,"virtual_id_number":734002,"owner":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-17 10:20:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-13 01:54:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-13 01:54:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-10 20:24:20.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-24 01:04:44.000000000","updated_by":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"reviewer":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"state":"REVIEWER"},{"updated":"2020-10-27 14:37:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-03 04:19:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 22:21:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"e7a98f6b8e91fc8cc9952fec1e7f095aaefad5e5","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-08 03:03:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92d14e486e4eae5d4f0915dadb5f31fc3dd6b095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 05:37:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7428578ade774db69b671e98b8612708 : FAILURE in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e3a9badeeb8467b8b76497c007ec550 : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a927f5f01524ef49b3b915447ae8a50 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/099bc2893816471c9cf7df6a140b11b2 : FAILURE in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf00b6ec916a4400b2db462c78b97752 : FAILURE in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c610f3e92745949f8b12108da36602 : SUCCESS in 10m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f68973270fc747a684c0b63fab4d5eb6 : FAILURE in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f9e430713cd14029b7c826dbb9fdd664 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb732cb6bc0b4acd99aed0cea31ce97a : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a101d1cc4d54149a89ee63149cf5682 : SUCCESS in 35m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87307b3da37f4879a178ebfa9c6c1991 : SUCCESS in 58m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/40700db6a9994a61b9cb0c13e7a88403 : SUCCESS in 44m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e451cd99fc8c44fa842f99888d7f43f8 : SUCCESS in 40m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6299fc99344043d8a7982028c3703d7f : SUCCESS in 45m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e8a3ab6376b4f0eaf9ea797f6bd7e64 : SUCCESS in 1h 38m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aaaf9af7397545c0bfe561eef35e24fb : SUCCESS in 1h 38m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f42ecba41dd7491cb3d0ce4f2f60aabe : TIMED_OUT in 2h 32m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e1f0be62a894102b293183a73c2e114 : SUCCESS in 55m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c0e55d0da3b562e7aad3b466da314af56be004e","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-08 12:04:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c002f8388487d6f09767ce30e08c0ebda6db7e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:13:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/793d08d9ecb74a36bdb5cef2a491a323 : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d627b435e9c42d596d74b925d6f6145 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6417e85023394aa7a881457fd1fed12b : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/682819f40b2d41d6924c73b83ff045d6 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc91b9af8d843f8b7090252189e58a1 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c171102f1b444e0cb36cc56a70162ec2 : SUCCESS in 9m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b3c0775f731d43969c89c690a2d57aad : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ee4b4b3c33f340838a6e0fad80f535f2 : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5cf2c4478e5410a87af8abcd01eb39a : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00fb15c0efbd494ab8ea76944cc2e618 : SUCCESS in 58m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a798560fb2984f3b9de730554797a5a7 : SUCCESS in 1h 51m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b43d42e1393442ecb15ac5d3e3c33443 : SUCCESS in 1h 07m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4abc6504e85f49cd8832b6833311b04b : SUCCESS in 1h 00m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0edb07c6316c470bab6e06281fdec274 : SUCCESS in 1h 03m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae16ddb3632946568551f6a867c7d07f : SUCCESS in 1h 47m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec26625e95f3427b891ffcf3882805a9 : SUCCESS in 2h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cda3edb354e74cfbbd2bb931a9f43748 : SUCCESS in 1h 17m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1c87f4744124558802088f22df31c23 : SUCCESS in 1h 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91ebd14187c83a0978df5c833039d1ea4c0021b1","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-06-13 01:54:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a3d9c91fd53cb01a982c8e7bc29a855146981c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 03:19:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc51171c8377477098776ff6dfce2e17 : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3867c4977f8e4604bc0a0b63f08baf61 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7c81135d4f4939808b5e8549d4c860 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f8c3244ff194d75be93ce2bcf8faa09 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3567440197234850b664e8863c12eeef : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60b5e8638c14e61b831a962ac670cee : SUCCESS in 10m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3274dfda09bc4a2b9616470621675ba4 : SUCCESS in 7m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/07fbfca898414261b408751242921f3b : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/085a1f9575d8487a8318fe15e8de7c8a : SUCCESS in 5m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff76e43983d34c6188de7e48ab65c991 : SUCCESS in 36m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab3f23df9772498bae4e341f27cc63de : SUCCESS in 1h 06m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1f5ce71341c47a2a1d7c500ec372bf1 : SUCCESS in 52m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5b8dd76e73d34fe69afd01f6f82a1721 : SUCCESS in 44m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a859dcff86aa4bf59355f0454d86f47c : SUCCESS in 45m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6ac3f73ed404afda1b6950e83a60ba8 : SUCCESS in 1h 23m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a13d19c380f24ccc81687b27763c78de : SUCCESS in 58m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91b2822991cf4ecf92e254d719ec757c : SUCCESS in 1h 08m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fedb966d7f224489a8210dc857c13b26 : SUCCESS in 57m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"830d35bd452d32cb26e81f95302f3e9aeb384a22","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-09 14:40:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"59410a5b230ea08b5a8be0fac3f860ac05ff2490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 17:56:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a28f368ed24f40268a58d4319ccb35af : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b7d9f6be3ed4654a89dd4fa0276dc39 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3fa87d3e0ed47db87fefd84d6a7fd43 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/213eae9d47bb4ab58263401c5599d42f : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1396cb6d56f40f3a117531b1d3e1d92 : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6567de3111a34f5db59a23b1d91e0f22 : SUCCESS in 15m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9488aaec696047789598d7297c2d385b : SUCCESS in 10m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a814e978a14143d1a1abab6fb0788682 : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb62f38e3fa64a77bc1b89c83542dcc6 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65790ce01140422892653b9289ff8473 : SUCCESS in 43m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94323ee90fff45e48060c51f3ce88905 : SUCCESS in 2h 07m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05de659550ec41dd9b47921c24136744 : SUCCESS in 41m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a9c3a23e77fd47d4b3748ef26fec49bf : SUCCESS in 40m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7878f0e673a444f0b99b3486dc110713 : SUCCESS in 37m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f4dacb757c64d66a0efb2f021509c91 : TIMED_OUT in 3h 04m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/64d8063a45a845a9bda61fd001ee5a33 : SUCCESS in 2h 19m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1022d024f0db47ed9aca6273ddbf65f0 : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0eb62f33148e4a9bae9fa9dc650581fd : SUCCESS in 1h 40m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5177ebc7025c4435a170a3190475bcd7 : SUCCESS in 1h 26m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9844cf9474ac485494ac14504d730d57 : SUCCESS in 1h 46m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5a374c1621e54bc3903a5cda22600c9f : SUCCESS in 2h 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"786469a95f65d8123b94cd5bda9b88f638f7456f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-10 20:24:20.000000000","message":"Patch Set 4:\n\nAre there any tempest tests for this that could actually prove this doesn\u0027t break things in the amp?","accounts_in_message":[],"_revision_number":4},{"id":"37c18df262b0c87a4c6b076ef36b06f70778c512","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-09-24 01:04:44.000000000","message":"Patch Set 4: Code-Review-1\n\nIn my opinion, you should divide the patch into two patchs, one patch does only one thing. Then, could you commit a new story to descirbe the second thing in detail, what bug it fixed and how to reproduce the bug.","accounts_in_message":[],"_revision_number":4},{"id":"9a0b9ec8a0810fed649f603d92ef7fc6d46232e6","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-09-27 08:28:17.000000000","message":"Patch Set 4:\n\n\u003e In my opinion, you should divide the patch into two patchs, one\n \u003e patch does only one thing. Then, could you commit a new story to\n \u003e descirbe the second thing in detail, what bug it fixed and how to\n \u003e reproduce the bug.\n\nTnx, I will split the patch in two patches, and about the second thing I wrote another story: https://storyboard.openstack.org/#!/story/2008192","accounts_in_message":[],"_revision_number":4},{"id":"d3d7002c094a3ed7fe0939163b439f5839bf3704","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-09-27 08:34:40.000000000","message":"Patch Set 4:\n\n\u003e Are there any tempest tests for this that could actually prove this\n \u003e doesn\u0027t break things in the amp?\nthere no tempest tests for this that could actually prove this, but I commit a new story [1] to describe the step of bug recurrence\n[1] https://storyboard.openstack.org/#!/story/2008192","accounts_in_message":[],"_revision_number":4},{"id":"433c0d5febe7e2b35aebb09786c51f96b959755f","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-09-27 08:55:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"888c20e38dc0428cbff7c239ea56b23b58c7acf0","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-09-27 09:25:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"faa86372d18d70627c8c61c93b06406a5942d266","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-09-27 09:31:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b7e5b9f6a70d80954330883a82df1eb524061d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 12:08:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9121800742547d59447f61f916a72e4 : SUCCESS in 10m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc7a022a5cba4696a3edc0144bd495c4 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29848162cc4b4dbfadac29f6c627d1a5 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6beea337589745a68554d6477c0456ef : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80890a7fe16d48258cc10c2e329f3f1a : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f714a5b459654e47baed6e630159983e : SUCCESS in 11m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2cc4bb8a5dfd4922a94813ed9ebca784 : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c5327b9e4a5b4df79cbb1d943c52671f : SUCCESS in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c572a40a7f94f459fd29cc0babbbb46 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dd9691bede44c87b976df5895de400b : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/169824f727e1489b96ee9fe766dbe538 : SUCCESS in 1h 41m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/453abdeae34b449687c7e82fd6928a24 : TIMED_OUT in 2h 34m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06b1d7120ccb465bac6cef55ba77b1f2 : SUCCESS in 41m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/70846750f1aa472d82b5bb5a74a8d4c0 : SUCCESS in 31m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28987150d9084f74945ea7126191ee06 : SUCCESS in 39m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/087ed9545c4d451fbceeeb0411780170 : SUCCESS in 45m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/58a8c561fac548348af9ac2234aeb6f2 : TIMED_OUT in 2h 34m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c333ff749654dea866e3a55fe0b45bd : SUCCESS in 2h 09m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/56f54a094f9f40378c1696a4ff0d9205 : SUCCESS in 2h 06m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d470097f16e44f5ba66737424a031900 : SUCCESS in 2h 03m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a2cedd3818a8476489643831e6cba366 : TIMED_OUT in 2h 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bd7730b45d9840f4a307719d679fb267 : SUCCESS in 2h 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6fc162c4cc376d906950b7c90705ed3480de5846","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-10-09 05:04:07.000000000","message":"Patch Set 7:\n\nreset","accounts_in_message":[],"_revision_number":7},{"id":"26b128354e659c4aa4589b5bbb0c15b4a037e084","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-10-09 05:12:01.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"31a752d43aa2c51cd1fe5f4496fe4d46f1eb5df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 08:01:54.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dde0e73b32a482692e1a3c63546a5c8 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a47e4be478e4021b1f6ed45c61aede4 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee0724f360c435b9add9ff3b4eeb0d7 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e200646fa9b4059a4ca78b4a252db32 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b90b916264e47a89bd9ed11998f5011 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c15b909fe4b42e4934f9c59f13088bf : SUCCESS in 22m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/600a701448a64fed9a08d743e72649c9 : SUCCESS in 4m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/26405354df3c44d49d371aa2f2c7d383 : SUCCESS in 7m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/732833bdeca74175bc24c25754fcf370 : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4fd32ded2dc9482085e038cce75b915a : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5be426af061345dc9d1638a52333deca : SUCCESS in 1h 34m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2163e392c7d442982198777c97b6dd7 : FAILURE in 2h 18m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ac242b6b2974ac4a1d875064726db61 : SUCCESS in 36m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9e8ae39f333e418dabecd7b0d2e929cf : SUCCESS in 58m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/787386ee4aee4793967440ec93df9365 : SUCCESS in 41m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59603cbc80f04e929c8d8682860c449d : SUCCESS in 1h 00m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/039d41ed45be46c88e809febc47289cd : TIMED_OUT in 2h 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e6e5d616343408d9a621ef6aa5057f3 : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3d0ad0d4e313411a90aa3bb2a51f3598 : TIMED_OUT in 2h 35m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5d954a9eef5f476986396b789699ec86 : FAILURE in 2h 12m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8115918502d543d28472121f04720b99 : SUCCESS in 1h 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/92157dd0b4644371a9559f0574c90105 : TIMED_OUT in 2h 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a4a945c1c65913389ab6d945f193eac6327f5701","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-11-03 00:59:45.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"87fe1e5daee29821ad192957ffceb564f8f2c805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 04:19:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f154edd7d62845ac9a28282e8c0bb633 : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/676ece155ac5474aa3f70bca6d58f85e : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5477cbd89b3045e6bb2b115105d44564 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a8b6ea08beb492d8fedbb1a471ce226 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5985766c6efc4215af8f2bb003983ca9 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1a66bf9fb7c4e7a867010cbbd66d7b9 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0433994b6bc14256824d28471bc363e8 : SUCCESS in 12m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8aa70d77d90e41d39921f7ac63dc53c5 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/827913eb1d7c40d98268200b69e1a1d8 : SUCCESS in 9m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5a645a94f7774a1490e690abedae1da1 : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8172401315544d21ad4a94d717280f2d : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5cb8289bd5c641cc897890d3a67cc2bb : SUCCESS in 1h 25m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c27984a8b62d4290abf74561f38cb509 : SUCCESS in 2h 14m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/807a5ad2fe9b44208869122c2c460ddf : SUCCESS in 37m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/910148f79f2b41efb37b555e080f51db : SUCCESS in 41m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dc5286055e9423d9c7cb4e4d30fb1f7 : SUCCESS in 42m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/574f2a78a0e14851bfc73da878407129 : FAILURE in 56m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ab093b2eb9d48b999bbefd165ea2e1c : SUCCESS in 1h 43m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3a3eb83d9ec549cea81c2abae908c9a2 : FAILURE in 2h 25m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/edfa2111fdb040eca3af197e63907fbc : FAILURE in 1h 51m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d4eae45ca5647beae7de19c5034396a : SUCCESS in 2h 18m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/518606514b7c4cee91ce4af9cc35c98b : SUCCESS in 1h 54m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9aaec9ad8d0d4f6e9af4ea47cbc4b77f : FAILURE in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8c1d1b5b247cecf36e10eb56d7682451dd644b95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-01-17 08:06:07.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"2ed0d6bc0c0976dd94fbd9d3f7dc0ec754ac0106","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-01-17 08:15:13.000000000","message":"Patch Set 9:\n\nHi, octavia team:\nThis patch fixed bug already influence our app\u0027s business. I wish that octavia can improve this patch\u0027s priority so that this patch can be merged as soon as possible.","accounts_in_message":[],"_revision_number":9},{"id":"74908b87a88182aacdc9c0d2a758165c2090d416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 10:20:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0568d8105184687a468f25f987077cc : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/666707d060c24df387719aa610f9cb1c : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72d3b8f71c04590808df8e1acc24daa : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a7b12ef04444ccaa250ebdd609a5ce9 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf25d74833f4fe0a4ff9980f65c86df : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d81fad52d4914e15b5a31fa15a3529eb : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d094008b4ff54dc3a0d1227d835964eb : SUCCESS in 11m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0e03f4c887a74777b4a59220ef40dbbc : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/da632fe796754cef92bf46301c99a69a : SUCCESS in 9m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46d6ece994f9477ba1c09a7c29e2be07 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3fa5882d15254b389f9ab1cd8b7a9bf5 : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/294eff934f5c4d9ab20bc78777f5fcb2 : SUCCESS in 1h 07m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c11305d3dd0649ab8512ac776b30525e : SUCCESS in 1h 33m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c368d0d7f9e4426b97161eceee5636b : SUCCESS in 35m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/545dc28118ff4029bf52751774e39de6 : SUCCESS in 40m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c15c5087ede49d5bf294750c7f0b978 : SUCCESS in 34m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/94849a7ebbda430ba5276903bef190df : SUCCESS in 54m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/28156d1546ae434f840a15254294f7a0 : SUCCESS in 2h 09m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db1955876ba54d16a10f552103c26e03 : FAILURE in 2h 07m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1cde9f742725432aa2c370742dc23077 : SUCCESS in 1h 42m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/54ae2971f2de49fda41d9a522a68718d : SUCCESS in 2h 09m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/90a3a57aaa584da4b0eeed613fc84aa2 : FAILURE in 2h 05m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/79303e19f2d24ccfb20bb70f2815ccfa : FAILURE in 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a7c67d3dbdad929835d465ce9847bc892f4cc187","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:21:01.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"87322714a399f335a04797cd70cd8c3336dfa3d1","revisions":{"7b84f0ede4ebb18928956db277812f5d0b42988b":{"kind":"REWORK","_number":1,"created":"2020-06-08 03:03:52.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/02/734002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/1"}}},"commit":{"parents":[{"commit":"3980c904037651c646df2a1af816ef32b01f9db2","subject":"Merge \"Remove Babel requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3980c904037651c646df2a1af816ef32b01f9db2"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-06-08 02:23:27.000000000","tz":480},"committer":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-06-08 03:00:25.000000000","tz":480},"subject":"Handle network deltas when delete member","message":"Handle network deltas when delete member\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b84f0ede4ebb18928956db277812f5d0b42988b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b84f0ede4ebb18928956db277812f5d0b42988b"}]},"branch":"refs/heads/master"},"fdb499f080ba2a9abd345c3fbabeb8b6d638dba4":{"kind":"REWORK","_number":2,"created":"2020-06-08 12:04:17.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/02/734002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/2"}}},"commit":{"parents":[{"commit":"3980c904037651c646df2a1af816ef32b01f9db2","subject":"Merge \"Remove Babel requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3980c904037651c646df2a1af816ef32b01f9db2"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-06-08 02:23:27.000000000","tz":480},"committer":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-06-08 11:13:51.000000000","tz":480},"subject":"Handle network deltas when delete member","message":"Handle network deltas when delete member\n\nThe patch does the below two things:\n1. When removing members, the nic of amphora ns\n   associated with member will be deleted. \n2. the naming method of the nic in amphora ns\n   will be changed to avoid name conflicts\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdb499f080ba2a9abd345c3fbabeb8b6d638dba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdb499f080ba2a9abd345c3fbabeb8b6d638dba4"}]},"branch":"refs/heads/master"},"ccdc038766b2cfd04c91e00eb17c5202e9164696":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-13 01:54:56.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/02/734002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/3"}}},"commit":{"parents":[{"commit":"6418ae00c9267f363a17f69d5fb92adaa9c08a8e","subject":"Merge \"Fix netcat option in udp_check.sh for CentOS/RHEL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6418ae00c9267f363a17f69d5fb92adaa9c08a8e"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-06-08 02:23:27.000000000","tz":480},"committer":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2020-06-13 01:54:56.000000000","tz":0},"subject":"Handle network deltas when delete member","message":"Handle network deltas when delete member\n\nThe patch does the below two things:\n1. When removing members, the nic of amphora ns\n   associated with member will be deleted. \n2. the naming method of the nic in amphora ns\n   will be changed to avoid name conflicts\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ccdc038766b2cfd04c91e00eb17c5202e9164696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ccdc038766b2cfd04c91e00eb17c5202e9164696"}]},"branch":"refs/heads/master"},"3a22744f8cc69d1a0156d8aca8966ea21916b466":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-09 14:40:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/02/734002/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/4"}}},"commit":{"parents":[{"commit":"9a732565e94e1d34067b3b737131a163be0da370","subject":"Merge \"Refactoring amphora stats driver interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a732565e94e1d34067b3b737131a163be0da370"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-06-08 02:23:27.000000000","tz":480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-09 14:40:09.000000000","tz":0},"subject":"Handle network deltas when delete member","message":"Handle network deltas when delete member\n\nThe patch does the below two things:\n1. When removing members, the nic of amphora ns\n   associated with member will be deleted. \n2. the naming method of the nic in amphora ns\n   will be changed to avoid name conflicts\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a22744f8cc69d1a0156d8aca8966ea21916b466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a22744f8cc69d1a0156d8aca8966ea21916b466"}]},"branch":"refs/heads/master"},"ea74ab2369e912df10642bd004f6c356c2a979f9":{"kind":"REWORK","_number":5,"created":"2020-09-27 08:55:47.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/02/734002/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/5"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 08:55:36.000000000","tz":480},"committer":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 08:55:36.000000000","tz":480},"subject":"Handle network deltas when delete member","message":"Handle network deltas when delete member\n\nThe patch does the below two things:\n1. When removing members, the nic of amphora ns\n   associated with member will be deleted. \n2. the naming method of the nic in amphora ns\n   will be changed to avoid name conflicts\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea74ab2369e912df10642bd004f6c356c2a979f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea74ab2369e912df10642bd004f6c356c2a979f9"}]},"branch":"refs/heads/master"},"00e0a1c90312794dbd79667aaecc5647da08f188":{"kind":"REWORK","_number":6,"created":"2020-09-27 09:25:48.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/02/734002/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/6"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 08:55:36.000000000","tz":480},"committer":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 08:58:32.000000000","tz":480},"subject":"Also handle network deltas when deleting members","message":"Also handle network deltas when deleting members\n\nthe original code only handles network when add members,\nthis may result in some issues as I described in the below story.\nthis patch adds codes to support remove its network interfaces\nin the amphora namespace while deleting members\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00e0a1c90312794dbd79667aaecc5647da08f188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00e0a1c90312794dbd79667aaecc5647da08f188"}]},"branch":"refs/heads/master"},"c8541b3e12b96a470adccaec5450e2169e553ebf":{"kind":"REWORK","_number":7,"created":"2020-09-27 09:31:33.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/02/734002/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/7"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 08:55:36.000000000","tz":480},"committer":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 09:30:50.000000000","tz":480},"subject":"Also handle network deltas when deleting members","message":"Also handle network deltas when deleting members\n\nthe original code only handles network when add members,\nthis may result in some issues as I described in the below story.\nthis patch adds codes to support remove its network interfaces\nin the amphora namespace while deleting members\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8541b3e12b96a470adccaec5450e2169e553ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8541b3e12b96a470adccaec5450e2169e553ebf"}]},"branch":"refs/heads/master"},"8ba2561e77c443c23b0af1d2d95a01863d47a400":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-11-03 00:59:45.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/02/734002/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/8"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 08:55:36.000000000","tz":480},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-11-03 00:59:45.000000000","tz":0},"subject":"Also handle network deltas when deleting members","message":"Also handle network deltas when deleting members\n\nthe original code only handles network when add members,\nthis may result in some issues as I described in the below story.\nthis patch adds codes to support remove its network interfaces\nin the amphora namespace while deleting members\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ba2561e77c443c23b0af1d2d95a01863d47a400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ba2561e77c443c23b0af1d2d95a01863d47a400"}]},"branch":"refs/heads/master"},"87322714a399f335a04797cd70cd8c3336dfa3d1":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-01-17 08:06:07.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/02/734002/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/734002/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/734002/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/734002/9"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 08:55:36.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-01-17 08:06:07.000000000","tz":0},"subject":"Also handle network deltas when deleting members","message":"Also handle network deltas when deleting members\n\nthe original code only handles network when add members,\nthis may result in some issues as I described in the below story.\nthis patch adds codes to support remove its network interfaces\nin the amphora namespace while deleting members\n\nChange-Id: I9348cfc8e87351d94c225f07533b7676dcbe50e3\nStory: #2007762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87322714a399f335a04797cd70cd8c3336dfa3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87322714a399f335a04797cd70cd8c3336dfa3d1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
