)]}'
{"id":"openstack%2Foctavia~735053","triplet_id":"openstack%2Foctavia~master~Ibb7f0f416a013b98edf72a5803aada71015cfade","project":"openstack/octavia","branch":"master","topic":"typo","hashtags":[],"change_id":"Ibb7f0f416a013b98edf72a5803aada71015cfade","subject":"Fix some typos in the explanatory notes","status":"MERGED","created":"2020-06-11 06:05:27.000000000","updated":"2020-06-11 22:06:14.000000000","submitted":"2020-06-11 22:02:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735053-1591912969635-1e1c92d6","meta_rev_id":"df9605240b06cb62abc8bcf5ca33cd657acd9fd4","_number":735053,"virtual_id_number":735053,"owner":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-06-11 06:58:38.000000000","_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 22:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-06-11 18:28:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"value":1,"date":"2020-06-11 13:02:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-06-11 17:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-06-11 18:28:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 13:02:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-11 17:45:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-11 18:28:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-11 22:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f52e97e7472275589cf8e03db4bc3ffdcb631008","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-06-11 06:05:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"112238256ccb42c546b2c895082c8d868f8f4ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 06:40:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbccb2ee83534f1d9e490d6401f69ae1 : SUCCESS in 12m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d50d5d7f83b2491ca7a266ff53acf591 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f3fa9389dc6428c913a377f78a1a36b : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d4773a1174f4e2685936bbe3a7a9d59 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c1128456d9f4096b465872c1638c4b0 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323d8c332791480185b6798fb45a18de : SUCCESS in 18m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/07aa1157727a48de91323046a9228b6c : SUCCESS in 8m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9eba9ae2e6bc41d288ba99f37ef98e0d : SUCCESS in 11m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b631ec979524635ac9f65f29c3aa5ee : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d9fccdd55c77d7066809c9fc9103c1dfb9eee9ca","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-06-11 06:55:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f35f78b50b9688cece0a18077c7792e91e26a5a2","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-06-11 06:58:38.000000000","message":"Patch Set 2:\n\nRecheck, thanks a lot!","accounts_in_message":[],"_revision_number":2},{"id":"a87eba630913147e56188b36062d4ee7dfbf413b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 07:26:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12678e1b73bb43378830df66c99e4e38 : SUCCESS in 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b96c8f1edf34bc6a6707a3244ba2e07 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3907a0857928439a85a45616839bbb10 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8472189e3ddf4afbabf5de115bbbe9a0 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/980db9c7f16f406290972646fb420dba : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db89dfb99807456aa2f46d0744cf411c : SUCCESS in 14m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b6355ff22d9b4d09b282f96278e6a4b9 : SUCCESS in 8m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0f102fdf0f5f4ff5a3e03366b8d26de2 : SUCCESS in 9m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eda5658f6e9d468aa8bda8accbc18e33 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":2},{"id":"231a1c0c68d7a433e6b53c7c205ce14f028965ce","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-11 13:02:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6c83a889635cddc37a2cc1daab92daba81bebb74","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 17:45:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"85c5b7a98f2597e89ac91e31b21dc3f138c1d339","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-11 18:28:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8db506210eac9ca4161a1b33a1889f857f80066d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 18:28:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d0e6ed4511da54909ed9d85b46f8b92fa435d9ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:02:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2480dbb5696c425798f121397fc41f56 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d9cf7655354f38b8a45a0d955cd5ae : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/690ed5044b33414e8b81f95581955431 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/282706b1517749158c047fdd987a394c : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73add3e0901749d580836b7fed91c47a : SUCCESS in 11m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38216d023e324838b51174704f8a84cc : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38779517fc0a4d5dba3bb64851061e6f : SUCCESS in 46m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a92a02b6489422aa9826e8cafdb87c4 : SUCCESS in 1h 06m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9b307ca643648628be7df5a3bf8559a : SUCCESS in 58m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/636388b4f7414baeb8cabdc9534ba708 : SUCCESS in 51m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca2e06f2d4dd4a4faaf764cc3595028e : SUCCESS in 1h 03m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d509cf56c9847aba0c312ebe3d6bd7d : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2ce32dea236f4d1ab2950f5e24d54bdeafa48949","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:02:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"100a5a247f23eacc1a80cfcbc14a9be5d5365e0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:06:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5886bc656ff45928be0c2ebc0e9e4ca : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"76616f35e724bc361da2e4b48799378a38c0e6a3","revisions":{"0e00119b098af6dd8598014477fe1e0d8fd85e66":{"kind":"REWORK","_number":1,"created":"2020-06-11 06:05:27.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/53/735053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/735053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/735053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/735053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/735053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/735053/1"}}},"commit":{"parents":[{"commit":"aee9cd6fe4dc1a407ae696ab7037b6a414c8a0da","subject":"Merge \"Use unittest.mock instead of mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aee9cd6fe4dc1a407ae696ab7037b6a414c8a0da"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-06-11 05:56:53.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-06-11 05:56:53.000000000","tz":480},"subject":"Fix a typo in the explanatory notes","message":"Fix a typo in the explanatory notes\n\nThis patch changes \u0027defiend\u0027 to\n\u0027defined\u0027 at the 193 line in octavia/\ntests/functional/db/test_repositories.py\n\nChange-Id: Ibb7f0f416a013b98edf72a5803aada71015cfade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e00119b098af6dd8598014477fe1e0d8fd85e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e00119b098af6dd8598014477fe1e0d8fd85e66"}]},"branch":"refs/heads/master"},"76616f35e724bc361da2e4b48799378a38c0e6a3":{"kind":"REWORK","_number":2,"created":"2020-06-11 06:55:32.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/53/735053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/735053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/735053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/735053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/735053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/735053/2"}}},"commit":{"parents":[{"commit":"aee9cd6fe4dc1a407ae696ab7037b6a414c8a0da","subject":"Merge \"Use unittest.mock instead of mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aee9cd6fe4dc1a407ae696ab7037b6a414c8a0da"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-06-11 05:56:53.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-06-11 06:53:39.000000000","tz":480},"subject":"Fix some typos in the explanatory notes","message":"Fix some typos in the explanatory notes\n\nThis patch changes \u0027defiend\u0027 to \u0027defined\u0027\nin the explanatory notes in octavia/\ntests/functional/db/test_repositories.py\n\nChange-Id: Ibb7f0f416a013b98edf72a5803aada71015cfade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76616f35e724bc361da2e4b48799378a38c0e6a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76616f35e724bc361da2e4b48799378a38c0e6a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
