)]}'
{"id":"openstack%2Foctavia~735200","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Iaa1b5b8057ae1faa63a2ad3783ec6237284a3a84","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Iaa1b5b8057ae1faa63a2ad3783ec6237284a3a84","subject":"Add DIB to required projects for stable/train","status":"ABANDONED","created":"2020-06-11 17:23:19.000000000","updated":"2020-06-26 16:49:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4b9c66f8364d303c19f473990430b93e6f436746","_number":735200,"virtual_id_number":735200,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-12 00:03:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 00:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d59f9f471cb11ded91048c74f33baec71f011837","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 17:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d67be8ee49086414e732a92489ac9be0ec5deabf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 20:01:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a59e52b32e0bc50cc2a9f931fc4e4f99621e5ec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 21:23:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8618c48a5d608718a64f5e9373c0f477fcb4f6ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 00:03:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2c147db5f144a6c871306445fea4cf7 : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1e22e46d9f84b2c9804c97941fc6097 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f073d474f064545a8fdceafd9532ed4 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35aedcd26bb2467c976e2555310df272 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86c3f40fc2484fa1bcfaded6686dc3da : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be1567101d764352a0f5cc6770b2da8e : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba4b3fc1ab84b8db1270b7201562dc3 : SUCCESS in 14m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/08294e737eac43d48cb1f579da397025 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/baa899cf9aea40e1aad42ed1a0f56c49 : SUCCESS in 10m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0bc7f25341b74317bdccc74fcd089f13 : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93b38f0201d7468d93405042f9fc3df9 : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a1d2706b0e541a1999a9b73a20540fe : SUCCESS in 45m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/45e61513b29a4bb59be0341ce263cb4e : SUCCESS in 51m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/824f8b222eb04b709f6eebbf81b4f1ea : SUCCESS in 1h 56m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f400fec5d6f44f33b27545d827dc1ced : SUCCESS in 1h 58m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c3c4b54aa5234bc0a3459b1284f09fdf : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d498f587ef9c46c796e5f489c35d7cc8 : SUCCESS in 2h 03m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0dd0620835504019bb4dc5b1bdd7d6e1 : SUCCESS in 2h 17m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb26a4d97aa64447905dd0514975f6a6 : FAILURE in 25m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2b9699c9023428ba69fe3dca9cc9160 : SUCCESS in 43m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7531aeccf3f848ac8c89b1fcad6fce6d : SUCCESS in 1h 06m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8935ed48d23d423397797abbde3ba232 : SUCCESS in 52m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1592a026b47a4e93bb7dcfaf21424ece : SUCCESS in 1h 03m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ae30a2a08f374bc9a0bff3ecf45fa688 : SUCCESS in 1h 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b9c66f8364d303c19f473990430b93e6f436746","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 16:49:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cda46257a61c97f90c8039ae3e7601161a32c150","revisions":{"16a749b0611b41e0c0c04fcf6a9ca8671703cce8":{"kind":"REWORK","_number":1,"created":"2020-06-11 17:23:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/00/735200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/735200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/735200/1"}}},"commit":{"parents":[{"commit":"e358637a10e6b3c2df336d8dad42cecbd97e3dd5","subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e358637a10e6b3c2df336d8dad42cecbd97e3dd5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 17:21:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 17:23:18.000000000","tz":-420},"subject":"Add DIB to required projects for stable/train","message":"Add DIB to required projects for stable/train\n\nA previous patch[1] changed how diskimage-builder is installed for\nstable/train. However the grenade job was missing diskimage-builder\nin the required projects definition. This patch corrects that.\n\n[1] https://review.opendev.org/733684\n\nChange-Id: Iaa1b5b8057ae1faa63a2ad3783ec6237284a3a84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16a749b0611b41e0c0c04fcf6a9ca8671703cce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16a749b0611b41e0c0c04fcf6a9ca8671703cce8"}]},"branch":"refs/heads/stable/train"},"9108faca1bcba9fcee0ced22bc319a07e6430de7":{"kind":"REWORK","_number":2,"created":"2020-06-11 20:01:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/00/735200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/735200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/735200/2"}}},"commit":{"parents":[{"commit":"e358637a10e6b3c2df336d8dad42cecbd97e3dd5","subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e358637a10e6b3c2df336d8dad42cecbd97e3dd5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 17:21:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 20:01:32.000000000","tz":-420},"subject":"Add DIB to required projects for stable/train","message":"Add DIB to required projects for stable/train\n\nA previous patch[1] changed how diskimage-builder is installed for\nstable/train. However the grenade job was missing diskimage-builder\nin the required projects definition. This patch corrects that.\n\n[1] https://review.opendev.org/733681\n\nChange-Id: Iaa1b5b8057ae1faa63a2ad3783ec6237284a3a84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9108faca1bcba9fcee0ced22bc319a07e6430de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9108faca1bcba9fcee0ced22bc319a07e6430de7"}]},"branch":"refs/heads/stable/train"},"cda46257a61c97f90c8039ae3e7601161a32c150":{"kind":"REWORK","_number":3,"created":"2020-06-11 21:23:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/00/735200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/735200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/735200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/735200/3"}}},"commit":{"parents":[{"commit":"e358637a10e6b3c2df336d8dad42cecbd97e3dd5","subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e358637a10e6b3c2df336d8dad42cecbd97e3dd5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 17:21:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 21:23:45.000000000","tz":-420},"subject":"Add DIB to required projects for stable/train","message":"Add DIB to required projects for stable/train\n\nA previous patch[1] changed how diskimage-builder is installed for\nstable/train. However the grenade job was missing diskimage-builder\nin the required projects definition. This patch corrects that.\n\n[1] https://review.opendev.org/733681\n\nChange-Id: Iaa1b5b8057ae1faa63a2ad3783ec6237284a3a84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cda46257a61c97f90c8039ae3e7601161a32c150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cda46257a61c97f90c8039ae3e7601161a32c150"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
