)]}'
{"id":"openstack%2Foctavia~735510","triplet_id":"openstack%2Foctavia~master~Ic782faba3eb907d29b6735ac0a6d6a8a2e104e00","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ic782faba3eb907d29b6735ac0a6d6a8a2e104e00","subject":"Fix error on devstack cleanup","status":"MERGED","created":"2020-06-15 06:50:15.000000000","updated":"2020-06-26 21:41:08.000000000","submitted":"2020-06-26 21:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735510-1593207522685-5872017d","meta_rev_id":"85f6000f7739cc601cb361ec1424d80e83872bf3","_number":735510,"virtual_id_number":735510,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-06-22 14:11:38.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 21:38:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-06-26 19:30:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-22 16:20:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-06-26 19:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-06-26 19:30:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 16:20:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-26 19:10:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-26 19:30:48.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-26 21:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9e93d72f6e8d64258722d8666a5c4f740c0bb12","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-15 06:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6c78499394dc99978564f327302ed8d15711d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 07:21:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8fd2e0584ba4c33b92d799dc67fa48d : SUCCESS in 12m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb8d676a42da4cfd9c43eff43702b546 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34739d8c85124c6fbc0a6c29de3b42e3 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46906227f6ba456aa979a62505ce5f83 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64de2ba4fa854516b66b8a1024de0f35 : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62259132ab8c46f3a11c9900972d9500 : SUCCESS in 15m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c02550953c0f48a7a242687fe749a4ed : SUCCESS in 8m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/04dc56b4981b4de8bb79b6f1d26ddc8d : SUCCESS in 10m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/572ae470ae9c48aea08c630c2e893dd1 : SUCCESS in 10m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2465829ebef14f6aade5c26596c85649 : FAILURE in 14m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb4ab2a3a4e846d9a2539c81b062d995 : FAILURE in 20m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bfb4f365678408d8f3c105afe914c1c : FAILURE in 14m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0304a1c39514433fac07c29cf377858a : FAILURE in 14m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50027f741b354a9d94dc0434bc7ad647 : FAILURE in 14m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e94d241106ca42aca1cd8c8b6cfaa669 : FAILURE in 17m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45214518e2f94a9abb04419d0120eda6 : FAILURE in 15m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe6719b0ff95416d8a4e1d937e8276a0 : FAILURE in 19m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/574a0d133e6542628fe6bcf88204ed64 : FAILURE in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"427d15d356177abc258d4f2fd0c8ea0501b7ab88","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-16 13:58:19.000000000","message":"Patch Set 1:\n\nrecheck uwsgi update","accounts_in_message":[],"_revision_number":1},{"id":"09a10aa598ffa3479c59b4f0785dfd69e2d4b818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 15:09:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01fec8c2d1eb4ae5a097133e3872b83f : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd2b92033783436d8b51b11a05b87ac3 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08db101289eb418585be7a35cdca18cb : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62809abd571b4e99aa0807da1f9b8a80 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ad2715903804b1b9fed3e8868010459 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1bb66bdccd4a38898a9ba9653894b2 : SUCCESS in 15m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/21c9da94b4d74f42926915a9f027a01c : SUCCESS in 8m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/290117fb1f33462bb1bf7b05c66a2e51 : SUCCESS in 9m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58a0c276e9b54088948a0a795ddcd5b7 : SUCCESS in 9m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/33a6f6fb2bca404a9673b2f59329d57b : FAILURE in 32m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd93fe4c7e104edfbdf716b927b3ed01 : FAILURE in 41m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2364c103f2d648c08898b8b7c759c772 : SUCCESS in 1h 05m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/18315d379fc04fb3a2818b4b7005db3c : FAILURE in 39m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b24131796be9421d8ab1f187a14a17a6 : FAILURE in 36m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d8bcfb3ddc84a0e9150985e726bd6ae : FAILURE in 18m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8adbfebe32094787a45b48b7a2d0a964 : FAILURE in 44m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72210e2e214541b39df92ece2a6760c9 : FAILURE in 40m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/af373539679742fd9da1e1182e513d8c : FAILURE in 49m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f67e5a27d7297dc8002cc3ce63a31d3523d53347","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-22 14:11:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14c6773b77ab1bc5ef5a25fdf196d835f7320b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:11:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3be6f247ed944019949a644c977d7573 : SUCCESS in 11m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2746d93b8e6749698323b69aa026364f : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c6a7387ec3447eba5ca5ec5bbaee185 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9c187ee12c74646935e91c3e0c4e28d : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c049a6e437e4287b2ca546c2189eb86 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea66fb2270a54c4bb5301a8c0693cea7 : SUCCESS in 14m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1157ca6f2e3c4b3494dd8e93c473eff4 : SUCCESS in 8m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a36b9d991a444114822ca38633208109 : SUCCESS in 11m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/01f05792a78243f5a6112111378b36be : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a32811948cb84279abdf490bea1fdbdd : SUCCESS in 44m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e218c41f970a48dd811ad16b35b608a0 : SUCCESS in 1h 38m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e54feca8e8894e89b9a9feec67ed903c : SUCCESS in 1h 00m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2770a4027d274ab5aa6bf00b933284a3 : SUCCESS in 56m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63b401d82d774b2ea214a7d5b4fd0c73 : SUCCESS in 1h 06m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da8897dc894f4bf48eada23f17fbbebd : SUCCESS in 1h 34m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e765b93eee4447879e44b0ccd46109d2 : SUCCESS in 1h 52m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1d297b5ad96f4462a1098fc0e072cd71 : SUCCESS in 1h 45m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/55026f4e683a413a90f46107609b6e31 : FAILURE in 1h 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab897d0f74d32bc5baed79863caa0d35200e2a19","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-22 16:20:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d1e286c3951a4f5aef4071273229b85e6b3ef05f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 19:10:59.000000000","message":"Patch Set 1: Code-Review+2\n\nOk, so I guess the other approach still has issues on an instance reboot, so let\u0027s go with this. I\u0027m not sure the keypair delete really matters as I think nova cleans those out on restack anyway.","accounts_in_message":[],"_revision_number":1},{"id":"c018346e9e46845502d47a1cb86d2405ed7160b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 19:30:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ae28360c63c155590490e6ea3463c1f791c7602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:31:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3301af4e6d91ef5d06adbba89c948db5136c59a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:38:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"367bd2694c0fe767ffc2f9336c0ee7362940a3b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:38:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a154c0d0d32e453cb728705d04976b60 : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/444a7715914f4b7f8e0d7e8212b17453 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19cfc53f60ca47bb8cfb3fd5a724c258 : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e78f731915c0412c9d3bb49f310ce42d : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60e794116754ca0b5be98ad05f92a26 : SUCCESS in 13m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4910de57e00a41fc9ad0412e5e0c1b7f : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6a3244b7d764702be737a6b7c589e96 : SUCCESS in 49m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efc9b7ccb5124185885688b8c884559b : SUCCESS in 1h 35m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d078bac662e412595e94a02bb6218a9 : SUCCESS in 55m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/32bb0d9c0a894bcd8331f063ee1a18cf : SUCCESS in 51m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8bb71b90d2214d6dbeb74804b2537164 : SUCCESS in 1h 12m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d59d783c167f4d46b8bcc5204bde9dee : SUCCESS in 2h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"85f6000f7739cc601cb361ec1424d80e83872bf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:41:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab266d25100449ebd51d50d6f9c6a4d : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c176cf60d46f5f5f743d457d3c941902c1bcfad7","revisions":{"c176cf60d46f5f5f743d457d3c941902c1bcfad7":{"kind":"REWORK","_number":1,"created":"2020-06-15 06:50:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/735510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/735510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/735510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/735510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/735510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/735510/1"}}},"commit":{"parents":[{"commit":"aee9cd6fe4dc1a407ae696ab7037b6a414c8a0da","subject":"Merge \"Use unittest.mock instead of mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aee9cd6fe4dc1a407ae696ab7037b6a414c8a0da"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-15 06:43:03.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-15 06:44:32.000000000","tz":120},"subject":"Fix error on devstack cleanup","message":"Fix error on devstack cleanup\n\nChange I7ebf4137feb04827490dffc0dac3d6e4c8888075 added \u0027set -e\u0027 in\ndevstack/plugin.sh, but on devstack cleanup, some commands may fail\nbecause of non-working services (i.e after a reboot).\nThis commit allows \u0027openstack keypair delete\u0027 to fail on clean up.\n\nChange-Id: Ic782faba3eb907d29b6735ac0a6d6a8a2e104e00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c176cf60d46f5f5f743d457d3c941902c1bcfad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c176cf60d46f5f5f743d457d3c941902c1bcfad7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
