)]}'
{"id":"openstack%2Foctavia~735914","triplet_id":"openstack%2Foctavia~master~I7fbf063a5b49eeb556d9dc544d3853c34b235c03","project":"openstack/octavia","branch":"master","topic":"healthcheck","hashtags":[],"change_id":"I7fbf063a5b49eeb556d9dc544d3853c34b235c03","subject":"Add a /healthcheck URL from oslo.middleware","status":"ABANDONED","created":"2020-06-16 13:34:24.000000000","updated":"2020-07-02 09:21:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa0c8f21257d7d30f32f5f117f49deb63e47ffe2","_number":735914,"virtual_id_number":735914,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-16 14:52:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-16 13:48:32.000000000","_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2020-06-16 13:38:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-2,"date":"2020-06-16 15:03:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 13:38:39.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-16 14:52:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-16 15:03:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"983fde9f48b8bbfae471721bad7e49a29917ae3e","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-06-16 13:34:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d5546dacd4f76897535886d8016df04984177cc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-16 13:38:39.000000000","message":"Patch Set 1: Code-Review-1\n\nHasn\u0027t this been already implemented? Am I missing something? See https://review.opendev.org/#/c/711127/","accounts_in_message":[],"_revision_number":1},{"id":"83e0a14473c98a832862c8a011c31a2fc351a825","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-06-16 13:48:32.000000000","message":"Patch Set 1:\n\nAt least for me, it wasn\u0027t there in Ussuri. Is this something that was done AFTER ussuri?","accounts_in_message":[],"_revision_number":1},{"id":"8385c90acfa23e911030952dbe1f6fabae01f578","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-16 14:08:16.000000000","message":"Patch Set 1:\n\nIt\u0027s included in Ussuri. Make sure you enable it -- see https://docs.openstack.org/octavia/latest/admin/healthcheck.html","accounts_in_message":[],"_revision_number":1},{"id":"7084e338815dbd865feea4f44e04a8220e04fc4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 14:52:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4410c708d6e94f4694598285e96bf64d : FAILURE in 12m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c18a63fd6434f67886930edd4f9d8df : FAILURE in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8d6af7fc9a455f94ad16fe58d5668b : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6e259095544aa489e38706106a9850 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91540518a00e4bb9aa3bfe0b4cdc4d7d : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eaf0edebcde4839bcbd7907724a7300 : SUCCESS in 15m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/693ad8ca3559498aa80b17f9220c0370 : SUCCESS in 11m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2d69523d24e14dfda9240077ad4d500f : FAILURE in 12m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3b6636e7b4b4f78be12840440aabb80 : FAILURE in 12m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f210d0cc3e04242aa769529c0dfe38f : FAILURE in 22m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b905485d62f4f9989234107e21dfb8a : FAILURE in 36m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b474b41f8c04129a120d0c25e906e66 : SUCCESS in 1h 05m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e9f576553ea4b12821aac051030419f : FAILURE in 39m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bda41df35c8747159f5c1174d1de3e4a : FAILURE in 43m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9443ede5d5e94a7e944180cead675c32 : FAILURE in 16m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7e029c7cee5d40beaab2f94599cb85c1 : FAILURE in 28m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a54bb212e1664c749e4c9fe2a1c07590 : FAILURE in 36m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/35b08a88e5e54001b69dd5ace664a8a7 : FAILURE in 42m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd3b1f1f13d9c8bccf72ec4e9a561a8ada1c62ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-16 15:03:37.000000000","message":"Patch Set 1: Code-Review-2\n\nAs was already mentioned, this was already added to the Octavia API here: https://review.opendev.org/#/c/711127/ with documentation here: https://docs.openstack.org/octavia/latest/admin/healthcheck.html\n\nWe did not simply wrap the app as that would only test the wsgi framework (uWSGI in this case) and would not actually tell if the API was functional. Also, since this is an un-authenticated endpoint, it poses a slight security risk (it should have rate limiting in front of it, etc.), so we make it an opt-in feature.","accounts_in_message":[],"_revision_number":1},{"id":"d568727048250932817083ba02a584b9408a85ed","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-07-02 09:21:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b778517dbebcf7481e070784938531e74a88d2e","revisions":{"1b778517dbebcf7481e070784938531e74a88d2e":{"kind":"REWORK","_number":1,"created":"2020-06-16 13:34:24.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/735914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/735914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/735914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/735914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/735914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/735914/1"}}},"commit":{"parents":[{"commit":"24acbe099bacacbe1840d8964bd32309d47a7081","subject":"Merge \"Improve terminology in an old spec file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24acbe099bacacbe1840d8964bd32309d47a7081"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-06-16 13:33:13.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-06-16 13:33:13.000000000","tz":120},"subject":"Add a /healthcheck URL from oslo.middleware","message":"Add a /healthcheck URL from oslo.middleware\n\nThis patch adds a /healtcheck URL so that operators can configure haproxy\nin a nicer way for Octavia.\n\nChange-Id: I7fbf063a5b49eeb556d9dc544d3853c34b235c03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b778517dbebcf7481e070784938531e74a88d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b778517dbebcf7481e070784938531e74a88d2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
