)]}'
{"id":"openstack%2Foctavia~736137","triplet_id":"openstack%2Foctavia~master~I5e6aee49f434820881051874c9ad2628b4fcada7","project":"openstack/octavia","branch":"master","topic":"uwsgi","hashtags":[],"change_id":"I5e6aee49f434820881051874c9ad2628b4fcada7","subject":"Use uwsgi binary from path","status":"MERGED","created":"2020-06-17 09:04:34.000000000","updated":"2020-06-17 16:06:55.000000000","submitted":"2020-06-17 16:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736137-1592409759405-ff50f346","meta_rev_id":"b6c5f0cdef22f63ba3cf8fdbb3020b7b9757517d","_number":736137,"virtual_id_number":736137,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 16:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-06-17 10:43:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-06-17 14:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-06-17 14:03:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 10:43:50.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-06-17 14:03:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-17 16:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4719e855538c1d958e48ae00f221c8185f980228","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-17 09:04:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fb936b4d069f0f0101e5af89e88b5a68040789b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-17 10:43:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3915a2bd9bc1a4fc4f048f5049adf06af10bd73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:13:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/028495e1e7f049f4a3c15fca1069edfb : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c33234316e7403f9f83c301837212b6 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820d41c3dda4462da7691d4e5b428a70 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dd664554ec143bbacd040e8695dbed5 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/007da67631e9436e8985280675b725d2 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2fdc1b88cb49788aa98d447a095ece : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b0d90692237d48599e372c1b69b1947d : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3651178fa2d6408ba55d0d3e28ec4173 : SUCCESS in 8m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed49081044c042d8a5d4b37b10932ebc : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d1c7e44facb4da7ab8c83ffc21f7098 : SUCCESS in 41m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/314d2e6de4384bcbba1117914513e3a5 : SUCCESS in 58m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6617cc389a7f458cbb0c8162fc7cf38f : SUCCESS in 48m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9149816a133549f2ba93568c388e2bed : SUCCESS in 56m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5644c9c3f3741959eda53a7275cb77f : SUCCESS in 1h 04m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b7162766bf84e6db8b0957486053a8a : SUCCESS in 1h 31m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b1a933c036f04612a523bf9d21445dc5 : SUCCESS in 1h 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bba4a6c9e5f240e8a0d28680b8d9bc02 : SUCCESS in 2h 01m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9b7579ca896c4cc1aa74aa445088b069 : SUCCESS in 1h 02m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c71232c74be2fbc4c6cf460baf1443cfcfb46cd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-17 14:03:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nForce merging this as it is a needed gate fix.","accounts_in_message":[],"_revision_number":1},{"id":"8367e4b495514f4c6fd295f288aad1d98b513a1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 14:04:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f2ce9808a9846bd356f7d4b43ccd663ef41e4f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:02:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85695f303bb54a0f8c863d7be7e77857 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a111297fe4403d942b8e3a81ff4ffa : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e470266f4e2493eae0d324667196e8d : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88c240835278468993509035501547fb : SUCCESS in 13m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ee70b146bc412591c3a67326f171f0 : SUCCESS in 15m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6597671231e45a689fa32e7c1282586 : SUCCESS in 9m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/438db0112d43454f96db9a51d65932ec : SUCCESS in 58m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31c1bc583e0f4055bbe4289827858a3c : SUCCESS in 1h 15m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59b4bd15c4da44138f5a18bf818d6c6a : SUCCESS in 1h 11m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/91a67d63b1d344b7835e496d4f6f4a47 : SUCCESS in 1h 01m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b22a77eddbe45eb8ce12b155444c780 : SUCCESS in 1h 20m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ba0a566354cd427a9f7abb8c44708646 : SUCCESS in 1h 45m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c3396a8b151878611b7f9e64ad81f95a701b0df5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:02:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"40decfc3d7d29d2e38ffa5608aa6adc3f9319e63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:06:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89e982ad0174c2e8d4e25b43a54a8b6 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5951ced5fe51262a57bf6a9717e75b8b6c63ca5","revisions":{"e5951ced5fe51262a57bf6a9717e75b8b6c63ca5":{"kind":"REWORK","_number":1,"created":"2020-06-17 09:04:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/37/736137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/736137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/736137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/736137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/736137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/736137/1"}}},"commit":{"parents":[{"commit":"24acbe099bacacbe1840d8964bd32309d47a7081","subject":"Merge \"Improve terminology in an old spec file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24acbe099bacacbe1840d8964bd32309d47a7081"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-17 08:59:29.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-17 08:59:29.000000000","tz":120},"subject":"Use uwsgi binary from path","message":"Use uwsgi binary from path\n\nIn-line with devstack patch [1], switch invocations to find uwsgi in the\npath.\n\n[1] https://review.opendev.org/#/c/577779/\n\nChange-Id: I5e6aee49f434820881051874c9ad2628b4fcada7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5951ced5fe51262a57bf6a9717e75b8b6c63ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5951ced5fe51262a57bf6a9717e75b8b6c63ca5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
