)]}'
{"id":"openstack%2Foctavia~736696","triplet_id":"openstack%2Foctavia~master~I6f372a3f82fdf4f41e661e640e4a983cf484ed6d","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6f372a3f82fdf4f41e661e640e4a983cf484ed6d","subject":"Fix listener API\u0027s test_create* assertions","status":"MERGED","created":"2020-06-18 12:18:43.000000000","updated":"2020-06-19 10:01:02.000000000","submitted":"2020-06-19 09:58:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736696-1592560733170-eae5d268","meta_rev_id":"2944999c896134098473982d7df851126e679d73","_number":736696,"virtual_id_number":736696,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 09:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-06-18 13:59:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-18 14:02:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-06-18 21:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-06-18 21:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-06-18 13:59:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 13:59:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-18 14:02:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-18 21:51:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-19 09:58:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4ffc177ecbccc6cae838c5748fb169e645ffbdb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-18 12:18:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ce45425c8f834e0dfa8bf60ac53bea7e68075e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab8f044e04934a77ba265fedb89844df : SUCCESS in 10m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/101d0bb2174849caae8df7e77926c74c : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb4da272c6d34b2bb68436bef7b752e3 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b7a3962d65d41aea55972590bfe7f68 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/952fff0ed29542c1a288acf1e3b52728 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae00e1ef64241848a030b3b6dfe8f96 : SUCCESS in 17m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4dcd274d14154583b2d1a766aa789f6e : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e710e5260b624b949976dabcdd9d8865 : SUCCESS in 11m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4cab0c2f55c4474588f16e1a6c499641 : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8e146767421e8b0e7fb704eb7af1262ce927cd67","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-18 13:59:50.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\nNice catch!","accounts_in_message":[],"_revision_number":1},{"id":"5837e5942315bd3689a3297e7a63346822df4b31","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-18 14:02:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a21439d5f8595d2e94e188a53b0102a1ca2221b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-18 21:51:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"1a18032d1f3ce011ffc99fb37a5636d078f15a3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 21:51:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f119ae9fd7928b3efa0e6dfdc3631bf04b83e8d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:49:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae3259eb21134505b411e1a333891d1a : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93de1df2aa3c482f9b517168042b0366 : SUCCESS in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3b2171d25f74e6f97c5169cb33828c0 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15ef1ce3640d4070ab80206586330676 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643b31ce0f264cba93bd4f97a4f11973 : SUCCESS in 16m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/067addc280d94682a95057bbd9e8d8e1 : SUCCESS in 10m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0969852f38244ec0ab72710fe1c22854 : SUCCESS in 46m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/199c3762d4a0467c9ba73bd5889d41f2 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0e533348e0614ed88b4a0774dc5ec159 : SUCCESS in 35m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fc94a061676c4855b365315395d41c68 : SUCCESS in 48m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e5bfd570de94c0f80e6c36ccc1f9545 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c09bd2fce844286b591f2d3bc7c1e91 : FAILURE in 50m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f32f41158243f7b23fa28fdc86e84127ea3ee54c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 07:50:06.000000000","message":"Patch Set 1:\n\nrecheck\n\noctavia.common.exceptions.ComputeBuildException: Failed to build compute instance due to: No Glance images are tagged with amphora tag.","accounts_in_message":[],"_revision_number":1},{"id":"513bb99d596706254b1ca528ef3677b29f94d20b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/478113100bce40cd8525444ac30c7e2f : SUCCESS in 13m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7029232a96d45f1b7d382499dce4bb8 : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b358b8637727495a994cf14f0b145617 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9848dd5810354936b1e0229abf2fffa7 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0b6a29c584d44a3bd6c1e17a80dc267 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0d1c43aeab4a7491929e0befce6915 : SUCCESS in 12m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0da34cc9c15042259373ee4bf8aa12a5 : SUCCESS in 11m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2fed1255ac164a848b0affeda62cce7b : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af93a46c4e0d475980ef983598c9b11e : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"223c43b332fe8185f67b46fcfbcde9dfd41842de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:05:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34216c67cf5b1a7ebd8698cdaffbfdeeb331e423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:58:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36681d63d3fe41149ce4f4019ade3b6d : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8474be9c2647d3977675084fba0730 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35e93366e78043169756aaeb9a41c144 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fe1b57f0bb54889ae4ba24b866d18a5 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c336d7dbf68e4a52b4c99b252c15ab61 : SUCCESS in 12m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80936130a1284965befacab53800d2c3 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f62c81326c714a8fad0b20e77417ad0c : SUCCESS in 49m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5069567dea4b4f28a6e37338f96c364b : SUCCESS in 1h 47m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e75a37f622494d77a51fd60eb4bf85b2 : SUCCESS in 57m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e3b343ccbf548e2be40e69d753b5600 : SUCCESS in 58m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d0ccfd72d2046b99d713c1c63d0ad17 : SUCCESS in 1h 06m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f0c7ae953aa649f792e58b7a7897cffe : SUCCESS in 1h 50m 18s","accounts_in_message":[],"_revision_number":1},{"id":"57d8195e378ecbd9d77472cff802f00c11e8fee1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:58:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2944999c896134098473982d7df851126e679d73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:01:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5c1aa3da208498bbd3ac7254280c5fb : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ee9bc0d65e1d9d41f2079c10b3361aec6b61fcf","revisions":{"3ee9bc0d65e1d9d41f2079c10b3361aec6b61fcf":{"kind":"REWORK","_number":1,"created":"2020-06-18 12:18:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/96/736696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/736696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/736696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/736696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/736696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/736696/1"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 12:03:06.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 12:17:33.000000000","tz":120},"subject":"Fix listener API\u0027s test_create* assertions","message":"Fix listener API\u0027s test_create* assertions\n\nAssertions were using the same expressions on both side: optionals and\nlb_listener are both parameters to the API (and the lb_listener dict\ncontains all optionals items).\nThose assertions should compare the parameters to the API results.\n\nChange-Id: I6f372a3f82fdf4f41e661e640e4a983cf484ed6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ee9bc0d65e1d9d41f2079c10b3361aec6b61fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ee9bc0d65e1d9d41f2079c10b3361aec6b61fcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
