)]}'
{"id":"openstack%2Foctavia~736818","triplet_id":"openstack%2Foctavia~master~Id9e19375ebca6a720e6a85006f5e8948d3aed760","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Id9e19375ebca6a720e6a85006f5e8948d3aed760","subject":"Fix invalid DOWN status when updating a UDP pool","status":"MERGED","created":"2020-06-18 19:45:50.000000000","updated":"2020-08-26 20:16:23.000000000","submitted":"2020-08-26 20:14:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736818-1598472850349-f63fd487","meta_rev_id":"d55e5336748ed6877ce7086fa3901cadeb56d59e","_number":736818,"virtual_id_number":736818,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 20:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-08-26 09:51:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-26 19:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-08-26 14:00:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-26 19:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-08-26 09:51:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2020-08-26 08:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 13:41:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-26 09:51:34.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-26 14:00:01.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-08-26 19:04:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-26 20:14:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52cec16843ba847000f7f16999103280c856c4b8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-18 19:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ad4a55bbe6a6f33db99388177df9790f1d1f286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:12:30.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/579aa8de281643d1a76fa29683e2c52c : FAILURE in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b32b2bb5f374c2cad6b9737b441f14f : FAILURE in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429e5381764b4c82838475ab7d443e88 : FAILURE in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3402d27c89074f2693119cbcf196602f : FAILURE in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34f096d048df4b66875301390ef86435 : FAILURE in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bdd18db1ab64075aec0c144e0e4972b : SUCCESS in 9m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a439c216d4234e92af4a0c7de37e039c : FAILURE in 8m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9ddd71a8edda4ec9a482a3eb9c6ce138 : SUCCESS in 11m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39167b7887b64bbc8855f52f9ff33ae9 : SUCCESS in 10m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/106ac2d2212643ee90ec19e90998ea12 : SUCCESS in 38m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6738c5eb2e3c4b8ca9cd8a40bd9dffa6 : SUCCESS in 1h 07m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49f8f147992d44b2a9045abaff2278aa : SUCCESS in 33m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f5c4cf7e93194785b807a6f5dca4b267 : SUCCESS in 51m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d9712168ea849f0ac3e3a50e7a15374 : SUCCESS in 32m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44aca79267fb4bbbbf93b73b04841ea9 : SUCCESS in 2h 23m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2f035e21db064390a4271c6caef18b51 : SUCCESS in 1h 24m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1c477cffd8d4c1687a9eed5eae29dc9 : SUCCESS in 56m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f2deeb3cabb414ba63312e09acd2b48 : SUCCESS in 1h 09m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9fd862090e23287982f8d1942da2de82efc62087","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-19 07:33:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"206159586362accb5a3e2cf252ec943a5654003a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:42:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c071073735b493c839681c775b35a87 : SUCCESS in 13m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b75132354e374041861eff222766f3fa : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8697d5803184f2aab3051ae4221540b : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7d17c6985534c07b7207c0d0132da41 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c2ccbf7240a454f9290f60cc2905133 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6da075426b74d4d82d124ee6ceebc72 : SUCCESS in 10m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ccac008a68b3442e8fbea33219d987c2 : SUCCESS in 7m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a3da2e702fb346e58ade5d94e4fc44a1 : SUCCESS in 10m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/076afed1975c40bfbe8c0cdf408fa408 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d4def6f2d3f4884afa13728d966b88d : SUCCESS in 51m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02739e0ea74543fda43ab20d5c54d5a6 : SUCCESS in 1h 58m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/940b17f584a74c728e8f85c3c1eb012f : SUCCESS in 59m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/612a4fb65d4644698d0c73df614c9e1f : SUCCESS in 1h 00m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5139ad6d386642cca57057c54ddb5f4c : SUCCESS in 1h 23m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5dede053ff5b4b69b868f0326e28b62a : SUCCESS in 1h 15m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce9e4082c0e848128bad05797b5f1b83 : SUCCESS in 1h 56m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5abc898836e4ea89d1e1003321bee92 : SUCCESS in 2h 01m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/01409797ffd748adb5c37dadf3e395df : SUCCESS in 1h 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c9b2c356bf9619e2bf6aad6ac3156d70297a9d7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-19 11:26:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d40426b3b03cb2c6086e7e2706c9a3a3494649b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:42:12.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3954d2c6d9a4fdab6862fd3179f57ef : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e216153e5ba74f7fb2e96750258405e0 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/134ac92a12bb4f489b101784176fcd4c : FAILURE in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54c01d69e41349668f231e0ba2ab67eb : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39811794c374426ea2cc00dde748ea35 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eedc14e3654d4a16a1d852476986659f : SUCCESS in 13m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f5329679f7ba457b8e16420b5e25f1ff : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/48eb2cebda564b3ab7648664c60e6df6 : SUCCESS in 8m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e34911b09c9b4779b8cf361d8c5079a0 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/732ffb88086c44e5921667f90fad53e8 : SUCCESS in 47m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d679452952c496db9b5a9e7313fbb4d : SUCCESS in 1h 53m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86af01c1d5d842a0b03759b53394498f : SUCCESS in 1h 01m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e54bb10e1404db0922776cdb514bd9d : SUCCESS in 54m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be2f06f75b4a4e6ebeb9e3641df5d660 : SUCCESS in 1h 01m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/908c3ae1e136475395dcb7a20503b573 : SUCCESS in 1h 40m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/edc4fa246eff426380f3abbb10ff7f96 : SUCCESS in 1h 56m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8227b39ed29343d49291a21f54de4857 : FAILURE in 2h 13m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8d1875b9674f4e408be7f80732cf40ec : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"10c4b0c1b8c5cd08dedffdb827ba08c04be4013e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-19 14:13:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"41fe5583b4d185e143577887d31b9966041249fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 16:28:11.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83ff9c7738754335a808cf5cf90f3766 : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d936798d105b431182dbc866aa331c1e : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3aebe44f0a4ab48e45bc46062f89b0 : FAILURE in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/862b0431f3c44f7fbb80f26d90aa16b2 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/227a9f6eee1f4783a5344ae7c49dd178 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4379d37ff524976955c2375891d909e : SUCCESS in 10m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a3e9a9729fb4c7b9faa04bde0d35d65 : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3675378c2e6f48de96d7fdf584503e8d : SUCCESS in 12m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7902514608ab4ceabd630b126bbc4844 : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92779bf0ada840cbad17d98b679d083b : SUCCESS in 49m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99178c7ff26a41e68fc80070df8d68ee : SUCCESS in 1h 23m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b7e4cdfb3def43d4918d1006b87f5412 : SUCCESS in 57m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/64d7f3604cec453ab83f430895b4b73a : SUCCESS in 1h 04m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e58d4ad93c54f5aa125c2f5a45b550e : SUCCESS in 1h 13m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5dc06b468fa44c0c86b14a55abf6392e : SUCCESS in 1h 27m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1e525844ade2427198ac393bffbce6ca : SUCCESS in 1h 48m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f04914c4feb40b2af0aa0630a9f2b89 : SUCCESS in 1h 49m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/baa53cd8058f40db91aedf63b02510e2 : SUCCESS in 2h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16fbb7eaf1218ed19fb37efee9e5bab4eac244db","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-19 16:38:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"93572d6b780c3ff82d5673a8d36b2731e68d6b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 18:49:33.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61e7a2a7940344b1a8d44cc2fe639531 : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42f35e24d592440880398f125d54c1e5 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b718a944e9b42888fc3f9bd21d26f86 : FAILURE in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c1434eb4c724e4dbef4f15da3e3a3e8 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e78b90576ed240799127294e1d412ab8 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21d0b2bfee9415fa13c45124c2725f8 : SUCCESS in 13m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c50ccce22c2b48c380d3758cab9bf39a : SUCCESS in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c7c1b9115cf5489b85047f1b0a9932f6 : SUCCESS in 13m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/755f49f87ce444bfa60040508c6ebf22 : SUCCESS in 10m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd22b8efddf946d080df90559660056b : SUCCESS in 46m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d31dd53e70d4a858393521d81a0c58c : SUCCESS in 1h 48m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11a6d09f8b5b4982980ee0f54c339434 : SUCCESS in 1h 16m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f6c0de7ea05b4655a935117fec737f07 : SUCCESS in 55m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43fb880cc7994bfc948de0fe48a57650 : SUCCESS in 1h 13m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3abf3b63eda846e1b0334aec8c5e43a3 : SUCCESS in 1h 41m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/251e346469214308b29e9a28e8629dd3 : SUCCESS in 2h 04m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1bc494ce5614a198f3a5ea4f569f4f9 : SUCCESS in 2h 00m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e785a2fd589d48888e6da15e55fbb48a : SUCCESS in 1h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0130e9bda6241e9efb7dbaf856b755b83c600f3e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-22 06:06:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"34ebce71e0b54b21fd907633ae20182530dbc5fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 07:51:14.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e074b381cd4a619fd2ed6c4c263919 : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2580f9386134d6c98a15f7899ff8507 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d78e9cd2cd84475a862f1fbefe90b93 : FAILURE in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b12b89e8d7574866b7953650fe7f34d0 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94400bc609e94b5db829adf0b786e6a0 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02fd2a3965624d14bea709d6f858af24 : SUCCESS in 17m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b78e020389b14c1e937793aa87f6df07 : SUCCESS in 6m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/19da90fb271b466f858ea8e1a89ede9d : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cb67a91b5804274a74e785da70a9924 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34fc233b9ffa4e9ab2d065d6679892ce : SUCCESS in 39m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b3a54e9913e4f8b88f50f66aa4b0923 : SUCCESS in 45m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2d3b36f65894cdc8f49427ac0386b2e : SUCCESS in 47m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f72885412754485da813272ff1e2fcd8 : SUCCESS in 51m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88760384ed684c2dbb2f413fda6945d2 : SUCCESS in 1h 03m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d1f0417356b445848a2b522e5b88dd3f : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/14c77746cb9f46aa816b80c08fb8b2e7 : SUCCESS in 1h 06m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72935b9b449442138dc336068fd9f3db : SUCCESS in 1h 27m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/360c3ce15cfc4bf893767f2a1140dd41 : SUCCESS in 59m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b323aa9dc732db075a0e5d95287f29267eaf47ad","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-22 07:52:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a3679f7d7b1e161756a0c9c94cdeae131fa032f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:53:13.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/821917a2b1124a47858a104bdef81740 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef938f41e2ac42deb1f2f751dc8f14ec : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5669dd46c9f45ca9b1cf202cf5172f3 : FAILURE in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9522adf43dce4cebbe65d3bc8e6fa6e0 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5978c837a8014271827513bcf57fcdc5 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3758ba891e14667a7d3a0be49fe0915 : SUCCESS in 14m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/81598fd0ce564c39995d29bcbfad11f8 : SUCCESS in 7m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/42ec704d177f4d608ecbde7b85b28e98 : SUCCESS in 8m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f9e4b90d6c548ce85bf85c2756d442e : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/880cc10569d94f21a4ae75ca8478883d : SUCCESS in 41m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4194e8f6c8d4620b628ee54dfe7da01 : SUCCESS in 1h 41m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76864a5ab1334d879ffe4fd540d1a535 : SUCCESS in 1h 08m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fdf6dc9c22a4456aa4d840ff7ff90b30 : SUCCESS in 45m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbd4653cb06c4f798e64c5a8582ce4ef : SUCCESS in 1h 04m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ddc7b85477c94a2b8333501692f228c5 : SUCCESS in 1h 54m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/624aac19a8694657b5b28d74d47df27c : SUCCESS in 1h 48m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7b315068f724ac589b5fb931b579646 : POST_FAILURE in 1h 19m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a9332329ca7491683441f63e696e3f8 : SUCCESS in 1h 39m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"345a905edd0f03695f2b6b8fdfd61e1a9d86d284","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-22 10:42:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2c4a1165d1549e3be533139a22c981362733c3d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 13:10:33.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b072645dc26b49ceb41a8e7f41bdc156 : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3da9bf6cdf3a41df991e6e25eac81e76 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/190940bedd3148fe88aaa460c63230e5 : FAILURE in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9ac0394124643269656fa7e9cae8463 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f4ab63250146158e8d87e08eb86728 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4fb0ef8293e4c5abee9ba9265f694cf : SUCCESS in 13m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d6415fb7c8340ef866d9f5489501160 : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/119e59cbb54940839cc63f8f155bc825 : SUCCESS in 8m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b118e347f6fc415a9a726c1cfdebf2f6 : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8142943d1a94c14a62d57cf3787ba29 : SUCCESS in 39m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fd33d4debf647bcac4789550dcc953c : SUCCESS in 1h 39m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/782b823dd48545c1a009ad5b532effbd : SUCCESS in 52m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/89d27f90e8574fdc827d266a63d912d2 : SUCCESS in 50m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5efe6b4352542d490a83474a17c5912 : SUCCESS in 52m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65b41b9a24ec45d68dfec28630fb3e9c : SUCCESS in 1h 39m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/503861bdda154508ad5152fd17d99714 : SUCCESS in 1h 43m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3bb5e180892448b99bef7801c0ead35e : SUCCESS in 2h 17m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1f2e010ac22d4084950065addf1fae16 : FAILURE in 2h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7b234e79c18c9e1ed52c58de6665673dc8a1d94b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-22 13:33:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"eb99a2d22580aa234bb572ca0a4e6c3f14a8c3e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 15:54:14.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dd9966063ea43b2976659ccafac1923 : SUCCESS in 11m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d43badc205ff484d86c164ca018019b7 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab560fb1999244b5a923fec8e36c3e29 : FAILURE in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/967c79f064714656a07d34ac1d166889 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9e88930145c4bc695e70b953605e973 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8193ba09a4684b0899219ddc984ccb57 : SUCCESS in 14m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/31e1ed53f4b440e2b23bbe0d590d7ad6 : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3423d02a3f13461bbac6181cd53239e0 : SUCCESS in 9m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd8e608c4a044129b6c96752b41a70f9 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02a31bb69bb04acb99ef33e1c8888a41 : SUCCESS in 45m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbcc0eeeb3994bc69e8b6f1e38ff8a20 : SUCCESS in 2h 00m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b98a188b1e474d009296c0250a274f8b : SUCCESS in 1h 17m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/537a81dbd5a24711a2f7231a8a38b091 : SUCCESS in 56m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf2f33db4a724d559db7c768034d3955 : SUCCESS in 40m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14fe02c8f16c479a832d0819255b3008 : SUCCESS in 1h 18m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6451147e7504b2ba8043c807857a62f : SUCCESS in 1h 06m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1cacca0758004210a5987b757b0ae8dc : SUCCESS in 2h 08m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72e5b2d0ecf345f098fc89f62596e213 : SUCCESS in 1h 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f48187808fd89def84b11d910672e1b3b6a6300","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-24 06:40:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d7259878862e682a12af63d05b08cc51a65e7afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 08:45:04.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d73eb3f899dc43e7a128c19691038ed3 : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d051c5aa82ee46f68166b963cfd96782 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc40198dc8143d49572c0ee386ef489 : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf1916b312e846b7856a7654f835a432 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cccbbb14a477453cabab1f5e6ed62396 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34708dd223549c1add54e071d163766 : SUCCESS in 11m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7edee821ab714dfdb367d32cd55fbe0d : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/558464230a3f4e2fb2e68dbdc4e3c504 : SUCCESS in 16m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f40e44ae76b5407aa19681ba537c8c1a : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e5ee203083247eaba008ca1b34a2fc8 : SUCCESS in 1h 00m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2b941f714e14fd383e4ad58132470e2 : SUCCESS in 1h 41m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e6b3a3096e9543a5afcbce59ad94df9b : SUCCESS in 52m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7f5e0ebab4ff4a329f3d8590ab37f24a : SUCCESS in 52m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d15e0f0691b54bb3be635593cae21b79 : SUCCESS in 58m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6ed2a0ec7c445b9bfe67b261ff570e6 : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1911287116a742c1a0f862c1425bb7cf : SUCCESS in 1h 57m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b02d72e009e4496595f8728bc1f8d21d : SUCCESS in 1h 20m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1f7d1740edb04ac3820842c453ccbc56 : SUCCESS in 1h 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5b8bd559c073bae6e95128b51aa1b4470104e937","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-25 12:21:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5e0d13ec1eb84916553521a48ca4060ca177515d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:35:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ab56b0bb3a54ab58445a78f24b1fea4 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e3bef9194a5423ca6d9fa24818485ea : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a6d1ee18874a8f8a4c47dd69a282ac : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64c690819e3c4b7bb35a07be57dbd15c : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3396480b60914055b11fb009b07a9fbe : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de7a1d6c61c48d5a97a6887f967104d : SUCCESS in 13m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bd340deb35cc48b4bd70cd6b77c8629c : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3e498a8f1b3d4f6985c01ce72d27be80 : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1691d6d71ab4aaf8dbbdbf3523b97ab : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/879be36472d5455b9daa589a214eb9d7 : SUCCESS in 48m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5d5331c78e64cbbb3d2a4561dc334b0 : SUCCESS in 1h 00m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ab1e84dfb0642bd9c000e670f24d0a4 : SUCCESS in 57m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/82370cd1a8404267a3996bad42c8246a : SUCCESS in 47m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf3ef1a23ee947928608873c4140a394 : SUCCESS in 1h 02m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dce2c20d44424da681b0d51a2da34bf1 : FAILURE in 2h 05m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4824cc94f77d4497919bf73506e23d2e : SUCCESS in 1h 45m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/51f86c2bb49042ee8fcc4acdcde3bb50 : SUCCESS in 1h 53m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2052694c82b542dd8fafb5e0a5a643d2 : SUCCESS in 1h 57m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"473ba4cf93f0f0f3013c44c14317268fdc1fe46d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-29 08:35:19.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"875d81d210e49ab224758d706f98d3ee9662eaa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 10:51:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/215b85acfb2b4617809daa3fb661ac8a : SUCCESS in 13m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0fd326b5fb9430983c3257ff981a821 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4a8eafb19949489ba3784faab2c1cc : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55fc251d02554ce582dcd0ea8a2522d4 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dab2d50d81e4c9692cfc7219fa5658c : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebd793420ac4d4fbfbf039394fc5122 : SUCCESS in 14m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/772cab09d7124dc9a9a29349c5b87a1d : SUCCESS in 10m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6feddaf6738f4ad5b44f2273b9a85ba4 : SUCCESS in 12m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/278fdbede5ba4a3484acda30ce08d622 : SUCCESS in 11m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fef0a988349b4f13b8b9267786d13fe6 : SUCCESS in 36m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d97fe869e2634b02a258d6347383bce4 : SUCCESS in 59m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/28b435bb67fb472bb5ea828d73dc7d2a : SUCCESS in 1h 14m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/92a93829e9414de09dc7f3190ff06115 : SUCCESS in 54m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11c565c2d75a44e1bdb8269920b3a614 : SUCCESS in 1h 05m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67def13743a74b54977cf541f8d7eda9 : SUCCESS in 1h 34m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba8fa91c4c2b4f688e6c50d3832b6a4f : SUCCESS in 2h 07m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a7d1b6469f194f768feba6f6d7917b78 : POST_FAILURE in 1h 30m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1736c476bcfb435c84544ad85b217646 : SUCCESS in 2h 00m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"105bcbdbddffa703e625583caa9a4cd1bdbca2ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 18:52:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nMarking -1 because at a minimum we need to document this new state.\nThe other comments are discussion items I would like opinion from other community members.","accounts_in_message":[],"_revision_number":5},{"id":"016a134de37e52ef3cdbe1e0efeae976085e3560","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-20 12:22:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ca57b5b6faa9b3e1a00902e3393e6cc41ca20ad7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-23 16:49:25.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis should have a \"fix\" release note.","accounts_in_message":[],"_revision_number":5},{"id":"6daea96448928727bfee26fa73595edffd80e7f7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-24 14:44:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"02f653cac1130e520edfc61a414996ed0af1ac6e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-24 14:44:43.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n \u003e \n \u003e This should have a \"fix\" release note.\n\ndone, thanks","accounts_in_message":[],"_revision_number":6},{"id":"a548842aa22da54c6d4387682b345dcafe67baae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 16:12:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/105d5278b05540838528b918c62a3505 : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca887a31727944838a074348462a3748 : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6911b93a9e234e58b1e21d74f941997a : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e743c8e1a4c427d9c141f4a1ba99f3a : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7c098f6e81141f09084cad07481c580 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790f299a21e142d6bff3108079ddd149 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/957cab73879a48f8bcf50aa693696273 : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/50e42c8135154c1d9f2f08e5a732390c : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/795fcd528c424d37af6f3f2d92737fc6 : SUCCESS in 12m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4f1da7e0bb78465ebe9613f32c9dfb8a : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72b619b573874335bc070ab14ce2d706 : SUCCESS in 48m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6216505323cf4bb9b33d038df99ab8ca : SUCCESS in 1h 21m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d9de96f95aa94f28bc8c09b6428b1c76 : SUCCESS in 43m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/92fc8063744849a3856214155b88e892 : SUCCESS in 32m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f37aab620c04c96ab002267de2c106c : SUCCESS in 51m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e55874b33139427e80fddc81acbcdae5 : SUCCESS in 1h 22m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2f68927e56064e6393e4b87078a1b795 : SUCCESS in 1h 12m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dbf96ce4c19741c383900390457738ac : SUCCESS in 1h 20m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7e6443f8e9fe4f4a817ad41ab588ee39 : SUCCESS in 1h 10m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1924e313a94b48f4b49dc12ab97c1f6a : SUCCESS in 44m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fbcd3a5acfb941fda299aad7adecf0e7 : SUCCESS in 1h 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fa47abb119387422de2b92cc7b522a636649fe37","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-13 12:31:45.000000000","message":"Patch Set 6: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bf050cff52895b2fb3029c87f1bf6e046610bfae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-13 13:41:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bff14ccd626b8aeb8c2f943446c22b010d494982","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 20:58:52.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nI don\u0027t think this patch needs to import oslo.config into the keepalivedlvs_query.py file.","accounts_in_message":[],"_revision_number":6},{"id":"005df820f5d4db26b6aa16ca737e0f9264211b8a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-26 08:02:10.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"538d1e8d0feaa169a5a67c0b3f2418dbc5c0ee7c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-26 08:02:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1c0118d65220861c5723643c8087bf47d0a15f2a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-26 08:06:26.000000000","message":"Patch Set 7: Backport-Candidate+2","accounts_in_message":[],"_revision_number":7},{"id":"b51a76152b5d3a4c6e9c87b650987599cb65afc3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 09:51:34.000000000","message":"Patch Set 7: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"17ea0da07c8140e30a2c4d5a6607aca8c3444f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:02:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ac451451e184d41ba6ec2866b597c04 : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dde50b9cf21044d489ef41f5422542df : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d2befb12bac49c9b4d38db29991f0cb : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616394e56a46462ca8911799ad751a83 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d7465a85c034bea8e783112c3a19fe6 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b3429965574325aff82cbfc1d81a0e : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f41f95a6507749bba22409ad27830fc3 : SUCCESS in 8m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/637bf45600e44f078544a5fe0d8d746a : SUCCESS in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e68b42d9cdb847c893fc7e8d5cb7760f : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fcd906266eb409892869d6c163ad63f : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1e3a0027a1640f4bcd8ef5468c43def : SUCCESS in 42m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0492b7366f9849efac194f0c141529ef : SUCCESS in 1h 00m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d55bb6c62394bcb985ed8b383d6aaac : SUCCESS in 43m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2c2ad4111b8541b09f8303fc1d32155b : SUCCESS in 54m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f569755098794118acf3b93475e7ff10 : SUCCESS in 33m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e79c7105db141789494706f88c6eb5e : SUCCESS in 1h 53m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73e7839198024d84a82ed3ee1ae8a134 : SUCCESS in 1h 05m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/70244dddb15d45f4af11887a27433126 : SUCCESS in 1h 13m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c0940b5424fd44a2a7660bc5d0263f2f : SUCCESS in 57m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3b7a659e54024d2fb6526902db2ecc01 : SUCCESS in 42m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e26f0a933f9a49bbaa6e35bb456729a1 : SUCCESS in 45m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d8cedbcbe6a34fc79ca1903e4fcb7bf1 : FAILURE in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f4f7db2ffa4598ff9f31841c7028f80b6cc83529","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-26 14:00:01.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"09426a98a7a106940c4c31d76611a0dfa2088761","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-26 19:04:02.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"5feba0ea41c30ffcf53650c1fffee87bf205d223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:04:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"84cad99deab970ed3c388e86c434fc71e93b64a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:14:10.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7af0519327b24a73a74e3c65fbc67dcc : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944a2b5772b24072920c7106e5b6301a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0755058294745d88873afe66bdac5de : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7134ffd9c04e40c49c9fbb1ac069997f : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a666d7f3f2f045a7a37680c5628dc1a3 : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df1fbfa5ca364055ae6577e70c3cf28a : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c126019a18ab47e88f63ab0fa100e589 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ea68e2254734f28a03e92c20e02fefd : SUCCESS in 45m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73d69c407bb84d1baf081dba2a3eaa52 : SUCCESS in 57m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd5067c8559245b58259681d0e2573d7 : SUCCESS in 42m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5994844560794adeaa484e204ecc25e5 : SUCCESS in 38m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a42188e479434e3785a9bae332b6d46d : SUCCESS in 40m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/26a0d99bce8d4ee7b88cb64bf48744ba : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":7},{"id":"d73fd3842112ae5e6f94e299fa746c154211a701","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:14:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d55e5336748ed6877ce7086fa3901cadeb56d59e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:16:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef447bc25fd64d0b800e717e7181abed : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad2b3d9e354b46c5a857ddeacfb393a6 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9fb58eb9f4d2574106a48e3cd3c439d5345517f9","revisions":{"b1be37e58210da5d71384304e224c20af0ecac3d":{"kind":"REWORK","_number":1,"created":"2020-06-18 19:45:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/736818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/736818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/736818/1"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 19:44:31.000000000","tz":120},"subject":"WIP Fix invalid DOWN status when updating a UDP pool","message":"WIP Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can  be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1be37e58210da5d71384304e224c20af0ecac3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1be37e58210da5d71384304e224c20af0ecac3d"}]},"branch":"refs/heads/master"},"149ab69359ed6d9c4310a329aa8f99e1c280b121":{"kind":"REWORK","_number":2,"created":"2020-06-19 07:33:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/736818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/736818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/736818/2"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-19 07:30:04.000000000","tz":120},"subject":"WIP Fix invalid DOWN status when updating a UDP pool","message":"WIP Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can  be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/149ab69359ed6d9c4310a329aa8f99e1c280b121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/149ab69359ed6d9c4310a329aa8f99e1c280b121"}]},"branch":"refs/heads/master"},"7c07e8e771ec0d3ed1fa8f9b303772c6e92c7832":{"kind":"REWORK","_number":3,"created":"2020-06-19 11:26:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/736818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/736818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/736818/3"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-19 11:25:54.000000000","tz":120},"subject":"WIP Fix invalid DOWN status when updating a UDP pool","message":"WIP Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can  be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c07e8e771ec0d3ed1fa8f9b303772c6e92c7832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c07e8e771ec0d3ed1fa8f9b303772c6e92c7832"}]},"branch":"refs/heads/master"},"da9eeabdad7b8e22dd3cea3d3398eaa3f8278930":{"kind":"REWORK","_number":4,"created":"2020-06-24 06:40:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/736818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/736818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/736818/4"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-24 06:40:18.000000000","tz":120},"subject":"WIP Fix invalid DOWN status when updating a UDP pool","message":"WIP Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can  be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da9eeabdad7b8e22dd3cea3d3398eaa3f8278930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da9eeabdad7b8e22dd3cea3d3398eaa3f8278930"}]},"branch":"refs/heads/master"},"d1298db411a0da4beca3b39082c78680fa53e2ae":{"kind":"REWORK","_number":5,"created":"2020-06-25 12:21:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/736818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/736818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/736818/5"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-25 12:01:41.000000000","tz":120},"subject":"Fix invalid DOWN status when updating a UDP pool","message":"Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1298db411a0da4beca3b39082c78680fa53e2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1298db411a0da4beca3b39082c78680fa53e2ae"}]},"branch":"refs/heads/master"},"22f71613805b580ad07eef0653de0e419fc93a5a":{"kind":"REWORK","_number":6,"created":"2020-07-24 14:44:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/736818/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/736818/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/736818/6"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-24 14:30:55.000000000","tz":120},"subject":"Fix invalid DOWN status when updating a UDP pool","message":"Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22f71613805b580ad07eef0653de0e419fc93a5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22f71613805b580ad07eef0653de0e419fc93a5a"}]},"branch":"refs/heads/master"},"9fb58eb9f4d2574106a48e3cd3c439d5345517f9":{"kind":"REWORK","_number":7,"created":"2020-08-26 08:02:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/736818/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/736818/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/736818/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/736818/7"}}},"commit":{"parents":[{"commit":"2d97ebcd24a786a58e67d1f7af344d29263adcd9","subject":"Merge \"Allow amphorav2 to run without jobboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d97ebcd24a786a58e67d1f7af344d29263adcd9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-26 07:56:57.000000000","tz":120},"subject":"Fix invalid DOWN status when updating a UDP pool","message":"Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb58eb9f4d2574106a48e3cd3c439d5345517f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb58eb9f4d2574106a48e3cd3c439d5345517f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
