)]}'
{"id":"openstack%2Foctavia~736825","triplet_id":"openstack%2Foctavia~stable%2Frocky~I0213424eaafe079e7f9a9aaee8e29514c7e840f4","project":"openstack/octavia","branch":"stable/rocky","topic":"octavia-grenade","hashtags":[],"change_id":"I0213424eaafe079e7f9a9aaee8e29514c7e840f4","subject":"Remove install DIB from Git option and n-v grenade","status":"ABANDONED","created":"2020-06-18 20:06:33.000000000","updated":"2020-07-24 11:42:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8180918b5b4578a02bbbe87339963eded565e8cd","_number":736825,"virtual_id_number":736825,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-19 10:54:51.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-19 10:59:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 10:59:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7accbe21fc33babb2a476c6578cc32d5d7a7dd06","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-18 20:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66124a8df0be2067934f94717bc0f241a979c930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:36:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2f3aed9edf34d2ab42d8c00ee963e4a : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6984034348d44187a339f27c863b5c76 : SUCCESS in 10m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8c0392028d048d180ba4b7cff93133d : SUCCESS in 6m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/62989b49a0e0471e9137cb1036994623 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e202b4570e34d65a0f592138bedb1b9 : SUCCESS in 8m 07s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/661a74bc8a3c415f865ecf16f3907487 : FAILURE in 21m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/38debefd0f5f438dac733992951d6542 : FAILURE in 18m 42s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4478cb8746734746abb4a8d8d03e09db : FAILURE in 19m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/71ce22eeb4ad4374a700b71202776c01 : FAILURE in 17m 19s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16a18b1a6fd34d27ac3a5a2e3ce44b5a : SUCCESS in 9m 22s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f1daf662f8f64711898ea9f2c1f24a9e : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55c32604805b49c9b21ca7650ff94084 : SUCCESS in 7m 14s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/374d5cff70c2436fad11d03c6365698d : FAILURE in 20m 15s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/30eedfbd6155434fa62b351e951124c6 : FAILURE in 13m 58s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b055cc2e5b3a483a8968cf54fe625371 : FAILURE in 17m 06s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/8ddda90aca4044498a08647693d1a95e : FAILURE in 19m 52s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c0662686e3fd49a2b8d828f2027211e5 : FAILURE in 15m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3db226646d9f403883746bfdb687a6ae : FAILURE in 16m 36s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/12d70cb42b2c45f5bda60682e3a238fd : FAILURE in 18m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/002cef4cdd2d42eba631eb8eca3a3bbb : FAILURE in 15m 52s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/07d1e75d3ab146d0a4cfbdb6eb292a8d : FAILURE in 19m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9e5e8b4e82d0418492365b894ab3e3d5 : FAILURE in 22m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e8e04a5013e449ca988e0b6b4c251f6 : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":1},{"id":"322f591c8b6c2d5869fdba77b03a8d54e0b7be4a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 10:35:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2bc75215ca8a379f0dffee75442f97d45e17fae","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 10:54:51.000000000","message":"Patch Set 2:\n\nuWSGI issues in devstack. https://review.opendev.org/#/c/735615/ may fix it.\n\n2020-06-19 10:49:49.257 | Successfully downloaded uwsgi\n2020-06-19 10:49:49.460 | You are using pip version 9.0.3, however version 20.1.1 is available.\n2020-06-19 10:49:49.460 | You should consider upgrading via the \u0027pip install --upgrade pip\u0027 command.\n2020-06-19 10:49:49.546 | + lib/apache:install_apache_uwsgi:97       :   local uwsgi\n2020-06-19 10:49:49.553 | ++ lib/apache:install_apache_uwsgi:98       :   ls \u0027uwsgi*\u0027\n2020-06-19 10:49:49.556 | ls: cannot access \u0027uwsgi*\u0027: No such file or directory\n2020-06-19 10:49:49.562 | + lib/apache:install_apache_uwsgi:98       :   uwsgi\u003d\n2020-06-19 10:49:49.566 | + lib/apache:install_apache_uwsgi:1        :   exit_trap","accounts_in_message":[],"_revision_number":2},{"id":"9f0decbe55ca2842494220c7d39c816c89a7dc26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:59:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a520d1c266844a05a6084b152b6b14e0 : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838e00e59e264278aa1df388523486d5 : SUCCESS in 7m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2483ad6122c4b44b86a6ab267121943 : SUCCESS in 6m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ba2f90283a26462083c25cb64a920107 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6013c31c2984273bb19f5c16aca4a60 : SUCCESS in 6m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/56ee5498af724afa8d2ce983ae767c68 : FAILURE in 16m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3bb4963b51e947d19fe178d099bf8e0a : FAILURE in 18m 26s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ba5736f60f14e70a652811e9315c1fe : FAILURE in 10m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f5eda369871640a4a1f3ffe7e72a2bff : FAILURE in 17m 41s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d44a4d0e2f0349d8a784b6ce187cf1ef : SUCCESS in 8m 59s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3fb4da2aaf274e59ad88f252fe95f4d3 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/67e191f1630249b494212aff61cf993b : SUCCESS in 9m 21s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3391434e8fc54437820331046b7cabe2 : FAILURE in 14m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/66ed18b141e940bbb1f779486b236517 : FAILURE in 14m 21s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e01345e959cd454482d7955bf0092822 : FAILURE in 18m 12s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/ffc451ce590d40a3a04926dacc747edc : FAILURE in 11m 51s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8981bfaf0555455d84652154fba4b11c : FAILURE in 13m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a06629f6d0e8416f8c08e30a0f641a06 : FAILURE in 15m 53s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/8312c62a3e404fbeb6fadb371b2d3c5e : FAILURE in 15m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2f491e048444e03abc321661a04bd24 : FAILURE in 14m 23s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/0755df29602a46c6a77dbc271537eef3 : FAILURE in 15m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7bebf45d98ed4c8486a50f620760b5d1 : FAILURE in 13m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f2a2ae540f34b3488a935a80827bf2e : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c0eb8f0f82d159ada1a56c040d2be6ac2d59acd9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:42:08.000000000","message":"Abandoned\n\nRocky Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"03f80f6f8d266ce9ccdbc46bf9af11cbdac7162f","revisions":{"fd968ab75518fff31540e2690c9b4601f8f540a5":{"kind":"REWORK","_number":1,"created":"2020-06-18 20:06:33.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/25/736825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/736825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/736825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/736825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/736825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/736825/1"}}},"commit":{"parents":[{"commit":"126ad303ac998f2fa36de572bb37cb36f366913a","subject":"Merge \"Fix the amphora noop driver\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/126ad303ac998f2fa36de572bb37cb36f366913a"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-18 20:01:27.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-18 20:06:10.000000000","tz":120},"subject":"Remove install DIB from Git option","message":"Remove install DIB from Git option\n\nDIB no longer supports Python 2 and is upper constrainted to 2.30.0.\nAllowing to install DIB from Git would cause nothing but trouble. Just\nrely on pip to a functional DIB version.\n\nChange-Id: I0213424eaafe079e7f9a9aaee8e29514c7e840f4\n(cherry picked from commit 0d2c567e3e52cea38f5e56ac10c07e4f13900735)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd968ab75518fff31540e2690c9b4601f8f540a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd968ab75518fff31540e2690c9b4601f8f540a5"}]},"branch":"refs/heads/stable/rocky"},"03f80f6f8d266ce9ccdbc46bf9af11cbdac7162f":{"kind":"REWORK","_number":2,"created":"2020-06-19 10:35:02.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/25/736825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/736825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/736825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/736825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/736825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/736825/2"}}},"commit":{"parents":[{"commit":"126ad303ac998f2fa36de572bb37cb36f366913a","subject":"Merge \"Fix the amphora noop driver\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/126ad303ac998f2fa36de572bb37cb36f366913a"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-18 20:01:27.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-19 10:33:01.000000000","tz":120},"subject":"Remove install DIB from Git option and n-v grenade","message":"Remove install DIB from Git option and n-v grenade\n\nDIB no longer supports Python 2 and is upper constrainted to 2.30.0.\nAllowing to install DIB from Git would cause nothing but trouble. Just\nrely on pip to a functional DIB version.\n\nThis patch also sets octavia-grenade job non-voting.\n\nA recent uWSGI release broke devstack master and stable branches. It\u0027s\nbecoming increasingly difficult to maintain branches in Extended\nMaintenance, especially the grenade job which depends on an older\nrelease. The OpenStack QA team has decided [1] that the best thing to do\nis to only perform upgrade testing between the current source contents\nof adjacent Maintained or Development status branches.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015499.html\n\nChange-Id: I0213424eaafe079e7f9a9aaee8e29514c7e840f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03f80f6f8d266ce9ccdbc46bf9af11cbdac7162f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03f80f6f8d266ce9ccdbc46bf9af11cbdac7162f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
