)]}'
{"id":"openstack%2Foctavia~736826","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I0213424eaafe079e7f9a9aaee8e29514c7e840f4","project":"openstack/octavia","branch":"stable/queens","topic":"octavia-grenade","hashtags":[],"change_id":"I0213424eaafe079e7f9a9aaee8e29514c7e840f4","subject":"Remove install DIB from Git option","status":"ABANDONED","created":"2020-06-18 20:07:15.000000000","updated":"2020-07-24 11:41:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d3000795dd7800a8f79f18bf3299a983b3ac1b1","_number":736826,"virtual_id_number":736826,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-18 20:31:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 20:31:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06ac33c4bda61b7563fb7fb964ea05f9ea9053f8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-18 20:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5bb139ee046bd8dc41cbdaef66218568d81bac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:31:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9854634f4a454f5d86a79a813459ec68 : SUCCESS in 8m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66ef7e51c49747589ad6f13f4cc7201b : SUCCESS in 7m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bbbdd148ac614293867eb260a43f0de0 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd5edb62cd954f96840a7db4a8dfacb7 : SUCCESS in 6m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f847c0f605ae4c9d83f21369797c3c67 : FAILURE in 17m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a27f69f53c6a48178e4f045019de0c6a : FAILURE in 18m 41s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0c263d87c9344d28446e82b09b4080c : FAILURE in 15m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/24fcfed43e0348c5a9b8ec5819a60636 : FAILURE in 15m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43b316bdf3bf44e3af0a5cfcc5664a89 : SUCCESS in 11m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7b3dba109a044e2b8051ebbcc5b630bd : SUCCESS in 7m 55s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0ba0b7c681d4488bb26e8cc688c63e77 : SUCCESS in 8m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34beac42a96d4411982e4352c75ef7ee : FAILURE in 8m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4a30dec4b77545fc920e0d6d3fbaebe1 : FAILURE in 9m 35s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e9241d8125764d299ded8542d0f280db : FAILURE in 9m 41s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5ead8a2ce0d44107874288ad9f1bb0ec : FAILURE in 8m 12s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ebf29301238d4b349e46c735e2e56c27 : FAILURE in 8m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96de41eb5a14401291a040d770725eaa : FAILURE in 12m 11s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/34cee3eda6674591a5051de4f64ae520 : FAILURE in 12m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ad3aaae04de43d7961aca063b0b15a0 : FAILURE in 14m 16s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b5f5c377a93b4de48bda9f01ec9081c8 : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e19cd58d6f051c8e5040933cbee815a526d4bbeb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:41:17.000000000","message":"Abandoned\n\nQueens Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"377861ed6023295ee9c6c17d37a4a0683bff3345","revisions":{"377861ed6023295ee9c6c17d37a4a0683bff3345":{"kind":"REWORK","_number":1,"created":"2020-06-18 20:07:15.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/736826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/736826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/736826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/736826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/736826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/736826/1"}}},"commit":{"parents":[{"commit":"299c54a5dee5898eb0712123b98670f4043bdec6","subject":"Merge \"Fix the amphora noop driver\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/299c54a5dee5898eb0712123b98670f4043bdec6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-18 20:01:27.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-18 20:07:02.000000000","tz":120},"subject":"Remove install DIB from Git option","message":"Remove install DIB from Git option\n\nDIB no longer supports Python 2 and is upper constrainted to 2.30.0.\nAllowing to install DIB from Git would cause nothing but trouble. Just\nrely on pip to a functional DIB version.\n\nChange-Id: I0213424eaafe079e7f9a9aaee8e29514c7e840f4\n(cherry picked from commit 0d2c567e3e52cea38f5e56ac10c07e4f13900735)\n(cherry picked from commit fd968ab75518fff31540e2690c9b4601f8f540a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/377861ed6023295ee9c6c17d37a4a0683bff3345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/377861ed6023295ee9c6c17d37a4a0683bff3345"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
