)]}'
{"id":"openstack%2Foctavia~736832","triplet_id":"openstack%2Foctavia~master~Ia0dddac00155e11098fcbd3e95bcae12f2fc63d4","project":"openstack/octavia","branch":"master","topic":"diskimage_shellcheck","hashtags":[],"change_id":"Ia0dddac00155e11098fcbd3e95bcae12f2fc63d4","subject":"diskimage-create compatible with ShellCheck.","status":"MERGED","created":"2020-06-18 20:33:56.000000000","updated":"2020-06-19 21:14:45.000000000","submitted":"2020-06-19 21:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736832-1592601166548-4b88bc5a","meta_rev_id":"442bcbf7ef45a0228c7c437d4941989b426485c8","_number":736832,"virtual_id_number":736832,"owner":{"_account_id":31335,"name":"Thobias Trevisan","email":"thobias_trevisan@sicredi.com.br","username":"thobiast"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 21:12:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-06-19 19:46:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-06-19 17:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-06-19 19:46:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-06-19 19:46:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 17:12:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-19 19:46:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-19 21:12:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9680688b4e2ca1e21245671d7bf38512834f52cd","author":{"_account_id":31335,"name":"Thobias Trevisan","email":"thobias_trevisan@sicredi.com.br","username":"thobiast"},"date":"2020-06-18 20:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adc265698b2348dd59db237998aef20685d6dc59","author":{"_account_id":31335,"name":"Thobias Trevisan","email":"thobias_trevisan@sicredi.com.br","username":"thobiast"},"date":"2020-06-18 20:42:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fa9f699f51217c18b149eff8d4fa59b41fd4a63f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-18 21:22:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"893453eae4670048fe6ff0f645ed8ccad348ac2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:58:56.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f94f76da42145a7a5f46f610e4bef67 : SUCCESS in 11m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5ff14ab092e48dda55510a92a64fdd6 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a33b8d039b244c98b422c641b3df2c96 : FAILURE in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f928a01bf5d840a581603e0a45ad6dac : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03aa536a26454785aadfdeaf82829003 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797ac155ecdf47c193282286e77a2e87 : SUCCESS in 14m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/71f94e36278c4059ae88084ac5e41ed5 : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c50ff928f835408f9365656934ed89cc : SUCCESS in 12m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e53b800099c4e9c8cddffff0d161915 : SUCCESS in 9m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6ed0ea3dc0440b8b614010a327300c1 : SUCCESS in 41m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76f95d7c0c2748669d676e3a44e5f7a3 : SUCCESS in 1h 42m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce69a8449fb247cabe9d381e70a53eb8 : SUCCESS in 1h 04m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f8a4b71893794aa880ccad7632fb3d13 : SUCCESS in 1h 02m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc5a05a2ffea42c5a65692ce9023b2e8 : SUCCESS in 1h 05m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5542539defd48ed8cba12e625da5c6a : SUCCESS in 2h 10m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/52b2dc755f614bfdad8718f60aefbbb6 : SUCCESS in 1h 43m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a86e3c6d534d4ad5a13fb1d86cb72e80 : SUCCESS in 2h 06m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b843667745b34cb2ba32744947c80535 : SUCCESS in 55m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c6372f88cdebf431ce596fb9e76f2a4fa40953c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-18 23:14:34.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"5dc1cd1f7e165c98613b92cad8209494fc9b8405","author":{"_account_id":31335,"name":"Thobias Trevisan","email":"thobias_trevisan@sicredi.com.br","username":"thobiast"},"date":"2020-06-19 13:22:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a78d591446abab6eb9e70430e39e603a2b1f7ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 15:25:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0358e27e87084177879915cf1ce19c20 : SUCCESS in 10m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a3482f149e3463f86677ee3102593d1 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b870197246854299958709e8fb384530 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3dc5788f38d40e0b5284593f7cf2429 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53178e19ebbb4e3ea5e3e3d03d24f439 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2be1f239fe47a6a36a3e4023bf7bf4 : SUCCESS in 20m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/308f993da59e4362b13bce6dfbc855cf : SUCCESS in 8m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e82014a0458b495aa7115c3889097c38 : SUCCESS in 11m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dcd1df8302a040718a9a59f76650f426 : SUCCESS in 10m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d04b2cb5c35645a39c52ab3715f5abef : SUCCESS in 47m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcea786823404e20be2733bb6edd7ea8 : SUCCESS in 1h 55m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9802fb44eefe412cb47654250de504a2 : SUCCESS in 1h 11m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ac8984b689a342b7bdcee6554d909768 : SUCCESS in 1h 04m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54aaaacf90b342a09f244f2147c7b50a : SUCCESS in 1h 08m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5c92bf7b4ff4bf0ba1bc8de83823069 : SUCCESS in 1h 27m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/480e8a79399743548130f75ceeb70e27 : SUCCESS in 1h 38m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6659b82ca1544256ab897606bf1a845b : SUCCESS in 1h 56m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/da489451685e44f28338fc8661e10ff3 : SUCCESS in 1h 48m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"24bfb7b4f11dc92f0756f1f3c60ff2b49e0640ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-19 17:12:18.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"7482ce63c6e8af637906571bba136b229f9730f8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 19:46:14.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)\n\nI don\u0027t think this needs to be backported but if someone thinks otherwise, it can be done.","accounts_in_message":[],"_revision_number":3},{"id":"9d1d547615087c1d498425a64e96a55a361cdf9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 19:46:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1cb9e6438a16a35fd58b84a88efc3403bcc58d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:12:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/869a777928fc4a64bc5753bccb238b79 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8450f0b39144081843093ac966aad83 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f404f14561a749dd84e449ede8d99cf3 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1138d80801994799accd75e8bf1898aa : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44adfd0dd29e4c73bbdca2ec8f75c526 : SUCCESS in 18m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e157909e4ecc4f2f89fea36b316b4b9e : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45adcab37384458b89623524d7a6e75b : SUCCESS in 46m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92505754f9244eb2b2ac65d7e89d2548 : SUCCESS in 57m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/145d9a13943f40a9b6dde51df0dd08fa : SUCCESS in 54m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fdc73316dcac41d29389e60b4eda6146 : SUCCESS in 44m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/875788a2bab14ef3b79c97b73cfe6248 : SUCCESS in 58m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/239e54d68f674ed5934da7a23fc005e3 : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":3},{"id":"2bead0971a6ff1d09a9eaab8a974c3dfd872a0f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:12:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6da6e0ba8337993dc5fafe735367a885efb9df48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:14:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338ae56b59f84a308600eabe1346a3cd : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d8d722b54d890caff988a4a97e75fa27b23adbed","revisions":{"86bbf5d51cb08519d66443844636b40fd6cce211":{"kind":"REWORK","_number":1,"created":"2020-06-18 20:33:56.000000000","uploader":{"_account_id":31335,"name":"Thobias Trevisan","email":"thobias_trevisan@sicredi.com.br","username":"thobiast"},"ref":"refs/changes/32/736832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/736832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/736832/1"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Thobias Salazar Trevisan","email":"thobias_trevisan@sicredi.com.br","date":"2020-06-18 20:27:53.000000000","tz":-180},"committer":{"name":"Thobias Salazar Trevisan","email":"thobias_trevisan@sicredi.com.br","date":"2020-06-18 20:27:53.000000000","tz":-180},"subject":"diskimage-create compatible with ShellCheck.","message":"diskimage-create compatible with ShellCheck.\n\nUpdate diskimage-create script to pass in ShellCheck (https://www.shellcheck.net/) tool.\n\nChange-Id: Ia0dddac00155e11098fcbd3e95bcae12f2fc63d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86bbf5d51cb08519d66443844636b40fd6cce211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86bbf5d51cb08519d66443844636b40fd6cce211"}]},"branch":"refs/heads/master"},"28795fbc872174fcf69c1ec0e334531281105f19":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-18 20:42:50.000000000","uploader":{"_account_id":31335,"name":"Thobias Trevisan","email":"thobias_trevisan@sicredi.com.br","username":"thobiast"},"ref":"refs/changes/32/736832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/736832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/736832/2"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Thobias Salazar Trevisan","email":"thobias_trevisan@sicredi.com.br","date":"2020-06-18 20:27:53.000000000","tz":-180},"committer":{"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","date":"2020-06-18 20:35:56.000000000","tz":-180},"subject":"diskimage-create compatible with ShellCheck.","message":"diskimage-create compatible with ShellCheck.\n\nUpdate diskimage-create script to pass in\nShellCheck (https://www.shellcheck.net/) tool.\n\nChange-Id: Ia0dddac00155e11098fcbd3e95bcae12f2fc63d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28795fbc872174fcf69c1ec0e334531281105f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28795fbc872174fcf69c1ec0e334531281105f19"}]},"branch":"refs/heads/master"},"d8d722b54d890caff988a4a97e75fa27b23adbed":{"kind":"REWORK","_number":3,"created":"2020-06-19 13:22:54.000000000","uploader":{"_account_id":31335,"name":"Thobias Trevisan","email":"thobias_trevisan@sicredi.com.br","username":"thobiast"},"ref":"refs/changes/32/736832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/736832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/736832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/736832/3"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","date":"2020-06-18 20:27:53.000000000","tz":-180},"committer":{"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","date":"2020-06-19 13:22:07.000000000","tz":-180},"subject":"diskimage-create compatible with ShellCheck.","message":"diskimage-create compatible with ShellCheck.\n\nUpdate diskimage-create script to pass in\nShellCheck (https://www.shellcheck.net/) tool.\n\nChange-Id: Ia0dddac00155e11098fcbd3e95bcae12f2fc63d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8d722b54d890caff988a4a97e75fa27b23adbed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8d722b54d890caff988a4a97e75fa27b23adbed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
