)]}'
{"id":"openstack%2Foctavia~736965","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ifcb6ad47e8a69db4c059d72b240d1750a0c5ff38","project":"openstack/octavia","branch":"stable/stein","topic":"octavia-grenade","hashtags":[],"change_id":"Ifcb6ad47e8a69db4c059d72b240d1750a0c5ff38","subject":"Set octavia-grenade job non-voting","status":"MERGED","created":"2020-06-19 09:55:16.000000000","updated":"2020-06-25 16:22:24.000000000","submitted":"2020-06-25 16:12:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736965-1593101532820-ae37cacd","meta_rev_id":"7fa9bae5fc073ef42072128280b71807de0452eb","_number":736965,"virtual_id_number":736965,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 16:12:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-06-25 11:54:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-06-25 14:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-19 20:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-21 04:08:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-06-25 14:27:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 20:35:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-21 04:08:37.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-06-25 11:54:48.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-06-25 14:27:24.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-25 16:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"911b752d8be5fe7020c6f0b914a173e233b39b13","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 09:55:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3751a01d4ad58accccff46a3696db960bd9c85a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 11:59:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48063abaded44a52802ca94c9c2d1e67 : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3d3f417f5a748b4b1f358c8ee26bd75 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12598fd3f9c74793a98a1232edeb4911 : SUCCESS in 7m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/197c00b738f940ff92bc94a552a4fc6b : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c79136df70d4bb9a737f355ea22fe17 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754b3b60d72540e0bd6ff57bb1cd01de : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e1998228804266bb3127a69bcfad4b : SUCCESS in 6m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bc1c233680a9469b81b41b97a3499c37 : SUCCESS in 1h 18m 18s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4ecb51337dfd42b6954699a3d27a1453 : SUCCESS in 1h 19m 57s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42d8b1a60d20488da1442817981967f6 : FAILURE in 35m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e246bfb998b443ac93bc9d1ab10eef87 : SUCCESS in 1h 27m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9040457d83e94725bbeaf4e80864ccc7 : SUCCESS in 7m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f4e7a7077be742ccbb8cde6f411c853a : SUCCESS in 11m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d8c62835b4b42798f7c3d2084dd0d86 : SUCCESS in 7m 00s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/781f322834cf4fd481e44934bf123aad : SUCCESS in 1h 05m 32s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8c64f31713294d9d9b2e79a2eb7ca294 : SUCCESS in 1h 40m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43ff96629f0a471d9373d87c1c9585f9 : SUCCESS in 49m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5bf88653d97540319e411b862fa435e5 : SUCCESS in 48m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1157a7867ad44aaaa56a9b2c04f9c5f : SUCCESS in 1h 58m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bacd8061ee7944d892b097e78558cee0 : SUCCESS in 1h 00m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5553af4fb5914fbfb315a3386206070e : FAILURE in 19m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e214264e2ee4e62a7f86ca869c91ba2 : FAILURE in 31m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6e2ab0b7f484eaebbcf06a3ea49dbec : FAILURE in 1h 02m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c3eab9889f8446d86a30c67e6ed888a : FAILURE in 52m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e60f59e26a514c66a6a59312ac0a9af4 : FAILURE in 16m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d40787d8169847728072d31ec042f45f : SUCCESS in 55m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d831c0f636d7e143832ea7497d0d2b6b06662c8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-19 20:35:49.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM\nNot ideal, but necessary at the moment.","accounts_in_message":[],"_revision_number":1},{"id":"3c1737f7058cdfaeaab7d9e0da2d31eeb58f195c","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-21 04:08:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f1f69b8c3ae49821d5f6d9543561ed2471429795","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-25 11:54:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"03c8bb3f7607d9126bb494116dfade8a985780c4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-25 14:27:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f3a1fcdd142bb83f550824e2eac3d118f329849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:27:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74f60715aeef7138e5822f594d2406014dd70d93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 16:12:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b58503f6cc96457ab8c708224b05bd02 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44f732e6d6c840e79368a32c8d31871d : SUCCESS in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8341102b3cbf4c79a3731f70046d1e25 : SUCCESS in 9m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/281010dd10ec49da9112eaf3d460644c : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0fa36b07b843399abe4298221c2268 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11329a10b4349bf809c1dbfc25529cb : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ffe707a8f23643548c9e53768137b09d : SUCCESS in 7m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f2ef9a97b68741efa12900605ffe8593 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43c231226e2b427493689bbda6002bb4 : SUCCESS in 8m 01s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86aff7964a964c7386a1f401d19e902d : SUCCESS in 1h 11m 12s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8706d04cb40648ff89bdc9e6e99ab7fa : SUCCESS in 1h 00m 51s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4d033b38402b46838dd17d892a6acbbe : SUCCESS in 1h 21m 13s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c5f27b94c92e463d825db926811da072 : SUCCESS in 1h 21m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7470e0a7eb7441b979de7274cb0a642 : SUCCESS in 50m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bb7835fbae2242d5918c142dfd8ef5f8 : SUCCESS in 47m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6998dd6081344876b7ae10141aabbff3 : SUCCESS in 1h 36m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c2cd41defe5e473eb0bcd435da159162 : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fa5c954311eb7464ae40b30d045ef4d9e3e2261c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 16:12:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f5304377bf5aa8dcaf535fdbb61170c2f3dfbd51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 16:22:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa73e652ff48465b9ad96793e8d8ee27 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04b3d6a165f77b9e978f98b947044ae6c79a7796","revisions":{"04b3d6a165f77b9e978f98b947044ae6c79a7796":{"kind":"REWORK","_number":1,"created":"2020-06-19 09:55:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/65/736965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/736965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/736965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/736965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/736965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/736965/1"}}},"commit":{"parents":[{"commit":"77b42c6260a8bc1f8a6c38c948bb647256bd915b","subject":"Merge \"Fix multi-listener LB client auth/re-encryption\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77b42c6260a8bc1f8a6c38c948bb647256bd915b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-19 09:46:16.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-19 09:46:16.000000000","tz":120},"subject":"Set octavia-grenade job non-voting","message":"Set octavia-grenade job non-voting\n\nA recent uWSGI release broke devstack master and stable branches. It\u0027s\nbecoming increasingly difficult to maintain branches in Extended\nMaintenance, especially the grenade job which depends on an older\nrelease. The OpenStack QA team has decided [1] that the best thing to do\nis to only perform upgrade testing between the current source contents\nof adjacent Maintained or Development status branches.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015499.html\n\nChange-Id: Ifcb6ad47e8a69db4c059d72b240d1750a0c5ff38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04b3d6a165f77b9e978f98b947044ae6c79a7796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04b3d6a165f77b9e978f98b947044ae6c79a7796"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
