)]}'
{"id":"openstack%2Foctavia~736967","triplet_id":"openstack%2Foctavia~stable%2Frocky~Ifcb6ad47e8a69db4c059d72b240d1750a0c5ff38","project":"openstack/octavia","branch":"stable/rocky","topic":"octavia-grenade","hashtags":[],"change_id":"Ifcb6ad47e8a69db4c059d72b240d1750a0c5ff38","subject":"Set octavia-grenade job non-voting","status":"ABANDONED","created":"2020-06-19 09:59:20.000000000","updated":"2020-06-19 10:35:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab5ae4d72903769fdd50cc7aa788f467dc0aaa09","_number":736967,"virtual_id_number":736967,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-19 10:26:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 10:26:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32d8132c28bab3d9f98664a428af42ccdb19fd4a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 09:59:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df0b9a0e8e22be06b6a5b608b8c7d5ec9cd354f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:26:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65eee54159b14df4a9561e853ef67ca4 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/358b9ce4658e4eeb994fce63888a228c : SUCCESS in 8m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62d52582ca294d5c911a211fffb8faec : SUCCESS in 6m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c37ecbabf7604abd913b28c9a5dba8fa : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e32ea308ad34eb1a8a7167fdd291245 : SUCCESS in 7m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0441b518f24345dab66e5e62df26bdfc : FAILURE in 18m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/648776a4118b48cbbe9021f7c15d6373 : FAILURE in 17m 48s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f3c217ce1fa4405970094f97ea3a036 : FAILURE in 19m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a5995b9f77ab48a2a55978d87297048c : FAILURE in 19m 04s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/79d576613a374abb80ef84df680471d1 : SUCCESS in 9m 16s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2f3e75f857264e7bb4aad63ed0fa4c91 : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d6d75a001224a888ff8e39a09628c4d : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95e59e6f8c87432193b71370ed589d23 : FAILURE in 16m 53s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2882c7120a12444fbf5974e17c469b01 : FAILURE in 12m 28s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/99dc01fa482f453fa0a78fb078787acf : FAILURE in 11m 35s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/ef33c4459f2c495f946cc8278b4fdfce : FAILURE in 12m 16s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b59afe26e1f94e3a9b4eb239aa8afaad : FAILURE in 11m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a0a911dfdfac4ef4bcd1b06b2c60d4ea : FAILURE in 11m 29s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/963fab8b044f4b568ce28711a307812e : FAILURE in 11m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b41b8c32b87847e69e7f39db1be1a14c : FAILURE in 16m 48s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b7482b38635e40d9ac02662b47321ef3 : FAILURE in 15m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0e1fea90433e4de9b620cdd1564070ba : FAILURE in 17m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b0f68073ed049e497edcbabe1f28e67 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"48bb80554830d461060aa00f395b2c52be856aa3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 10:35:37.000000000","message":"Abandoned\n\nSquashed in https://review.opendev.org/#/c/736825/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2e2c0c9a5ceb6eb530b392cebfbe622f3ec7f4a","revisions":{"d2e2c0c9a5ceb6eb530b392cebfbe622f3ec7f4a":{"kind":"REWORK","_number":1,"created":"2020-06-19 09:59:20.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/67/736967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/736967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/736967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/736967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/736967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/736967/1"}}},"commit":{"parents":[{"commit":"126ad303ac998f2fa36de572bb37cb36f366913a","subject":"Merge \"Fix the amphora noop driver\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/126ad303ac998f2fa36de572bb37cb36f366913a"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-19 09:46:16.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-19 09:58:57.000000000","tz":120},"subject":"Set octavia-grenade job non-voting","message":"Set octavia-grenade job non-voting\n\nA recent uWSGI release broke devstack master and stable branches. It\u0027s\nbecoming increasingly difficult to maintain branches in Extended\nMaintenance, especially the grenade job which depends on an older\nrelease. The OpenStack QA team has decided [1] that the best thing to do\nis to only perform upgrade testing between the current source contents\nof adjacent Maintained or Development status branches.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015499.html\n\nChange-Id: Ifcb6ad47e8a69db4c059d72b240d1750a0c5ff38\n(cherry picked from commit 04b3d6a165f77b9e978f98b947044ae6c79a7796)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2e2c0c9a5ceb6eb530b392cebfbe622f3ec7f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2e2c0c9a5ceb6eb530b392cebfbe622f3ec7f4a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
