)]}'
{"id":"openstack%2Foctavia~737111","triplet_id":"openstack%2Foctavia~master~I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd","project":"openstack/octavia","branch":"master","topic":"stats-driver","hashtags":[],"change_id":"I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd","subject":"Refactoring amphora stats driver interface","status":"MERGED","created":"2020-06-20 00:44:12.000000000","updated":"2020-09-09 02:13:10.000000000","submitted":"2020-09-09 02:10:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737111-1599617452973-c09e01d6","meta_rev_id":"ffbca5ce41608fe637c0a3b20389a42888ec660a","_number":737111,"virtual_id_number":737111,"owner":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 02:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-09-09 00:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-09-08 22:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-09-09 00:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-2,"date":"2020-09-08 22:34:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-20 00:44:22.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-06-24 00:00:04.000000000","updated_by":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"reviewer":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"state":"REVIEWER"},{"updated":"2020-06-29 21:53:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-05 16:08:33.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-08-11 16:48:34.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-19 01:26:38.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2020-08-19 01:53:34.000000000","updated_by":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"reviewer":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"state":"REVIEWER"},{"updated":"2020-09-08 22:34:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-09 00:22:05.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-09 02:10:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff9e2e9a53667ccd17faf7c8636192f17a4a4b3f","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-06-20 00:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3af7266b04da154ae4754f8499407ba3fc954c0","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-06-20 00:44:22.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"9ecb618c23c071e44030e657c4931b8f8837684a","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-06-20 00:44:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ded5b888ae47f3b7f236ac1f28fc911d38e779a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 02:25:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95a9e0aa062b43aeb94b37d89f573f54 : FAILURE in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e897719cd164d22bfccaf28ac57ab5a : FAILURE in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706059bf5e874eb6963a204b0475af84 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67a30122436143f1bc164a55c6ec8aa3 : FAILURE in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b30bc8f0bf471e93f6060a5daf3c03 : FAILURE in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7ba51fd73e74734b11fb6f5c45c57e6 : FAILURE in 10m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e0d7079f8b564fd1a4ac5ed55495df97 : FAILURE in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/53b0370aabe54defa9de03ebf8def2bf : SUCCESS in 9m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d47909d4471430c85e15eba6f302f4f : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc3899ad1894400aa055375c57bf859d : SUCCESS in 38m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce2b10e8319a45fca3ab59bbea13a736 : SUCCESS in 58m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64651374c68c477780be45f0a3c45efd : SUCCESS in 49m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/00e5d7d3f49a437ba2d6e4999e0ba62f : SUCCESS in 36m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84dff4a5e51f4b54b351110adcf88f38 : SUCCESS in 46m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/94764883982f4b9a84957a81308d01d7 : SUCCESS in 1h 20m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2397661ce06e4c2db64467deb0d213d0 : SUCCESS in 50m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e20cb0209d5f4fb9af56d4915ea9022b : SUCCESS in 59m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fe37e329822240bdb3e307117d14e02e : SUCCESS in 1h 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e22f37c2ad8b0678aba8581f33055aaff035f9ad","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-06-24 00:00:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2623b60afaf2f7b056b4e83e19a1671cf4b6c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:09:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcd7a9f93dcd4b519e36a3204e406c97 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f21d692e4c6e485abf50efb94db86db4 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0e264d8ffc4b6cb36e1226bb276f07 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/892fd9e2f940497f8227dfdfa98085e4 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a53c6b73da9d4d81877545ab210f02d5 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f7cad39fb34fb6aaee5a2e16833cb9 : FAILURE in 7m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ff3e7addec92473295a4e507040f3489 : SUCCESS in 6m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/69afa1e863aa4c1b8cc20c7383207aa9 : SUCCESS in 8m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f9317ae8cde45d4a19ed29395445ed8 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7136368de7b344a8b87b7dc6c46e2fa5 : SUCCESS in 41m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8809e3a9e764be3bc2750ed80a7cfcb : SUCCESS in 55m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a00a73388f2847dca7b26cf7ca8eb977 : SUCCESS in 42m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cb09ebb648834f0692e1a796938dda50 : SUCCESS in 47m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1add29f5ea1e43e399c6b95eeb551632 : SUCCESS in 31m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6245cdf15674032b11406ad789acd50 : SUCCESS in 1h 39m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/166283ec4d7e45b987c8c4f14192e679 : SUCCESS in 2h 02m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ebf2ce326f049fb9abd887197e04c5e : SUCCESS in 1h 58m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e8f06af1d1f54c0b89d93b45323014c9 : SUCCESS in 1h 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c079cdf0675bc73f78afdec1d3c027acb7762e69","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-06-24 17:38:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"259a6cd39f21a7d6bb2882c4296b92e5b7f5e318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:03:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8ca666b0c194592bfd07b6c9f66493d : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb979e4359ac480fac946cf591a15364 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2050baacf3cd465f853aef81d7396537 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab7af15b643345359669206eb0e74ada : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f217c2ff78ea4dd5ae0443460a98b208 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79841d3c6c75479b8f0fa988762373a9 : SUCCESS in 14m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c6181501f0fd4165ab454caa1b4b578e : SUCCESS in 8m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/15cac200f9ab448b88dde8a5ce10fa12 : SUCCESS in 8m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0703cd8016c34dda8e134f2288e8f86d : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6949499e5d914c97b04be787f9933590 : SUCCESS in 43m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d95413812d5468697098d888d74288c : SUCCESS in 1h 47m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e44f7254158a4823b5bed59af4a728bc : SUCCESS in 58m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e3567aefc2c446d981ae4e7dbfd8cf6d : SUCCESS in 59m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/854a141d3f7c4922afceb6b6c8a13dc0 : SUCCESS in 47m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d698d9f74634f8aa593514208b097ef : SUCCESS in 1h 39m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/454377dde2714fbc9ecfbd19a19f6d92 : SUCCESS in 2h 11m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7449825410d04f4ea2b607257d3eb41e : SUCCESS in 1h 50m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce7eb3f4b57340fdb9c8348cafbd5515 : SUCCESS in 1h 43m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e3f8937900ae44c0bfb868739763adb62f3bdf8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-24 22:06:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cf6b9364b9c27e757696cfd74e097c9293d20759","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-06-25 01:25:52.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"52b9995ed3f3388229064d7e110fd52bd4dc34c6","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-06-25 01:28:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9ce42e4b2ad7dcc170658751147222d7f29b23eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 03:43:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec5f542483634bd895e9177dbfe1128f : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/553c374ac91540528d983c1a156a0434 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9a39534385d4ef28dc88e0d2dc9f44f : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7f1a4923be42848688a9ae8c0104f0 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/262b3bd6482e4d398255cd2b4bf96375 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36fcf16dcf17483089632e27ce6ec0c4 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51291f66b71742bd8d67de1e2ce545bd : SUCCESS in 9m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c103b32128e54be5b91cbedb9792ba46 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0813746b895e4a4ba4e1bb1318663148 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0bb57f39c9594a27afcf70cecdda8a72 : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd81028572c5498f891ffd5f56e36322 : SUCCESS in 42m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f46cdb40e1fa4957a2798e22ab158f85 : SUCCESS in 1h 24m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8465dfd715244f99ac152496cd7f94e7 : SUCCESS in 1h 05m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b47fe5645fa5474f837d27e8694e94be : SUCCESS in 53m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7215e34c5ec74d188f55fb6fe6a84420 : FAILURE in 56m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/25d3dfbfe46249149a016bd8510a70d1 : SUCCESS in 1h 31m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fccc361144c49f4aa98ce7589c9ac1e : SUCCESS in 1h 50m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/377aa6bf2132454d920cca42136fc078 : SUCCESS in 2h 09m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f098c9ea277b4cd489d44ed438882b99 : SUCCESS in 1h 59m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a637bad822c6d7abc7c0b698a20f93e24ca1ba3b","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-06-25 21:47:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cc5bb4cbb6d8d7ef98a406d9613dcbdfaa360387","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-06-25 21:50:25.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"600d50be64d5dea5c036dd35dcd142be82c7244c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:25:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0770518192364223a2a284ca1dd67d9d : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68f5e7fda4804d5fb4d67c15c1848bde : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f2ee15718d94eab8dae1573add20b4c : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889323048c364f66a300b099a1a5ef0d : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d316462b10747c79284464f257c82fc : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e74c941ec1c146a1a8951605475eb73c : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d48652140354de490e6cc2e97ec48be : SUCCESS in 12m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1cd1bd6f989e42a3b432638f1df6c48f : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/80e45cd26f7c483d856c4f7f68dcd646 : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7970a2aab1fa4ae58a23a786401d21bd : SUCCESS in 8m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4db5a5ff0ec24650acb81a3b34affe56 : SUCCESS in 45m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6388f758ceb54ffeb5c03cc78f71c2b1 : SUCCESS in 1h 44m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91d2dd19a5104589aeba1db1d159be43 : SUCCESS in 55m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7a8b0eac6a0c4898bfec93c88dc5c0bc : SUCCESS in 50m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75a588159f24443495ba0b86103000ad : SUCCESS in 53m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5921dd116cc7403192d56f6492b37b67 : SUCCESS in 1h 43m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0b2b042c846d49709bb1e4070f3c5f60 : SUCCESS in 1h 06m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73e396cdd62f40cca2914f7e0be683a0 : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/782105f4158347c3bf3f51d514ab81d6 : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1c534857f097468309f47631753205c685ffdf28","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-06 23:30:03.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nThis is looking good. One concern I have is about the other provider drivers: https://github.com/openstack/octavia/blob/master/octavia/api/drivers/driver_agent/driver_updater.py#L139\n\nHow can we hook them in to use the new metrics extensions?\nOne issue I see is the current metrics extension is requiring an amphora ID. Do we need that? Can we abstract this interface to only need a listener ID?","accounts_in_message":[],"_revision_number":6},{"id":"2d8d9f0e1dad2e686efff0ec139421b222d8c1f4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-07 14:04:13.000000000","message":"Patch Set 6:\n\n(2 comments)\n\n\u003e (3 comments)\n \u003e \n \u003e This is looking good. One concern I have is about the other\n \u003e provider drivers: https://github.com/openstack/octavia/blob/master/octavia/api/drivers/driver_agent/driver_updater.py#L139\n \u003e \n \u003e How can we hook them in to use the new metrics extensions?\n \u003e One issue I see is the current metrics extension is requiring an\n \u003e amphora ID. Do we need that? Can we abstract this interface to only\n \u003e need a listener ID?\n\nI believe we do -- the DB version needs to store things based on amphora ID, so it needs this info. We could make it ... generic somehow? We can think about it.","accounts_in_message":[],"_revision_number":6},{"id":"f4ff2037ba84451211dd3a6d593990456af0e3f9","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-10 01:03:48.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"383021885ed0fe335e7f0dba290d0515cf560969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 03:14:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a796fe6bb553420498eba6b557dab0ee : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed814855a7ab4d32872654f969d72224 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ecf535c3b9f480bb6834614ae895dc8 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca4b464f2094b499b446041d4faf494 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbbc5b1bd30246319bb2b0faeb8e974b : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0a05d92533a46e2b86c3c0ef543f5ac : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0066eddcea24763b5a1f1f825f31b38 : SUCCESS in 13m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/19472875689b4bb4aa252b7f28c53c51 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/36d816aa6e88495da85cf2a66cf5e199 : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85a21e9dc65f4e018086a21cedfaf99d : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a326686ddeca447a84ee4583742a80a2 : SUCCESS in 45m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10d4fae9663e4fb4b9439c75a807880d : SUCCESS in 1h 46m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b1d652222904fb7b72fceb7aa1d7a2c : SUCCESS in 1h 02m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d69b51ce49eb4526beb46b47e7853fb5 : SUCCESS in 46m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a32d62eb01841cbadf09119151f2ad2 : SUCCESS in 56m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbf98382bc954b33a73a3e8a5556149a : SUCCESS in 1h 34m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/333f8c0f38ef4765b886e0ffbc61198e : SUCCESS in 40m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6f94fc7cddbe415caca8bffcbc946a28 : SUCCESS in 2h 03m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/906cfc4073ab4ece9b9785948b938a26 : SUCCESS in 1h 06m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/31f572df8b4f4d249c4690afc340e569 : SUCCESS in 1h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c3426809f024cc42aaac099a142a10a97bc5974d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 10:15:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c4c68bc2b722e488fcdab01331c68903f72b445f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 10:17:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2c1d0bc283c0a1efc413cb15153e39386b95f551","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 10:17:49.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6934969319844e90c5b3f41aee6c4959ce57a52f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 10:19:52.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f4795cbb5ef8405156bdc24b76a60d1cddb1e0b0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 10:22:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8a747bc71aea30718a21121d90503a24637e8ba0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 10:29:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bce3a5ec87695614421967baea15ad3d96143857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 12:08:40.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/30b8651c14874ef1803b4a5c6923ec74 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b935e354ce8a415ab5a9328c55fe1706 : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32efe7896a5146428323c547ecc55667 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83587a72a4924ad38c5af54a7bfb64ce : FAILURE in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/820603f27e974f2e9cf199be2709d8bf : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e286ef055dd4245bd7b4c2802bb8ba4 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2432f95ced46b899c9ba3480d0637c : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/684b4c2181f043c08e16a1a5664dc329 : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8184745159ab44e2b93c035b1f2c707c : SUCCESS in 7m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d228f282e4494a38b39acc24a82c9f0c : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bc58e1606f04d6ab973dbffde20cabb : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54595428aff44a31b33aad0bdb840ea9 : SUCCESS in 42m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77a845f9b7d247f083836b937f36879d : SUCCESS in 1h 01m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3afe294ca9c4a5fa5c4be75bfa6b2ee : SUCCESS in 57m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/61ed73c709bd4c20a16596ceb12c8844 : SUCCESS in 51m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a9e998f74984b7cb98a08fc1ed551e1 : SUCCESS in 57m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b39802c80cb49d2aa0a00bf7ad2551f : SUCCESS in 1h 34m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b9e49151a22342fa8329f0680336bcef : SUCCESS in 47m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a55a13ffdc904411bb7d42227d61220c : SUCCESS in 1h 06m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b32244f322f84714b1c7fce019949477 : SUCCESS in 1h 11m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ae136a88902c43109a20beb60fde405b : SUCCESS in 55m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c558db7bbfde5ada12f3fc8d1f786a0708a444ed","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 16:58:40.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"876b75d723f4b79d5e3f909358a6f0d7a38724b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 17:00:44.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fcdec572d4f4b5d5c98ea557aa663a2744fef7dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 17:20:56.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7570a05f5103d630346bf3f63d7f198dbcc6223a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:29:42.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a37ed768b05d4c949beb8a2089fada8c : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/491b91e22a4e45189b42f81820182fd0 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/085357cb1a4d4fe29bf650700ab953e0 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee97eb9e83b649858240257cbe10e812 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8202a4b16dd741d4a0e98195ba2cfff5 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26622e5eea8345989639173f8bd02dbe : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5131bfdfad7f47a883b169a7702309d0 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3753fe5f053c4c7ca7c294c619fbecab : SUCCESS in 4m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0a2166cc6d2b410b87a71046107c5452 : SUCCESS in 6m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5577567a38f341419491c009b0648c2d : SUCCESS in 10m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e7b12e6f666493fa4302d77750d126e : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01404e013471488e85108cdb83cfa6b3 : SUCCESS in 47m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82f174954c7f44acbc6d86af9e5846e4 : SUCCESS in 56m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6b7dc04b3a7467d96afa06bff1fc094 : SUCCESS in 56m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d5dce3b284c245da80c4c693ccf832b6 : FAILURE in 40m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fa054ca9ad249b3a09053a5cf87afa2 : SUCCESS in 31m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/23a259e3acde44d98c4401b3fecea4c4 : SUCCESS in 1h 28m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/57ee9864b0904d8082617936f26d26d3 : FAILURE in 57m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2aa4e38a788145c3ae78987ab068e061 : SUCCESS in 1h 00m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9ce338ec162c4ce18aeceb73c765dc6c : SUCCESS in 1h 07m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1eada65834a41a5b1c16ce953d7cc62 : SUCCESS in 44m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7957f96f9f464cc582f52fda08bef764 : FAILURE in 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3f4967cea9cb28381ce2a9ea1f94bbb0d2e391fd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 20:35:37.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2d4490424c23b45b483a7b1af94e95745b8f15c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 22:25:11.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f58a87bb888489086480337275d98db : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb41ad05302e4b109dba6ece4970b953 : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ba36d369faa4891a7e46850200e6301 : SUCCESS in 27m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7d6d86580e43a996636b345aa8d193 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28aa06e9c35f4b1faf6dd1a6b1c8cd2d : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d26b3460fd74065b2772ad4519b84f5 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6617249919d54c57a8899f067581c410 : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e66bbf7d232439b8eae2152acd7fe1d : SUCCESS in 7m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9347c24ad70541b99e861074e9ce039f : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9da0b1f3b58742ec9b6f45124c111af3 : SUCCESS in 9m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf324f4990dd429a85c649dd7cfc7399 : SUCCESS in 10m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7551f7fa38924a1b9c8792ba68e02760 : SUCCESS in 44m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd74e3e4655f4e2ba542f28a302d90e7 : SUCCESS in 57m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8c15d2914f14e1ebe5ff51dfe12544b : SUCCESS in 31m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e14738f859644194a14a2e3864d0fd42 : SUCCESS in 32m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b48d77499a2f4fc7853910172cae1612 : SUCCESS in 39m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/06a848e20421406bb15c496941b9323e : SUCCESS in 1h 41m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e19a8822e2684165a58fb5d41095907a : SUCCESS in 50m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40a297e2e61c42d682dcd6709bb06d49 : SUCCESS in 52m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf78943930dd4329b25b28e131b5488c : SUCCESS in 1h 14m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b8fd8386965f4cf68d164a5d12f18db6 : SUCCESS in 46m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bd4e17aa242349c18c38fabd881dd739 : FAILURE in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"23b4f99408385fc359b370fbd4040ddd849ebc2d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-20 23:38:57.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"381e73dbc96e7a00b964fb6c278692b33a15ba26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 01:32:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/80d7676da76b40d0b159acf409f86b9f : SUCCESS in 13m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/792c24bb49fe453bbb72eb2e77559503 : SUCCESS in 11m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa87e896d3d74b9595cf65150d85d42b : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f72e8023155421587757f628248bef7 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07a2f2fff1db42e391c8739e815acded : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb0dc355a2f94930948f7514c678bb3e : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6890220044174df2b8a5ab1da5e3f39b : FAILURE in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/462dc37c5a8a47968dd4b2b99103569b : SUCCESS in 4m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/95a80b58e95a4b4698c3e7755b5e7a9f : SUCCESS in 6m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ce0f0290c27c4b1bafe5f933cbd6ae03 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ff025cdbc634907b4c2bc123c8e5aac : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/acb0ed93acfd4ab8a94aa83a322e1953 : SUCCESS in 47m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2839ae7bde247caa77634ccc8f84786 : SUCCESS in 1h 06m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0db327736c794c639a53018e6a6b30f6 : SUCCESS in 32m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/81fa52c6c3e14c20bac94a0a4546bf23 : SUCCESS in 32m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d17429d1589b46688ed2721f80dd6c6b : SUCCESS in 33m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f2ae2809f26843dc991623e0ee80d09e : SUCCESS in 1h 36m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/555a57e3baf24025be55c5b6ceb559d7 : SUCCESS in 1h 07m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74a6fa02df7e451aa90ba11cdabc6c4d : SUCCESS in 1h 21m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50412e474c9346508e57438e53cc27f4 : SUCCESS in 1h 04m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/391736e8d73d4ddc9eddd46afb80fa89 : SUCCESS in 1h 14m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/699db283405a44c4b78731c1309747bb : FAILURE in 1h 51m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f5b3fbfdcf2b65fc8a27ab3bb404ed54d967d3d7","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-23 17:51:20.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"03d9835e1f633a51f66bbf55b1b09bf48b7ee18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 19:52:06.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0270181659414262b5aeb5e556ee0010 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/446d3d011dfb4ca3bb569917cff07a5d : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2a4ea49b77248999a835affcf1a1ebb : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/def0c0ddd11341d2966905d263fe819b : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb4222e1a11a44e8889d12037ec6009f : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8016d9535ac34f6d90b2f99c61a5e39b : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8616af6a3cd4da39e58e642c7e48ef8 : SUCCESS in 15m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ac35919931341478df956219eecece7 : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/01985f828b19492a9cf5139ad3d06963 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e7827e9c9eb44ed5810479ab8f2e1ebf : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1c11e890ea2416dab14f9271c581090 : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b923fcc8a46648c99909a9275386b31e : SUCCESS in 51m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2d1e129ceae40f79902955a510267bc : FAILURE in 56m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7520a336d5bc4ea09559de19a59ea5d7 : SUCCESS in 31m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dcbc0665bb604962a1e2d0dc4391d6cb : SUCCESS in 41m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea0cd35f850342ebada63b12bba2307e : SUCCESS in 39m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa505a9a23934991b3740925e31e32bc : SUCCESS in 1h 43m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b5ffe9504c7d42f0a710dbda1fee23a8 : FAILURE in 1h 19m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0dc72624c7347a5848640da757fb7bf : SUCCESS in 50m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9d109304da9f463387d9ed4e4d80b0c2 : SUCCESS in 40m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/183010ce3b5e46d7a138671c86b32ef3 : SUCCESS in 44m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6ca3532861be4e2283bdd6a5c26f0a27 : SUCCESS in 1h 33m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ced8715a1fad6ce6f86847eaa069b286fa392a32","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-23 22:38:45.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"74789d8b4c9b995b6c8ad8962e15aa43624dcb59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 00:31:57.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eb9daab6bc6f44fe968aec51a6254148 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97b1df96e59c44f6bffb6691d9cbbe45 : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e220f612a6fd472c93a190193595e9aa : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d873f96d1f421b94515cb2d352cf02 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10a42c13b87145a1be45a254c72d92e2 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1360376e4ae348129db5d44632bf134e : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f14579fb9f4b298ab87ba033fc320c : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07bf897527884ff9b96c619f14f26bc4 : SUCCESS in 4m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fc2ad9c49f34434984b469d855b5e26e : SUCCESS in 5m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0a4ef92d022747d0bd20a433f2aebc24 : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54d0400881164e7896f9799c9fe5abc9 : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ab3fb52a5de4ebcaa12211167ea4d8d : SUCCESS in 47m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ad97f98da60450185052c25401fca68 : SUCCESS in 52m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca63e37025f940feb38e8ec99b858e17 : SUCCESS in 41m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c704db3e6bda44d08b1b49b861b3b197 : SUCCESS in 1h 09m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc821d0c07f744469865d6efa24d5d3c : SUCCESS in 53m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95ed6a4e06c8498b88c34490ad12a13a : SUCCESS in 1h 47m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/502c485489d04a07bf81170f1b426663 : SUCCESS in 59m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/158d173854af44ebbf7b9046b3038585 : SUCCESS in 1h 05m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/841eacc692a44210bd02605088ba0319 : SUCCESS in 55m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d7f750b6499d4944881c3b3613920674 : SUCCESS in 58m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bc0e371888824a36a56ff704307badf9 : SUCCESS in 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"52ede4b32df73350598daf64397d23580c94562f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-24 22:49:13.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"937b92bb076dcc773928b554a627f8b13aed4b69","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-24 23:29:15.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"bff55e4b22b96d8b1a98ba9f91e6726892200b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 01:46:28.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/48ce032528a8494b8b87fa6f571e4354 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b652b658bcb140f69e86b4119ab7bb68 : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f398b8ff68314adea9ec6d57d30285ef : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b060f2b680d14ef886993df02916190b : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf1dc6e288441d8a64c1585a68c56c5 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60039b8e3bc948319c6eff0cf9cda7ab : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016ee6482a604deab449787894250b54 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f367d29a224c258c668a2ebb71689b : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1905c6330f274c28a4df32a16df75cf6 : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/83e1653c5e6d4efebe61b4f90e162a3e : SUCCESS in 9m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75f5a3c6b6e1499fa4b4bde35fa5a2a7 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db9fd7ffcfed4e3c9f95c23c4294188f : SUCCESS in 45m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e38148e0bfd4e55a616464c2a1bc173 : SUCCESS in 1h 04m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b41000bdc5ed40d4b078b4633e777ece : SUCCESS in 1h 00m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ac92f77762f74f38b044a519fc7295d8 : SUCCESS in 40m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cf890e194a343978dfbde058ccb085d : SUCCESS in 1h 00m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c9091aa7c804edd929c3243ea81cc52 : SUCCESS in 1h 40m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1b946aeaec314d69b822aac9fdd4358d : SUCCESS in 55m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c826eee77c25473f86faceb895b31c45 : SUCCESS in 2h 14m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/077d64431afb42659ef4ec4710c590c6 : SUCCESS in 1h 12m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96300e7dfe6e460da6a624df633209e1 : SUCCESS in 52m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a0f60c89fc304e75b92ad4daff5cf3c8 : SUCCESS in 1h 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e863dd186a3bbd99dc2963153ff2c882d2466a8a","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-27 19:52:01.000000000","message":"Patch Set 17: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"7ebb16491c4d48f2b42839403f469a457ff3cfe6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-27 21:15:31.000000000","message":"Patch Set 17: Workflow-1\n\nNeed to rebase this, and also need to look at adding a few more things to the model class we pass through to the StatsDriver interface.","accounts_in_message":[],"_revision_number":17},{"id":"1ea5e7e1fdea9673d323ae4e2b7d2008f429256b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-27 21:16:20.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6604b70df197df08140b1f9c8b46cef98191eb74","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-28 00:47:50.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"7aa062f68081e9c59ddad2b9f88a76cf16da8d6b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-28 01:31:26.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"2fa984811488ebbb89713f7542329cbaedc749f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 04:06:50.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df4aa2e652dd4897abe6b1b3be8f4ad6 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/901fc2a98fcb43e3a04c13d721bb81d6 : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5597fb5d7cc4abf9bcc426558f767e1 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b5e95f34004099b86130820a3de4ea : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79007059128b4956a0d0736bead7e7bd : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ff2f25b8f4c4d0aae4a9a311276fde2 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96107df1bf341818c7885a945149ea4 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dab7b88c7023479e86ad303fa3fc8b36 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb480cb2cf0745f681f49ff4d647cfe2 : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f3c2cb7f57454fa59e5019e4b170f1b9 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5304ecfaaae48918a326aa370b2da43 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a78c69c6891484db0836ddca4dd428a : SUCCESS in 46m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a773757d6f3f469792c2c526d8501b40 : SUCCESS in 47m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7697f129ab04cf4b6721cb41450f5c4 : SUCCESS in 59m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4b20e4e4d30340dea345bad336b22d0c : SUCCESS in 29m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d23fc7710734cc4adfa22717e31e3f9 : SUCCESS in 35m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/633d5b8575914abd8a47fc8d8739621c : FAILURE in 1h 35m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80a4be7cb630491a830434753f6bb6bf : SUCCESS in 1h 04m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/795262a4cd994a9ab0fca388e020575d : SUCCESS in 1h 07m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eec461aa31694d97b88f808d69805e56 : SUCCESS in 39m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eeb191ffe07f4ac88311c288b4f29f20 : TIMED_OUT in 2h 32m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ace936623eb34ea7addfa1da188643bf : SUCCESS in 1h 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"647a934f955bbc0b65830fff0200b8fe131f5ce3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-28 22:05:47.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"50045a9df7ef1597d34002fcf32fbab8515ea14b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-29 00:43:53.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"19ce95315ded46633fef2d1aeb1ee0dba42cad4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 03:29:05.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2e9fde235cf94c1a83e570209c041e65 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0bde468453f47c18bb047db6d1131d0 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b90ee0be9c94cf28696b511ca75d8bc : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bccbecf34bb4b85b32fa6b2e8332244 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06615d6098db4c859d1f39e5b563db60 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab952c74be4f4d8c9149bb00410b0596 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51e1b4baa8044a894ee21519d5f1157 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d19051c9e93545caadb9bbf94e959624 : SUCCESS in 7m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6ceee9a24f3e47f1acbc1787565f6d0b : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f120c670f5e946fcb790f21f7502a398 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59614e11da154e87884b8675ea2c7fbd : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8d1317bb64246308a8d965f7b21b926 : SUCCESS in 32m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08b01ba0c1314112a46b48bc2b1ea797 : TIMED_OUT in 2h 34m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49849b08115e4ef59431e9c9630ca545 : FAILURE in 1h 22m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0a2a0e1b29d641e8b43644029d8a95e1 : FAILURE in 47m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d968e8aed09c4ff38de22dc6e0a0d264 : FAILURE in 1h 05m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c119578677c4e29ba3b6209afa70ddf : SUCCESS in 1h 28m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0deb12e9f50f4f9493539837554e8249 : FAILURE in 2h 35m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/337fa47516da49af993440d73b202cd7 : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c55d9b454c784357a1c9167b46f52427 : FAILURE in 2h 17m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51b17602aeed4175986732670fd6b9d5 : TIMED_OUT in 2h 34m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f47d865b75f945728ce79df61a67bee6 : TIMED_OUT in 2h 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"3824dd12f3d0e7fa73d025d93ff9e9bc85c01c11","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-29 23:31:30.000000000","message":"Patch Set 21: Code-Review-1\n\n(6 comments)\n\nI am -1 on this for the test coverage issues and the comment I think is no longer relevant. There is a meta class that I think should be in octavia-lib as well.\nIt would also be super cool if we had a functional test that \"method_maps\" the (test sqlite) DB and noop stats drivers.","accounts_in_message":[],"_revision_number":21},{"id":"2b1c9c50e33970aa9d599e9352f5f0ce770a524a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-30 19:12:38.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"bc0eb7f2c1a3cc7a9141cbd21671201312180e4d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-30 23:13:36.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"06623a55bfab7ae00a68ec96f4127d197f7675f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 02:06:23.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b36ebb6e8b204e019e43291e73f77a1e : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc184329e0fc4b3cab1895bd099beb77 : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/651e50ac2cfe42708bf45908483c9531 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd883e3a1fd241d99f961c99ef91b7d6 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14f777c82b784404af7e496c6bf017a2 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d7d97114fed4be09d4c602ccd8648d9 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e237640c034632aa8cec82cea4b623 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43178c72b17b443e8735f61772bba532 : SUCCESS in 4m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a3b2dcc66bf3406ca84dc6d609718b1b : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4f32994e9b7e4d29a05e5a32b99a3204 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/508676905e334692af3b469bb1081941 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e3a87471f30402dafc4e290bb8a228b : SUCCESS in 39m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9dd766eeade48249cdde7256526d7e2 : SUCCESS in 43m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2b7b9fffe9e4df4832ae16186330273 : SUCCESS in 1h 07m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/39a1f84cd3264f1b9df1007681cfaa9c : SUCCESS in 55m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1bf25c79bf4247d9a68768eed5979541 : SUCCESS in 41m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/930a7a5baeb34fc3a1d10cb565783ab8 : SUCCESS in 1h 33m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d9341f70b06d48a4a6b801fdce09e52d : SUCCESS in 56m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff3e7cc9ac754f21a2fcc3dcaa3b5660 : SUCCESS in 52m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c42a033f9f8743a9a340df528d181548 : SUCCESS in 1h 10m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eac082d20f134028b5ccf1c6e77fb1f6 : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e8c0d6541b434f188a427d648b83f071 : SUCCESS in 1h 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"aa50abd9916ce41c6c2ad83cd3b416c0070e8d4e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-31 11:10:33.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"669d42954503b8f123aed42e00a2d25ffb97b836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:50:50.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77bcb540cc874a80896d24ca16c20539 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cce40a9f0cd04caf9749b9866bc5d9d5 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/891dea4041bb4dbc97c17fa805b5e497 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795a63a12c514672b7a1ae3589d897f2 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e04d90c30ea4a279449436246617d31 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d38b81fe49d24471b5dede73ced96a51 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2933592b7f0c4bfd809a3df5fdde1695 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a4d4ebf0eec4f02ace4699e85fd6916 : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e3871ef43b404bf5b9ddfcc2ee1c80f9 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/54210724bcfc432c85e9bab5bc223fe9 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a13118d36555456ebf8edbb0b38c0ae5 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c3d750c64ac4c819ae2e882947bc0a0 : SUCCESS in 40m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ce0763c999445a18040a46f1898938c : SUCCESS in 44m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/69b0fa02c9c54a81b29c77101862b554 : SUCCESS in 49m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ba8e08d4f5a94c5da36167bb708d36d0 : SUCCESS in 52m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23c5918ea8eb48c98ace900864b01e54 : SUCCESS in 40m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4f07e31a07e942508d594cc4164c574f : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/234bd0caac4d4ed683ddb1fcdc574a3f : SUCCESS in 1h 17m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d878999c33c14b91b577634809362b27 : SUCCESS in 1h 24m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fd0a11262a294137926ea0acd4e083a0 : SUCCESS in 50m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/204109af8eaa4f19ae541937e638cdad : TIMED_OUT in 2h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c047bab1ef114f229540bbb13a7e0d7f : SUCCESS in 1h 00m 58s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"0630b6af6bc3e2b34d585827c988a117ca4157c1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-04 23:13:10.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"8f6f480708c47a952420f5e162ee523d94a62c45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 01:42:32.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f8cc2270500a4507b8cbdf8ee2b8ec29 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44a11d4c486e4a2292da54b035bfd83f : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a32b379a09dd4b258eb73b81ff808089 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e62506568fb4d3e85222aa65ae622b2 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c77d3e022ca44eabad5cffaba7acc6a : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/882ba4a11fa34f0ca03648829c35774b : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c997c3d2df47e698f4d1e0d038dd6a : SUCCESS in 14m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2fb31161135474fbf34b24aaf722b60 : SUCCESS in 7m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eabd25aafc944bf7bbb9a25c5332d303 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d0a5bc8423524f56856957e6a3baa152 : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45bc3a06dca246f1b49d461864f37916 : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1845a5c83f544a4394d00c3cea2a00bc : SUCCESS in 47m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eda4216e93c940a28fcc8dd398abf9dc : SUCCESS in 56m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd009434b1f74f33bbd53546b3a68ac5 : SUCCESS in 45m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3d7b879c3031498bb33956873563ae80 : SUCCESS in 42m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1630dc45c98a4cb68b21ba73f1830971 : SUCCESS in 43m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c25c4a1e26c842548af5b428a600f344 : SUCCESS in 1h 36m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cac754032cee44848ec0d395a1a6ac5f : SUCCESS in 56m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84ae7df1c01d4300b8e20319859f2393 : SUCCESS in 1h 24m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f66dfbbda76c482c95d1b9476646386a : SUCCESS in 57m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dd728ae66bc04f50b9cb2c4d7233e1eb : FAILURE in 2h 23m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2175c679cea34d0f92764a73184dd6b6 : SUCCESS in 1h 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"e6c8f339d7861044033f15f9dcf8a4d6fd7a54a4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-11 00:45:12.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"d4ab4693c87a3ffc7df9d0a53948c6cf2a7e00b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 03:21:04.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff9fe1facff745a4a6da4aacfe8afc85 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daff37c1a80a48ed84aaacb15ac22147 : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeb2abf37f3c4ad8aea955e48e99e2a4 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2ca94f4da104a1f968cc77a6853f4ae : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ee89ac3757d457590150336f5d3e0a1 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/348c3d7f963c41be959cabec2a16111a : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdae3fa07e47406598e8ee2301085c11 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d0fc9fdee2140d9b5f1ff638e519349 : SUCCESS in 4m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5b10d383d944e2fb5ea020592af35c2 : SUCCESS in 4m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/502dc17117254730bbe91b87d7cc87be : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5fe8623dc7394930a56ddc8b9f955e90 : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e13407f2c35a4f45a9148c666ab0c2f8 : SUCCESS in 31m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43c9b345814c44bda64c2d8884cb54c6 : SUCCESS in 49m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3f299deff0f40a98d2372a7afcfd177 : SUCCESS in 40m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6ed22ffbee8a48028552b4bdd6f1039c : SUCCESS in 36m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fd815bcb9e747b98a387cec8fabe40b : SUCCESS in 1h 09m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/12cc30490bc741fe960fc31d20d619a4 : POST_FAILURE in 1h 41m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54f3948fac4748ccb887a6acf7d22ec0 : SUCCESS in 1h 30m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/347f648551484e20800d3014c2fa31ad : SUCCESS in 1h 26m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0a63ec4e23f34163bbc36b9214a9fd0e : SUCCESS in 1h 23m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/20db07f493f34c0bb9ddffe65bc4357e : TIMED_OUT in 2h 34m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d9902d6f2e3e44b6a350e7faa9ea041d : FAILURE in 1h 03m 37s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"02b63f706d609486ea68666d3e78b437be6dcf95","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-11 15:15:29.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"42e8df4b0616c2624c5b84d516cfbacfb041964d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-11 16:48:34.000000000","message":"Patch Set 25: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"98d93b10a175fc6bdafdb1a1a346510cf453dccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 17:50:08.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aac1afafa9df45f4bd1ca0a927d897be : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60196a73b62f492ca5b08fd271ba92c8 : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9b9c4b1c175496692f0aca0efa11920 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4ba0bd972f40649387b8e0566c7d9e : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7dedecde77246cc841dbd6a3b33b977 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d366a75cd434cfba556d1a1b8e43d55 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cf389f20ec84cc08c18ecc85f205295 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b0068fdab064d4b9ae9293a30589e11 : SUCCESS in 5m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9d25e8fa7bd74bbd862491304aea031f : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/30a0d0e794204747a460dcb4562d2c1e : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e28372f524ef4c2186327276862f3ba5 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ce8acda17fe421c8e2113566742bf52 : SUCCESS in 46m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/473002111b16404e860783599c3ee86f : SUCCESS in 43m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4455d2e69a9743e4a2d2724797ae52d8 : SUCCESS in 38m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2340c6f07f04454bb24c8452f6d13c08 : SUCCESS in 40m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/365a276ed2cf45f1a9eb08312cce2dc3 : SUCCESS in 42m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/25d7f3fc1a7b46f2ba8af9680091924e : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e90d1889abd34b858bb0710983f648ed : SUCCESS in 57m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ecf8e41424741f18ab3b5311203448c : SUCCESS in 1h 07m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3e76b126b37849a9b60a73d6a643354d : SUCCESS in 43m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ac2d351b097f40a1bec6e0b4ee7c7923 : FAILURE in 2h 30m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a40ca1a431064975b914fb5b3476c7ca : FAILURE in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"c8d327da378fc668959ff8f839d283d4b5b17ec0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-11 21:51:43.000000000","message":"Patch Set 25: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"a1a95eeb66eae4141aed30fd7392ff47eb8ba2c2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-12 06:44:31.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"df86764a45006ba805081f4d4f5a3c7df8669638","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-13 13:05:40.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"be96a4657eb3130f28aae75fe59fe2cdcf23cb6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:33:33.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/91efa8d931674a0da073dcdcb562d088 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/596e0f33c0064b4ca7947ee2498a412c : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5841f0ef2cce4598a19fc6db3dcc1afe : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb00c0e249574939afea3cf8f3cc1524 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1bcfcd2095b4edebafdb9b982db0e32 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42f606b580564c79bc46404d46535d10 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dce14b08e634f67aef4cf0384aff91e : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cad706451834ceaac83ec2a5fe93110 : SUCCESS in 6m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e9f0eb6abf8344788ff5f4e0f41f0e2e : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f44d7cca1e404fd6a01c4653a0b17993 : SUCCESS in 11m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2886b4e13ae44ed7a55114dd7bd62a08 : SUCCESS in 11m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5bf4188cf8de4f2c8dbe5db22ebf0a7f : SUCCESS in 53m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb13f8c9e1094de898d28708a126c83f : SUCCESS in 55m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f44f7b648d4b4adfb283ea89e26d200b : SUCCESS in 41m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bdb16c97c41b474eb5ebbafab59feca2 : SUCCESS in 41m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/254c65a56fdc4b7daf28b34b8700bcef : SUCCESS in 30m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ecac7e0eac542c89ae725e85d45809f : SUCCESS in 1h 25m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77e1403a01c140e38ae0bc24bce8fae6 : SUCCESS in 1h 22m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56bceafc4a3b451990d898fc6706d503 : SUCCESS in 50m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/468a1eba5d1f40e6bfd089fd3d360c8f : SUCCESS in 53m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d5a85d408da1469c97eb29313a055a8f : FAILURE in 2h 14m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b312026f491e45609bc47cf10c99de31 : FAILURE in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"9b9d61faba4252cd8182e3e7a5e8df69c93e4b1a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-13 17:37:49.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"158397c625f151a6351a5059dbd54fd7b76c8ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 20:14:55.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/16476f669207460188f8d3bfe7097239 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/779bbf96f0ce4590b76f8a7d3ccd6cea : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d97a1ccab874e78b4653d979f21d0ca : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7de75de2e4b4ac98093ce2091a094b4 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bdba7c73d8c418c8f0891bc14fcbe94 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/478dae6717784b55bd7dce429b63d3ee : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bbd418d4ce42099e76262ce6c31a50 : SUCCESS in 16m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a75789c4370f4c7ea385171a1b7ded04 : SUCCESS in 6m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7a88de9e911046c78360d902dfa37e86 : SUCCESS in 7m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4f11527424974b0f98a9adb643208807 : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88289572434b461892e7d61702ba57a4 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3a20e3d6811d4a3b8fcfaa6a6da27df1 : SUCCESS in 44m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee850f2b7dc64248ad71f576a9aacbc4 : SUCCESS in 1h 11m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/752681879aeb4bd9bf8348d7f00f282f : SUCCESS in 1h 21m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/21a0dc2fb97943368a550e47932bb65a : SUCCESS in 39m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c0919f63b7a41b3adaee4e6d28cc9f1 : SUCCESS in 1h 19m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/476c9dec083b4aae9be40641c975a851 : SUCCESS in 1h 19m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61b2b701418a4099a46b6cdd1e7e83e6 : SUCCESS in 1h 38m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/163a39502a6d4b96b02cfb21040179d7 : SUCCESS in 1h 07m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cbfe8e98d1b64cc783b71baacb58aeb2 : SUCCESS in 51m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/55852212318e48679d52ca5bc2265815 : TIMED_OUT in 2h 33m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4b4e80d1890c4117b26f91df82b4446b : FAILURE in 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"3c46daa855db30346cb7994a83822e191c4c1230","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-03 23:42:20.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"1d5c1a336f00f7315b16d2361d3325608833009c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 01:14:50.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a9987ebc404c4beeb28bbfe1c569cd27 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/814657c308424d1ba2b13d66d776c0c4 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbe81bd732b54338ac6271c24c9e1509 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe867b1e5b54579bb58669e5a720aa4 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2cf249de1514d048107f0dd92752984 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eda1dbeb2fe24c97a5c1598f219b9c68 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0226134a91d542d79a2852d440b48ab0 : SUCCESS in 14m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00fe86a9cb264589828a2cbe7bb471f1 : SUCCESS in 6m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b3d29f80a275459eb7956dc5d87649ba : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e9592b801e9e42dcbd43ff43a67487f0 : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a865051b5694e1a9618a08427dbd7e2 : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95ad44e1a2514b97a8ccb619411fb2eb : SUCCESS in 48m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2222a234b184ce3a9781dc7f74c9333 : SUCCESS in 1h 10m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f868403c81c4dcbb76787a017e6a384 : SUCCESS in 1h 00m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/43d5c8c84c714ac6b2bfb7fdff59a9a8 : SUCCESS in 57m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa1d95b8458946b885949ad10c88deb4 : SUCCESS in 30m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/96cb6bcd8365402cbf013cb9b495ab9a : SUCCESS in 1h 04m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f0fe2db76584b81ac4674e5d000f80b : SUCCESS in 1h 17m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88e4b871792e4b5cac79940fb7ee77d9 : SUCCESS in 58m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b89877af43d245ca975a0aace061e9c2 : SUCCESS in 1h 10m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/376fcb58da6346efa090520d820a9a12 : SUCCESS in 42m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/be34cc12ce5f4e50bf455e8eb71b6b20 : SUCCESS in 55m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f7fa5f24df7d4910aef9cacccc63b3b7 : SUCCESS in 1h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"842cb30bc6e6ae355967fb56c273bce8e12e2dc6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-07 21:33:39.000000000","message":"Patch Set 28: Backport-Candidate-2 Code-Review+2\n\n(3 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":28},{"id":"6d98a48d079b436ef0bd021a92d203771753121b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-08 21:39:53.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"d23acfb1fcad378bb3ef8154fce2f9352147f684","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-08 21:48:38.000000000","message":"Patch Set 29: Patch Set 28 was rebased","accounts_in_message":[],"_revision_number":29},{"id":"839885c685dcec77aa4ba54f25750b05a53a5721","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-08 22:32:17.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"8dc5b1855a20a1a7f8b227891b1d56913ddf348a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-08 22:34:33.000000000","message":"Patch Set 30: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":30},{"id":"7ba91d646b69e61657415bfb0177a6218b658d4a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-09 00:22:05.000000000","message":"Patch Set 30: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":30},{"id":"7fe7c0b3662d3c1f33097da09f22941443a7e568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 00:37:50.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b7a1b2140fc44ea486872440912b38ca : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f15533263a7843db8e93cb3ab6d5cd5c : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/604610f9341a4fbf83495c65c02972f4 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8284a3a87a41f884f2166200c742da : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26731b73e3dd4f4e97783c4ed935a00c : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d85e5cfcb744037a048f51d2f53308f : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b8a95fb28c4085a13d2ba0b8056236 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2312e518bd24f77b823d8a59a5207e2 : SUCCESS in 5m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/10686f9b20314a4b8628ff2ffc87ddea : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c1970fef401a450c8639334ccacbcef5 : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76f118e231094d05a584805c5183a78f : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b845e7dff1a14bcdb50a30477c761d30 : SUCCESS in 45m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0ea226f4ef24ff5a3d23390f40ee387 : SUCCESS in 1h 11m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ea2335644904fd38494ed4c8dfbef9d : SUCCESS in 56m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/81f54b6b879e46b3a4b0979afcb233e5 : SUCCESS in 51m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a91ace48520457b9cd06874f3918fc4 : SUCCESS in 51m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3af9110604c346dc92ca639766d9f6ec : SUCCESS in 1h 04m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/86c6bcf2bed844e6a74891d473403b65 : SUCCESS in 52m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e58e9b04cfbb4bb88e1f12617285f5f3 : SUCCESS in 54m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8df53116f57d4514a9885f2d8f009559 : SUCCESS in 43m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/53daa7c8f390439fafb9d11228cf62fb : SUCCESS in 53m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d74a25bae3e74770a432df9a700d54c6 : SUCCESS in 57m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/426de14459814ae3994ec224c7c8851c : SUCCESS in 1h 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"ca2ab536b591734a35417abd921f965610330c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 00:38:03.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"5add77572ebac2f87c6fa6763c8fca7da80c0c83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 02:10:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":30},{"id":"294fc3fb090215d574afe74efdecc040ba3da805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 02:10:52.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/033df4240c734c69b611dda7feb8e544 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e410939159364598bc566b65e60ad3f4 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e37f89ddf094fd8990036c14d54fe7f : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b39104948cf6405b85b5d0770f62f741 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55afc7bd82074954b610385b6e2d3f4d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4efd64be7e4e81999a52ba189b3678 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/965c7c51d17d44bab3454c8e833e46c6 : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/353e6c5096e345bbac83de66770ba333 : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ac11fb531874dd1a28c273ba3ca690a : SUCCESS in 42m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2009aa7a5dd1472fa7f18cf155b19ec0 : SUCCESS in 44m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d3ac9910fa24ff7ae80f4cf652af68b : SUCCESS in 32m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a14cd463ce7846cc9c108c28650e2ea6 : SUCCESS in 41m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08f90c99f6684d24b60e37d58fd9f570 : SUCCESS in 30m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b18bc70e4644118aae23210375bb14b : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":30},{"id":"ae100ea99213205d407085407d8ea462a4096555","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 02:13:10.000000000","message":"Patch Set 30:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951fb924b02c4d5fb2749c962411c03d : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c9b00d51ebc46dd890d18bcd479950b : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"5092597f6b78bc3f02812cd425aa9d95e9501be7","revisions":{"055bfe8579ea0513c0b56ae38bf7258bf516c727":{"kind":"REWORK","_number":1,"created":"2020-06-20 00:44:12.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/11/737111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/1"}}},"commit":{"parents":[{"commit":"77786595e5803488f46e9b4faae02b5bccdfe88d","subject":"Merge \"Refactor the failover flows\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77786595e5803488f46e9b4faae02b5bccdfe88d"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"subject":"WIP: Refactoring amphora stats driver interface","message":"WIP: Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/055bfe8579ea0513c0b56ae38bf7258bf516c727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/055bfe8579ea0513c0b56ae38bf7258bf516c727"}]},"branch":"refs/heads/master"},"29bbdb185d6f5df1911face571886146b6c2b865":{"kind":"REWORK","_number":2,"created":"2020-06-24 00:00:04.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/11/737111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/2"}}},"commit":{"parents":[{"commit":"77786595e5803488f46e9b4faae02b5bccdfe88d","subject":"Merge \"Refactor the failover flows\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77786595e5803488f46e9b4faae02b5bccdfe88d"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-06-23 23:59:53.000000000","tz":-420},"subject":"WIP: Refactoring amphora stats driver interface","message":"WIP: Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29bbdb185d6f5df1911face571886146b6c2b865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29bbdb185d6f5df1911face571886146b6c2b865"}]},"branch":"refs/heads/master"},"55769481cec9345b0cbdd148e861dd8764f75352":{"kind":"REWORK","_number":3,"created":"2020-06-24 17:38:47.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/11/737111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/3"}}},"commit":{"parents":[{"commit":"77786595e5803488f46e9b4faae02b5bccdfe88d","subject":"Merge \"Refactor the failover flows\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77786595e5803488f46e9b4faae02b5bccdfe88d"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-24 17:37:48.000000000","tz":-420},"subject":"WIP: Refactoring amphora stats driver interface","message":"WIP: Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nFixed indentation error.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55769481cec9345b0cbdd148e861dd8764f75352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55769481cec9345b0cbdd148e861dd8764f75352"}]},"branch":"refs/heads/master"},"7c626134b38f64b0131091122e2613b6e3d70b5b":{"kind":"REWORK","_number":4,"created":"2020-06-25 01:28:28.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/11/737111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/4"}}},"commit":{"parents":[{"commit":"77786595e5803488f46e9b4faae02b5bccdfe88d","subject":"Merge \"Refactor the failover flows\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77786595e5803488f46e9b4faae02b5bccdfe88d"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-25 01:26:42.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c626134b38f64b0131091122e2613b6e3d70b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c626134b38f64b0131091122e2613b6e3d70b5b"}]},"branch":"refs/heads/master"},"e469960922e4a6a9a62e38cb6f4df196d2a49d19":{"kind":"REWORK","_number":5,"created":"2020-06-25 21:47:22.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/11/737111/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/5"}}},"commit":{"parents":[{"commit":"77786595e5803488f46e9b4faae02b5bccdfe88d","subject":"Merge \"Refactor the failover flows\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77786595e5803488f46e9b4faae02b5bccdfe88d"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-06-25 21:46:52.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e469960922e4a6a9a62e38cb6f4df196d2a49d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e469960922e4a6a9a62e38cb6f4df196d2a49d19"}]},"branch":"refs/heads/master"},"f8bf58edbeb662f7d911d5eb67b2c872d1a44c5a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-25 21:50:25.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/11/737111/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/6"}}},"commit":{"parents":[{"commit":"41c628a084002017d2003926cf0e25ba3ffeee0c","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41c628a084002017d2003926cf0e25ba3ffeee0c"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","date":"2020-06-25 21:50:25.000000000","tz":0},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8bf58edbeb662f7d911d5eb67b2c872d1a44c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8bf58edbeb662f7d911d5eb67b2c872d1a44c5a"}]},"branch":"refs/heads/master"},"cdcb7823310466d052c95f44bba09cff161c2827":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-07-10 01:03:48.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/11/737111/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/7"}}},"commit":{"parents":[{"commit":"0317855d9d2c9cb1ee2b61c54cb8c8337aa6cb0f","subject":"Merge \"Add a router to the lb-mgmt-net for slaac\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0317855d9d2c9cb1ee2b61c54cb8c8337aa6cb0f"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-10 00:56:44.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdcb7823310466d052c95f44bba09cff161c2827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdcb7823310466d052c95f44bba09cff161c2827"}]},"branch":"refs/heads/master"},"da82d63627e2a026110b569228e0ccd276a236af":{"kind":"REWORK","_number":8,"created":"2020-07-16 10:15:49.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/8"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-16 10:15:40.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da82d63627e2a026110b569228e0ccd276a236af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da82d63627e2a026110b569228e0ccd276a236af"}]},"branch":"refs/heads/master"},"6800c91c5e16679f57f8557a543c715776f9682c":{"kind":"REWORK","_number":9,"created":"2020-07-16 10:17:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/9"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-16 10:17:07.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6800c91c5e16679f57f8557a543c715776f9682c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6800c91c5e16679f57f8557a543c715776f9682c"}]},"branch":"refs/heads/master"},"63b5120386b0dac319a8fb78a04702cc5da2329e":{"kind":"REWORK","_number":10,"created":"2020-07-16 10:22:20.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/10"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-16 10:22:12.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63b5120386b0dac319a8fb78a04702cc5da2329e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63b5120386b0dac319a8fb78a04702cc5da2329e"}]},"branch":"refs/heads/master"},"eed60a2eb42b3ddea4f583c158d5bcde9c7ed5d0":{"kind":"REWORK","_number":11,"created":"2020-07-16 10:29:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/11"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-16 10:28:55.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eed60a2eb42b3ddea4f583c158d5bcde9c7ed5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eed60a2eb42b3ddea4f583c158d5bcde9c7ed5d0"}]},"branch":"refs/heads/master"},"cfee9373145fbef7d502c45d748c2caca16eedbd":{"kind":"REWORK","_number":12,"created":"2020-07-16 16:58:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/12"}}},"commit":{"parents":[{"commit":"1072263b73810133ff2c0e25e4b0a085e8f60f02","subject":"Add non-voting octavia-v2-dsvm-scenario-centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1072263b73810133ff2c0e25e4b0a085e8f60f02"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-16 16:54:04.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfee9373145fbef7d502c45d748c2caca16eedbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfee9373145fbef7d502c45d748c2caca16eedbd"}]},"branch":"refs/heads/master"},"d0c7aed84d2f616993d489d912ddd8f4d994fa66":{"kind":"REWORK","_number":13,"created":"2020-07-16 20:35:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/13"}}},"commit":{"parents":[{"commit":"f14ccb52ed004f1c30c49e87407f2ce569c045e9","subject":"Merge \"Fix listener update with SNI certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f14ccb52ed004f1c30c49e87407f2ce569c045e9"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-16 20:35:25.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c7aed84d2f616993d489d912ddd8f4d994fa66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c7aed84d2f616993d489d912ddd8f4d994fa66"}]},"branch":"refs/heads/master"},"a0edf9bda75f532785571fbb509c60f3e909fb2d":{"kind":"REWORK","_number":14,"created":"2020-07-20 23:38:57.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/14"}}},"commit":{"parents":[{"commit":"c17cc78d6fae308b9fd81ebfc45f43e628d0b3c0","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c17cc78d6fae308b9fd81ebfc45f43e628d0b3c0"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-20 23:23:31.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0edf9bda75f532785571fbb509c60f3e909fb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0edf9bda75f532785571fbb509c60f3e909fb2d"}]},"branch":"refs/heads/master"},"0782c60fe4a29596748c8984a83dfeb991b17d00":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-07-23 17:51:20.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/11/737111/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/15"}}},"commit":{"parents":[{"commit":"1b70a0dbe5514ed79de1ed21985e9af1933129df","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b70a0dbe5514ed79de1ed21985e9af1933129df"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","date":"2020-07-23 17:51:20.000000000","tz":0},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0782c60fe4a29596748c8984a83dfeb991b17d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0782c60fe4a29596748c8984a83dfeb991b17d00"}]},"branch":"refs/heads/master"},"2e8094efe72e5d429e268a306fc1281559a56c44":{"kind":"REWORK","_number":16,"created":"2020-07-24 22:49:13.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/16"}}},"commit":{"parents":[{"commit":"f5584a0a0aa529c6345ef42166ed21508a17eeca","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5584a0a0aa529c6345ef42166ed21508a17eeca"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-24 22:48:47.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e8094efe72e5d429e268a306fc1281559a56c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e8094efe72e5d429e268a306fc1281559a56c44"}]},"branch":"refs/heads/master"},"5cbba60aa1880c3d8172627a56b9be028357584e":{"kind":"REWORK","_number":17,"created":"2020-07-24 23:29:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/17"}}},"commit":{"parents":[{"commit":"f5584a0a0aa529c6345ef42166ed21508a17eeca","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5584a0a0aa529c6345ef42166ed21508a17eeca"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-24 23:27:42.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cbba60aa1880c3d8172627a56b9be028357584e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cbba60aa1880c3d8172627a56b9be028357584e"}]},"branch":"refs/heads/master"},"7efeebcb45adce53ffb9b341a163958d3179b93d":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2020-07-28 00:47:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/18"}}},"commit":{"parents":[{"commit":"5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-28 00:47:50.000000000","tz":0},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7efeebcb45adce53ffb9b341a163958d3179b93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7efeebcb45adce53ffb9b341a163958d3179b93d"}]},"branch":"refs/heads/master"},"8ec513b4409742d5db17a8355279b0a6acb3b4cf":{"kind":"REWORK","_number":19,"created":"2020-07-28 01:31:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/19"}}},"commit":{"parents":[{"commit":"5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-28 01:30:51.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ec513b4409742d5db17a8355279b0a6acb3b4cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ec513b4409742d5db17a8355279b0a6acb3b4cf"}]},"branch":"refs/heads/master"},"d9cf2a79dbf86d7c9d51d2df772cd0e4c42007f7":{"kind":"REWORK","_number":20,"created":"2020-07-28 22:05:47.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/20"}}},"commit":{"parents":[{"commit":"5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-28 22:05:34.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9cf2a79dbf86d7c9d51d2df772cd0e4c42007f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9cf2a79dbf86d7c9d51d2df772cd0e4c42007f7"}]},"branch":"refs/heads/master"},"42ca88b7306c1aa19f0e5bec4de4e98655dcca2a":{"kind":"REWORK","_number":21,"created":"2020-07-29 00:43:53.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/21"}}},"commit":{"parents":[{"commit":"5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a2b83bc7920a18e85fe6c92c8eb927e0c3dc7d8"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-29 00:36:14.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42ca88b7306c1aa19f0e5bec4de4e98655dcca2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42ca88b7306c1aa19f0e5bec4de4e98655dcca2a"}]},"branch":"refs/heads/master"},"e4021ece8c4dcd22a57711ecdce9500cc2c93b07":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2020-07-30 23:13:36.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/22"}}},"commit":{"parents":[{"commit":"d15cccff2fd382475f971a646f5579bd43ef190b","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d15cccff2fd382475f971a646f5579bd43ef190b"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-30 23:13:36.000000000","tz":0},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4021ece8c4dcd22a57711ecdce9500cc2c93b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4021ece8c4dcd22a57711ecdce9500cc2c93b07"}]},"branch":"refs/heads/master"},"e11f859d1a22bf0917b5c98c607606b8ef1c2d37":{"kind":"REWORK","_number":23,"created":"2020-07-31 11:10:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/23"}}},"commit":{"parents":[{"commit":"d15cccff2fd382475f971a646f5579bd43ef190b","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d15cccff2fd382475f971a646f5579bd43ef190b"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-31 11:10:15.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e11f859d1a22bf0917b5c98c607606b8ef1c2d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e11f859d1a22bf0917b5c98c607606b8ef1c2d37"}]},"branch":"refs/heads/master"},"ea85b4b22e8e8db6274fd89b2e0c25649974c0d9":{"kind":"REWORK","_number":24,"created":"2020-08-04 23:13:10.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/24"}}},"commit":{"parents":[{"commit":"d15cccff2fd382475f971a646f5579bd43ef190b","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d15cccff2fd382475f971a646f5579bd43ef190b"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-04 23:11:57.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea85b4b22e8e8db6274fd89b2e0c25649974c0d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea85b4b22e8e8db6274fd89b2e0c25649974c0d9"}]},"branch":"refs/heads/master"},"4f6234229aae6a56062c2afa4e9213e9a5ba53fa":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2020-08-11 00:45:12.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/25"}}},"commit":{"parents":[{"commit":"6a762f8a072ab3b3eaede32c6d471d874f2ad945","subject":"Merge \"[Trivial]Add missing space between words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a762f8a072ab3b3eaede32c6d471d874f2ad945"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-11 00:45:12.000000000","tz":0},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f6234229aae6a56062c2afa4e9213e9a5ba53fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f6234229aae6a56062c2afa4e9213e9a5ba53fa"}]},"branch":"refs/heads/master"},"9ed974928d42b1ec53850005f6628d653a8a7a1c":{"kind":"REWORK","_number":26,"created":"2020-08-13 13:05:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/26"}}},"commit":{"parents":[{"commit":"aa52cb39caf05d9ce91b6569271e0b831c8a589b","subject":"Merge \"Add line terminator to certificate list file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa52cb39caf05d9ce91b6569271e0b831c8a589b"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-13 13:05:28.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ed974928d42b1ec53850005f6628d653a8a7a1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ed974928d42b1ec53850005f6628d653a8a7a1c"}]},"branch":"refs/heads/master"},"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12":{"kind":"REWORK","_number":27,"created":"2020-08-13 17:37:49.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/27"}}},"commit":{"parents":[{"commit":"aa52cb39caf05d9ce91b6569271e0b831c8a589b","subject":"Merge \"Add line terminator to certificate list file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa52cb39caf05d9ce91b6569271e0b831c8a589b"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-13 17:25:53.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]},"branch":"refs/heads/master"},"0e6a6532c3ecaf1b490227add41b8f552b5467ea":{"kind":"REWORK","_number":28,"created":"2020-09-03 23:42:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/11/737111/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/28"}}},"commit":{"parents":[{"commit":"4a4a2344de09d61aa253174d9a7ca242d116913e","subject":"Merge \"Add a message for the amphora image size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a4a2344de09d61aa253174d9a7ca242d116913e"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-03 23:42:19.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e6a6532c3ecaf1b490227add41b8f552b5467ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e6a6532c3ecaf1b490227add41b8f552b5467ea"}]},"branch":"refs/heads/master"},"3c8af16e9e4637039e59eb54274ab4b3bac35e8e":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2020-09-08 21:48:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/29"}}},"commit":{"parents":[{"commit":"954cf8afb3d4dc21c4cfe7729427cdbb7c98bb43","subject":"Merge \"Add DIB element octavia-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954cf8afb3d4dc21c4cfe7729427cdbb7c98bb43"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-08 21:48:38.000000000","tz":0},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c8af16e9e4637039e59eb54274ab4b3bac35e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c8af16e9e4637039e59eb54274ab4b3bac35e8e"}]},"branch":"refs/heads/master"},"5092597f6b78bc3f02812cd425aa9d95e9501be7":{"kind":"REWORK","_number":30,"created":"2020-09-08 22:32:17.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/737111/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/737111/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/737111/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/737111/30"}}},"commit":{"parents":[{"commit":"954cf8afb3d4dc21c4cfe7729427cdbb7c98bb43","subject":"Merge \"Add DIB element octavia-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954cf8afb3d4dc21c4cfe7729427cdbb7c98bb43"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-06-20 00:44:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-08 22:32:04.000000000","tz":-420},"subject":"Refactoring amphora stats driver interface","message":"Refactoring amphora stats driver interface\n\nPreviously the stats driver was responsible for parsing the health\nmessage, which should have been done prior to passing the stats to the\ndriver interface.\n\nRemoved the driver interface for the health updater because it is core\nOctavia functionality.\n\nStats drivers is now a singleton and can load multiple drivers.\n\nBoth the amphora health manager AND provider statistics should use the\nnew driver interface.\n\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I3a013aebd1eb89cd4f983fbf4f8ae8d6639548cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5092597f6b78bc3f02812cd425aa9d95e9501be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5092597f6b78bc3f02812cd425aa9d95e9501be7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
