)]}'
{"id":"openstack%2Foctavia~737155","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","subject":"Fix batch member create for v1 amphora driver","status":"MERGED","created":"2020-06-21 01:31:46.000000000","updated":"2020-06-26 19:25:50.000000000","submitted":"2020-06-26 19:23:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737155-1593199421047-88eb4db9","meta_rev_id":"1c54ebb4cecb67a95173b13dd1c4b0928fc49773","_number":737155,"virtual_id_number":737155,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 19:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-06-26 16:44:49.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-06-26 06:25:32.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":1,"date":"2020-06-26 16:44:49.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-06-21 01:31:46.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-21 01:31:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-26 06:25:32.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-26 16:44:49.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-06-26 19:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38674f5c3223a3243a7b58bc648f2a8d4df427e7","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-21 01:31:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9d0c0d4482dddc0e6c48492b2dd9c3fe738b61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-21 04:05:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e35aa65a5473431aaa572e4f590af5bb : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a911a46101441f5bd727c96e3e22498 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838dea8f7db94a58869aa14e3be1b774 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec99e3188a034adba06ba24b3bdea179 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f167bb3504b940a187c5921e0f5ef5dc : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3ef5f1296344b248ef41b89d949c449 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662c1b40e0d148f0a9720ca25d8bbbac : SUCCESS in 10m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b91b936ec37d4db4b147251690e910cb : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1059079a59cd4944b8bf1f3decf233ec : SUCCESS in 7m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a6b36aaf62ef4ba2805337a3bf88719b : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca534d95521541cc9e087c093c52482f : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7fa8097df4bb45a3bdfa1ca1e8658e0b : SUCCESS in 48m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/980275164c204d5bb77944c1a687bee7 : SUCCESS in 42m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb47617248554383a78713772c6b77b8 : SUCCESS in 1h 41m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/662dcf77f43f4a7281e3f9040bf4b206 : SUCCESS in 1h 32m 53s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a94cfa1b1d9846daac971cba92a785cb : FAILURE in 15m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/75a3b289996b405785daf8ef6fbf8f7a : FAILURE in 51m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/608a7662f61846b1ba29f394c1af4af0 : FAILURE in 49m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0fefbaf4a5345daa11401a48a5b23dc : FAILURE in 28m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f33345456a534cce8d33c69d330d16a2 : SUCCESS in 57m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dc01f8abc6a94f74af0540fbfb510f27 : SUCCESS in 1h 43m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5049d28204ee423b9cdb009dbf7c5239 : SUCCESS in 1h 01m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/816e846561434089a41e4906a4eb1e8f : TIMED_OUT in 2h 33m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/da47a2b74b0b448abc3edbf4e2defdf0 : FAILURE in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"810dc7abe7a0826a02d91a8766751877060fc894","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 19:35:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b41758d81892cd2e02c1ba722752ad329289088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 21:37:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a27c56a633fa4144a3f20baf2f3402b4 : SUCCESS in 13m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f64d5f4ace5496cb863abb7adc841fa : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f3ec24c9d2497d85dc23ebe291740c : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c0de0dacff142f690ea74e70c9ae0a1 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85d444c858024427ba9a3c24222e089e : SUCCESS in 8m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02f7e6f6bb4c483fbd0fc552bafcf576 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f4828dbd4e4d56bf547bb8d50e5792 : SUCCESS in 8m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/185d259d3b744f6b9b92c5e3eafed5ee : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0116fdedf2c643db8c3c5038717b4960 : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b8a04024bdfe486087f2dfadc61ea4fe : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36d8366eeb2441cdaa4435e560e514ba : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03025fcd5c754a599b0e416c47091682 : SUCCESS in 35m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9fec482822754b69ad41fa1ac1a687bd : SUCCESS in 46m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94f32f19694d4c26bc6c344339b247e5 : SUCCESS in 1h 44m 11s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e8eb1602b5ec47d5884d74ea4499637f : SUCCESS in 1h 43m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e6179adb90e74c04a0a39736cc69b80c : FAILURE in 14m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/827d170f3fee462bb9fad421c0d282bc : FAILURE in 57m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e494b84c420e44b6aa451470cf8b69f4 : FAILURE in 52m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2deaafc4a27415f96879e893d13541d : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eed608a726844649990bf3e6237cc11c : SUCCESS in 59m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a705b8df9a734424bfa3b5efe0b5e8e7 : SUCCESS in 1h 51m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7864fdbbe665442bae3ab649db2f1164 : SUCCESS in 1h 06m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a8cb903a17fa48df9882ef7fe86aff07 : SUCCESS in 1h 23m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/401d87fd79a541e7ae53b9a0ff14b14f : FAILURE in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4eb1b3251b949f26d694dca7ce8140167e17d3e7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 06:25:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"626ec14a12d9fa20f4fb760834baaceb1e88f27b","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-06-26 16:44:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbc160d8a26a60ca15a87af86f086d40f088a127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 16:47:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"914f4dd32b0598b46013f60bec9e85d17dd8090f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:23:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9966239120d44e9dbc359eeea9d50bd3 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdccaefc9076449ead226d4dc76d77f0 : SUCCESS in 10m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7cdcd22401954c1ea9b4c71ba21e5c51 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc5f7f1f47e145fe870fe8085b0ab252 : SUCCESS in 9m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/644b1e1f85b84d51a2fe55b8e424e3a2 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024bac7ee9cd4d04ab53f05870354583 : SUCCESS in 13m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4936203701af403c84b7f6c9af6e421e : SUCCESS in 9m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93faa7f6637b4fa5b3246424a84ab9ec : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7fe6da9370ad4935b294ffaa73a4eea5 : SUCCESS in 50m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a02ea8ccc31e494c8137fb87f33cb6a2 : SUCCESS in 48m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9925702531a8432d921d72e7ec7a923e : SUCCESS in 1h 46m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2d8d2abb7cac4ac3be9ab05d19b25403 : SUCCESS in 2h 30m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/47993859df3d4ee6b134e745589c73bf : SUCCESS in 1h 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"505dc155e64feaeeb77dd5b61b41f8682017d963","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:23:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1c54ebb4cecb67a95173b13dd1c4b0928fc49773","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:25:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca4afc61e18432db0fbcb8846e4eed7 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc2c3c481226253a37635466f13947a6a60111b0","revisions":{"cc2c3c481226253a37635466f13947a6a60111b0":{"kind":"REWORK","_number":1,"created":"2020-06-21 01:31:46.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/55/737155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/737155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/737155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/737155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/737155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/737155/1"}}},"commit":{"parents":[{"commit":"e358637a10e6b3c2df336d8dad42cecbd97e3dd5","subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e358637a10e6b3c2df336d8dad42cecbd97e3dd5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-04 17:45:35.000000000","tz":-420},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-06-21 01:31:41.000000000","tz":330},"subject":"Fix batch member create for v1 amphora driver","message":"Fix batch member create for v1 amphora driver\n\nA previous patch[1] missed batch_member_update when adding database\nrepository \"get\" method retries for new object creation actions.\nThis patch fixes batch member create to retry the database get call\nwhen new members are being created via batch member update.\nThis issue only impacts the v1 amphora driver as the v2 driver\ndoes not need to get these objects from the database.\n\nStory: 2007581\nTask: 39503\n\n[1] https://github.com/openstack/octavia/commit/48e85569f7e43e3dd5f09fd1ef4cb165526a92cd\n\nChange-Id: Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc2c3c481226253a37635466f13947a6a60111b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc2c3c481226253a37635466f13947a6a60111b0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
