)]}'
{"id":"openstack%2Foctavia~737157","triplet_id":"openstack%2Foctavia~stable%2Frocky~Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","subject":"Fix batch member create for v1 amphora driver","status":"ABANDONED","created":"2020-06-21 01:48:04.000000000","updated":"2020-07-24 11:42:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a3e0d2aec2178a90142a1578f6ba1f075f9ff9c","_number":737157,"virtual_id_number":737157,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-21 02:19:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-21 01:48:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-21 01:48:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-21 02:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80a4d95dd212e6ebd982280c69be3c228dd5da07","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-21 01:48:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"60ddb0350aec45b622404be45c094fae35624975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-21 02:19:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9fd431c903a41b7b9154238cd7e9c92 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fdc168214304dbd9837232f2b3765a0 : SUCCESS in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/feb4e74705944bd5a797284fca0a978e : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b997cf617ae44931b70864e82c801cb6 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b45754d1fff1455c86171ba18bb1a01b : SUCCESS in 6m 30s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/8d9a7da84ea84eb6a1caca788b0be761 : FAILURE in 15m 00s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/af0458428030427d87999a09b3c08d4b : FAILURE in 15m 23s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/826318f274664a3b93edb3d28ceba791 : FAILURE in 16m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/434981dea2c94849866c9543d3a8a03a : FAILURE in 17m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e2e8cc6c9f645c29c46fc64c197ed80 : SUCCESS in 6m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8edfac3015314613a25a7229ae93c755 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bea538fd0aae4abab9b2f40afcc9f269 : SUCCESS in 6m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/782ecc1ce5c84e58a3ebc8b5d05c14b3 : FAILURE in 16m 54s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/99002de9f45d4614863dbbf66704e9f2 : FAILURE in 11m 44s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/7843c796f638465a98bdf096f8f96afa : FAILURE in 17m 54s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d4d61179e2e244daaba97ff4ccc43fb3 : FAILURE in 12m 08s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8c3ba250d18b4ceeb6e54c08af650f8b : FAILURE in 14m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f159397ea93242bc84838520ef2845ee : FAILURE in 13m 33s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/1785826329d444cbb7b4e349f05dddd1 : FAILURE in 13m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4c8b694a9de4f0cbf3dc770fd1b0fa5 : FAILURE in 13m 14s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/efdec50f6a90456ebaf62c7d0bd76413 : FAILURE in 13m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/713710737fa646cc89911662b0ff3720 : FAILURE in 18m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e3eb4d0d6a6407ebded60f5b5143b79 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1a3e0d2aec2178a90142a1578f6ba1f075f9ff9c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:42:11.000000000","message":"Abandoned\n\nRocky Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5cbeef4be4d447881ae20aeb7fea689c2a89397a","revisions":{"5cbeef4be4d447881ae20aeb7fea689c2a89397a":{"kind":"REWORK","_number":1,"created":"2020-06-21 01:48:04.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/57/737157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/737157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/737157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/737157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/737157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/737157/1"}}},"commit":{"parents":[{"commit":"126ad303ac998f2fa36de572bb37cb36f366913a","subject":"Merge \"Fix the amphora noop driver\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/126ad303ac998f2fa36de572bb37cb36f366913a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-04 17:45:35.000000000","tz":-420},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-06-21 01:48:04.000000000","tz":0},"subject":"Fix batch member create for v1 amphora driver","message":"Fix batch member create for v1 amphora driver\n\nA previous patch[1] missed batch_member_update when adding database\nrepository \"get\" method retries for new object creation actions.\nThis patch fixes batch member create to retry the database get call\nwhen new members are being created via batch member update.\nThis issue only impacts the v1 amphora driver as the v2 driver\ndoes not need to get these objects from the database.\n\nStory: 2007581\nTask: 39503\n\n[1] https://github.com/openstack/octavia/commit/48e85569f7e43e3dd5f09fd1ef4cb165526a92cd\n\nChange-Id: Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cbeef4be4d447881ae20aeb7fea689c2a89397a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cbeef4be4d447881ae20aeb7fea689c2a89397a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
