)]}'
{"id":"openstack%2Foctavia~737158","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed","subject":"Allow DB retries on controller_worker creates","status":"ABANDONED","created":"2020-06-21 01:48:24.000000000","updated":"2020-07-24 11:41:22.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88dc23521271f3e149e71ec3e4ecdef5ff544185","_number":737158,"virtual_id_number":737158,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-21 03:50:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-1,"date":"2020-07-07 17:11:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-06-21 03:05:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-21 03:05:01.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-21 03:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-07 17:11:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"98e8fc04afb67ffd21d3f0032247309c6febe6f8","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-21 01:48:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"c251d568409b02b68a2c1b272250547b521df3e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-21 02:16:16.000000000","message":"Patch Set 1: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c2368dd5114a06b2ae61664545492f : FAILURE in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a2bef67d1704c60bc6aea5b7618f970 : FAILURE in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/43bd0863089c4108a38719e5151a3e4c : FAILURE in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c15cf4b79f34b4c87ad2e1a31510664 : FAILURE in 5m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/91716298b2a74830a63f63d98e7df067 : FAILURE in 16m 15s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3663c00d35a245e892e9e1669c44a11b : FAILURE in 20m 35s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1603c9fac1a642bfae075b2aa8d2a7fa : FAILURE in 15m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8763f428205c4a0dbe07ae95a3b6304f : FAILURE in 16m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/306388a1f0c742a892c69adfc5221885 : FAILURE in 4m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/daf3ee5b79ae4a0e841eca7252b5f61e : SUCCESS in 7m 07s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/eb8ea7add6394af7add6c4f97d66a748 : SUCCESS in 6m 41s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84cd970da7024e4fbfd93d08a11b8050 : FAILURE in 16m 22s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2c412d6c3f0f487cb61cbbc0c5d2fb8d : FAILURE in 16m 38s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/619efcfdc87e4553b046ab8b50dada8e : FAILURE in 13m 49s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/1947d756b23d47098ac57e33d5f93504 : FAILURE in 19m 48s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/50d458f86c854367af176a75d5680d8a : FAILURE in 14m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ffe3d0174454bb6a343b0a8458896e1 : FAILURE in 13m 24s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/33c9a8e9aa0e4dab91cd0f1f274ab5e8 : FAILURE in 9m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9981377b203e472da70bfcaad9c592dc : FAILURE in 13m 45s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/6a97cd2510204f7e8fdb08e91895fb20 : FAILURE in 14m 28s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/modules/octavia.cmd.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/modules/octavia.cmd.rst","accounts_in_message":[],"_revision_number":1},{"id":"c2d638a196486c2c5b9bf1f5edeef8ed089ae216","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-06-21 03:05:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"980cfd4d9cb5512f911932cb35352c1e6376e77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-21 03:50:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ccd009b6a464891b78302c4c4c1c9cb : FAILURE in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b013edad623e4dcab0dec3d66d49f120 : SUCCESS in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de2c08818a8d4b26b23ff504db234a76 : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/05e52382f5664f4689cdd2a6e80252ef : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136380daf1cf4c2e87fa694702353a0c : SUCCESS in 6m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6a8083f4754d4f4cab52e5813627e779 : FAILURE in 11m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/66bf783d20684c15868fe0b79855d45c : FAILURE in 13m 47s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b46bcdb22444bd19d1bfd4c5e92d981 : FAILURE in 12m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/06bf4b7f9cd8494aaaeff1411416329b : FAILURE in 17m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90882ac0278640bd943cb74c6716c4e8 : SUCCESS in 9m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04a12aa4685d4aeeb08040aa6217dd52 : SUCCESS in 6m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c0396d44369d49c5a98b13e3765ef967 : SUCCESS in 6m 56s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12a7582e954144848fe5ba61b47d8489 : FAILURE in 12m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/721f337a30fd4ad1b91eb89f8cf96252 : FAILURE in 13m 39s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e024b4c3ae60448d88bd85b15e9e0e16 : FAILURE in 14m 25s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/19a07dc2c150430eac3f3c3b751bbe49 : FAILURE in 14m 05s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ba1cab38806f456fa89a4f9ddefabf5b : FAILURE in 11m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a99fe05d4c0648fbbad144bb4df500a8 : FAILURE in 13m 16s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/b456c7ed2f4f449d9074fa5240f565df : FAILURE in 10m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce39003f04a1485f9fec3e01fd7b3ec6 : FAILURE in 9m 27s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/fb048177cab04e90a565472bce895bcf : FAILURE in 10m 40s","accounts_in_message":[],"_revision_number":2},{"id":"e22bbc7826ea49457b00bad68e23065e086c20fe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-07 17:11:59.000000000","message":"Patch Set 2: Code-Review-1\n\nI think this is an attempt at a cherrypick of 630a4e6a3cb60a8edeb4d8ad4a9cd7b31b42469a, but the cherrypick isn\u0027t defined in the commit message.\nIt should be noted that queens did not include the provider drivers feature that introduced this bug. So, I don\u0027t think a cherrypick is necessary here. Please provide additional information if you feel otherwise.","accounts_in_message":[],"_revision_number":2},{"id":"88dc23521271f3e149e71ec3e4ecdef5ff544185","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:41:22.000000000","message":"Abandoned\n\nQueens Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c54cb8f201526d17742f9f49da2f725e37f70e25","revisions":{"27d83ca016340f329ac1303bfbde71c82043e37b":{"kind":"REWORK","_number":1,"created":"2020-06-21 01:48:24.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/58/737158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/737158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/737158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/737158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/737158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/737158/1"}}},"commit":{"parents":[{"commit":"299c54a5dee5898eb0712123b98670f4043bdec6","subject":"Merge \"Fix the amphora noop driver\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/299c54a5dee5898eb0712123b98670f4043bdec6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-04 17:45:35.000000000","tz":-420},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-06-21 01:48:24.000000000","tz":0},"subject":"Fix batch member create for v1 amphora driver","message":"Fix batch member create for v1 amphora driver\n\nA previous patch[1] missed batch_member_update when adding database\nrepository \"get\" method retries for new object creation actions.\nThis patch fixes batch member create to retry the database get call\nwhen new members are being created via batch member update.\nThis issue only impacts the v1 amphora driver as the v2 driver\ndoes not need to get these objects from the database.\n\nStory: 2007581\nTask: 39503\n\n[1] https://github.com/openstack/octavia/commit/48e85569f7e43e3dd5f09fd1ef4cb165526a92cd\n\nChange-Id: Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27d83ca016340f329ac1303bfbde71c82043e37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27d83ca016340f329ac1303bfbde71c82043e37b"}]},"branch":"refs/heads/stable/queens"},"c54cb8f201526d17742f9f49da2f725e37f70e25":{"kind":"REWORK","_number":2,"created":"2020-06-21 03:05:01.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/58/737158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/737158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/737158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/737158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/737158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/737158/2"}}},"commit":{"parents":[{"commit":"299c54a5dee5898eb0712123b98670f4043bdec6","subject":"Merge \"Fix the amphora noop driver\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/299c54a5dee5898eb0712123b98670f4043bdec6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-30 06:58:36.000000000","tz":-420},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-06-21 02:51:50.000000000","tz":330},"subject":"Allow DB retries on controller_worker creates","message":"Allow DB retries on controller_worker creates\n\nTemporary workaround for commit ordering issues with the amphora driver,\nuntil the driver can be rewritten against the real driver interface.\n\nStory: 2002127\nTask: 19809\nChange-Id: Idfaca392b278a6efad36e51adaedc6c80372a006\n\nFix batch member create for v1 amphora driver\n\nA previous patch[1] missed batch_member_update when adding database\nrepository \"get\" method retries for new object creation actions.\nThis patch fixes batch member create to retry the database get call\nwhen new members are being created via batch member update.\nThis issue only impacts the v1 amphora driver as the v2 driver\ndoes not need to get these objects from the database.\n\nStory: 2007581\nTask: 39503\n\n[1] https://github.com/openstack/octavia/commit/48e85569f7e43e3dd5f09fd1ef4cb165526a92cd\n\nChange-Id: Ia3476ab7b24dc3fd6e29ff2abe6eb6bacd9908ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c54cb8f201526d17742f9f49da2f725e37f70e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c54cb8f201526d17742f9f49da2f725e37f70e25"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
