)]}'
{"id":"openstack%2Foctavia~737528","triplet_id":"openstack%2Foctavia~master~Id4141e820a8b34e379a8f60b53fc183680c55c79","project":"openstack/octavia","branch":"master","topic":"flavor-amp-image-tag","hashtags":[],"change_id":"Id4141e820a8b34e379a8f60b53fc183680c55c79","subject":"Add amphora image tag capability to Octavia flavors","status":"MERGED","created":"2020-06-23 12:57:47.000000000","updated":"2020-09-04 10:36:12.000000000","submitted":"2020-09-04 10:34:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737528-1599215655458-1d2acb16","meta_rev_id":"015da2634964eac4557e116a450493dc0ebc0b93","_number":737528,"virtual_id_number":737528,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 10:34:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-04 08:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-09-03 20:11:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-04 08:36:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 07:08:00.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-07-23 21:07:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-06 12:39:42.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-03 20:11:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-04 08:36:44.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-04 10:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5640d42b9e68dd799d33b138789a09c6f54210e1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-23 12:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b4e241403a08ae75b9028aed807cc68a3797f43","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-23 12:58:05.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"bbf04ad62a2e70ce1c6076f8348a7b865db362c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 15:05:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dbe0c481cb749dc86c9d7cf4b5c39c3 : SUCCESS in 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bbaa9e0549e4e7ca879d4e0b64afbff : POST_FAILURE in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da8419a80ef4a4fa939e2b0d03a1c93 : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b91429ec53394ee0913b711e0eafc76c : POST_FAILURE in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5c012943b1434ba98777b7f9303779 : POST_FAILURE in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77a22dee90b45f5bc0d620086c6fc35 : SUCCESS in 18m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f6d62ac6a064ac9a7a95c8d330d336c : SUCCESS in 6m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4845b1f9ff534c0e8d04f632f5fcc875 : SUCCESS in 8m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/824e24aeee19404491c15a9afd37ae70 : SUCCESS in 11m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b88a6f2238243b9908a8a1f53c190bf : POST_FAILURE in 10m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c295436e999421a881ffcf151a9be28 : SUCCESS in 48m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f66d98a5f4b04c7aa1159c567849b2cf : POST_FAILURE in 1h 56m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/328d9feddd1143e39f0bd54ed3e7c17c : SUCCESS in 59m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/85b4ad3a2f994ccca17e3b3c8b2d9374 : POST_FAILURE in 53m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d722e2db7ce946c5b3b21c671f8245b4 : POST_FAILURE in 1h 01m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1cfa27e076ab43efa42e83f358523401 : POST_FAILURE in 1h 36m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/36a000d0828b409c8647277947b01f4a : SUCCESS in 1h 34m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56543c0e35e24dde98ed46254613a872 : SUCCESS in 1h 48m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a4368ac08c324d45b9d68654a6cf7e78 : SUCCESS in 1h 53m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4d6d72e68ac69253c38c0ca0f03379f9d757b4e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-23 16:31:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9dd8e36170d00c87b9981a6ad6ce6b3ba153bfdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:50:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52e62da938d74eec93cd4e8f14c88a94 : SUCCESS in 11m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46b1150e8de1449abfbe67d5ff4e0869 : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db6868d88804e58821e42b050cf6e9a : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6d9ed9c6b174d389fd0a3a5414d4204 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa5d67594c064d3c8021ce4696421809 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d320b6cdd01a42ceaa9afde062558fbd : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bfa0f023a734043ba0c34ad8728251d : SUCCESS in 5m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/64c6152c17c24ea3ba5bc08b1e316c74 : SUCCESS in 7m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/83e0446450c74fc5bd8eaa93a4937e9c : SUCCESS in 14m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5a2ee70d4f54a12921ceb497570edf4 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9989810d548243e186e5b026d930882d : SUCCESS in 1h 05m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53ce2185cc4f4aaabb2674ef8312489e : SUCCESS in 1h 50m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6777e2055f3e4b64af35a50b1ebbf114 : SUCCESS in 1h 06m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7d67f3bb3cf749baa2738f95e11553d4 : SUCCESS in 54m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f683d1f43f84e838e35ccb75476c3b7 : SUCCESS in 1h 02m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9dd31f87ac544a10a4981e5084dfb1ea : SUCCESS in 1h 32m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a62efc8da2640a2ba8004e598417d06 : SUCCESS in 46m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37e72c55f4af41b9be6756158c47f5db : SUCCESS in 53m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7278edb4f0c446e5accbf0a91f6154fe : SUCCESS in 2h 02m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a8f2966cbf4b02ec1dc7453ccaf1b586e7610a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 18:33:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d062db186523fa2a18d1fcf298af9ad6dcb2fe85","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 19:33:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a9d35bc54448d54fc41b1d3e46dc81842dc94e9f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 19:57:35.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"294d3e62c609ec71f5a5ca267e0a3132f60f66d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 22:03:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ada38ebe6f3743828b41c61ba9849d4b : SUCCESS in 4m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49901c888ee1455b9a253c60c946c822 : FAILURE in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9377f03785ee46238bb69a0995fc736b : FAILURE in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d3dfef9f12423f93e4ba96ef6d8f52 : FAILURE in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e372891320b544cfb433eaf75ff2ec23 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9678aa70c6214a9c803d615d1eac98ac : FAILURE in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c2b26f7cac42da8216286756cf4d74 : FAILURE in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c49c5d2417a843079ff5e428e1c58b4c : SUCCESS in 6m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/001d6c507f0149ca9fd31ea6bde226a6 : FAILURE in 7m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c286cd27da514c72802baa2667446bcb : FAILURE in 10m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8dca69c66fd04788ae68e42ed091686e : FAILURE in 5m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f1922774ccb94e1bb2374a12c56f6ace : SUCCESS in 47m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f782a4c68b6841af852078aba88d9e74 : SUCCESS in 1h 57m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0caee9fbcb6e424aa894734c5a4f538b : SUCCESS in 56m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d061af8d2cce4a6bb32a7d561ea61ec6 : SUCCESS in 51m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48f24530c416465190d2b41d4beddaec : SUCCESS in 1h 05m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7850e18ac3cc401d8d0620ac3263ea5a : SUCCESS in 1h 41m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b4d3e78f634e491ca2281876ff52d50f : SUCCESS in 1h 52m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/598ce1ceb3cd4ec782b37ebee71cf926 : SUCCESS in 1h 05m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce18d5685a2b4df4b527cc32ea020b7a : SUCCESS in 1h 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5ba11404b48b322e9e43e7a4c23bac87d86842c6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 09:23:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c5fd039e41141e94aa8b9c452c90686d69ccf763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 11:42:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/67f2247436ce4a0296f3d3425ab531cf : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2b635f0b27a40fea34c99cabad936de : FAILURE in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91e0300db61f47e8813a7921b0d2526e : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719998f02ae24c79ae0e2736f334542b : FAILURE in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ade44a3c1649aeabff4c5496940d70 : FAILURE in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5e23988e66e4f689dc5eb502ade2cd2 : FAILURE in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e4f66419d3d480291c7f9cb8f910cba : FAILURE in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7dde1ee22254d3c8d6acaa7b1374035 : SUCCESS in 6m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6e38a897f4514ed5b10a6284121ffc42 : FAILURE in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ff2ca0644cc2487f87e1cb3cbde28236 : FAILURE in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/207d065ca42e4e72a50c0d5240717b81 : FAILURE in 4m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0ae197d61af419fb66ec2b97983f4bf : SUCCESS in 43m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de3e6ccc468144dba33ac33dd37d5105 : SUCCESS in 2h 10m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb56a459a0654e2f98fe0e4a465a0e1d : SUCCESS in 58m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cfbf11c232a549bfb57014d85d7940f6 : SUCCESS in 57m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68f574cbb3d940bbbbcc3f2d60a15a94 : SUCCESS in 1h 01m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/780cadb2782e43afafa04b0aa6ce616a : SUCCESS in 1h 55m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eed99c1f27234e6a990b82c360a813ff : SUCCESS in 2h 03m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/745ce31b71384073bdce5a28cb64746c : SUCCESS in 2h 12m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fae09bde8869433d89e63ed59ec240a0 : SUCCESS in 1h 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ef2f3dfecdcb1cba8b3a4a395e7e84fb91582d28","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-20 14:04:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9a708ca5536b0b31cf7a87e2cfa7af65c42de2b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 15:38:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/19784ebc59d24d5684087ac270cf1f16 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12286a4f579c443886fcf8bd5f547bc1 : SUCCESS in 16m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccccc0e650ce4498a8c2973f6ddb26a4 : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d017c290bff449ba33ea79cb1839045 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c545572182724f46bc020d5156317fae : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/025690d77de644f3a252537c3d5f909c : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb369b7e119414299672aff2b73c9fb : FAILURE in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab6a8888e1974d4d965a2a9d792e4b4f : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5fae905ea9cb4dacbd264e8d6b7d4f49 : SUCCESS in 12m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/138a1d7778b7428184d279dd100a4a18 : SUCCESS in 8m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa5072d3bdde48d2ada0ccb09b5eea14 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f87bd1089d64b64a2e747f6a4aa7d9b : SUCCESS in 45m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42e97df8e21d4e4dbdec9197d52a5069 : FAILURE in 15m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a884228d9bc4c67afcf2b6e0912a0b4 : SUCCESS in 43m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/20f9db4bd9aa44c78363b55653701c5e : SUCCESS in 33m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dee35c42c1554b0291f37cd14123505a : SUCCESS in 50m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3bc8279901946ce9ac3a91c4d79fcf3 : SUCCESS in 1h 31m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1378efa0b970423cbdef922c4f3f97a1 : FAILURE in 38m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f910364b4c3044b4bdac2376056cbfe2 : SUCCESS in 1h 17m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5fd1a3e66fa648b9ac92fe1f8f00aa16 : SUCCESS in 54m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb949938b3e74b41b9cfe3b75f2c9d0b : SUCCESS in 52m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a0ab45781400464d9cd17c2e24688b26 : SUCCESS in 1h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3f3c57e89bc23790d7f6f098cd49ba91c9a393d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-23 12:38:24.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c7f00b7df3a42f6a4ffbb54677d1e3825a496275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 14:14:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b206edae87014428a14519dc6ef300b8 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afcc15a9a944477bbd7fc0a3a7aefab4 : POST_FAILURE in 29m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fc7efda601e479081a1ce18876799e9 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc99798f7fc488b9ec5caaf2ac5ce7d : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92663cb53d594732843f959bbec3ef1f : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ee1a5d9e49e495387d57eb8636581a8 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecbc524117404cc080a35382eb7491df : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d3dbbf834cd4308bc31eed4e60ea615 : SUCCESS in 6m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a7272e43384a4a99864dd74a1478e091 : SUCCESS in 9m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/361aa43f2b114eb580ad93f5d3e35ff1 : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a9dab87e13ba48afbc5758c0edd20d78 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b434d69d4463455c9d6055c9666d11e0 : SUCCESS in 42m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb3b2ba5191e4b2aa2b10e95fac89c67 : SUCCESS in 1h 08m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62863a0a0f164ef58309ac5a4495cac6 : SUCCESS in 1h 01m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8dd686366d8e4bd3880b42529a22e0b2 : SUCCESS in 1h 01m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b68758dc6934675a5a88c02d4ff676b : SUCCESS in 40m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c0e22da427d47fb9d91844b2d1c6a57 : SUCCESS in 1h 30m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ee6e0e1483d64ca6887c8a3d977281f2 : FAILURE in 1h 18m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00dfbadc53b74e73a7d42182f29f2fba : SUCCESS in 1h 27m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b0fcf5144e7649f7b17899e0a4bb4ad0 : SUCCESS in 1h 11m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0766e5e593fb4876bf51f22eaf517d4b : SUCCESS in 42m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2cd033f645914ca8bc2f1d037df6eee7 : FAILURE in 40m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"741d2f78cc88b20348eb48bf7697653b7ddd31c8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-23 14:17:08.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"4775ddd3d163f3ff47c9b02bcf98d141b9a6be98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:51:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b6ff543aa17845c693e3cd68347eac2d : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/099200e6a63e4b108aaa647bdd70268d : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e86191a2f28c40b992b98b2b1c65c532 : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4904aaea7c4fc099ad04c919e72685 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aae710ee545c4c29bfb8a2bb42c34376 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fda71739c544f25ac9bb3ed08094813 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631f1efe1f9b403fb868d80244561605 : SUCCESS in 15m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/150d033a7dfb44a99fd698798bfed113 : SUCCESS in 7m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f4c8db85d6a5416399e384fb25c3a60f : SUCCESS in 7m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0fabc3f1e62645c7b2e77bfc8a7995c6 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aad483e4531b41b3be8346115c998d79 : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2643310dd75c478e99e2e166896e7122 : SUCCESS in 45m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bca1f70917634b25a252edb57f80a8a8 : SUCCESS in 56m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/943ce10fe265461cae38c4a4a4b90bcd : SUCCESS in 41m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/42aa6ebdf4cb4ab693f22bf35e202850 : SUCCESS in 33m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9c78f2d0c4d4e33adb5e36c9d1b1f8f : SUCCESS in 40m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e37cfd36794b419289f12507f9ec6452 : SUCCESS in 1h 39m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/31ab58c9843448298a3ad43adecc2d4d : SUCCESS in 1h 00m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8bf93ff8e194bba840d4e0d2ee9b351 : SUCCESS in 1h 02m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/85b1b1ea33744f08973c4fb21b554f16 : SUCCESS in 1h 12m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3956db6d39df48099f892a6fec4fce76 : SUCCESS in 54m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2a4a8d45fb5245c190afd7cf29a351dd : SUCCESS in 1h 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"af0b2bccb32c9da6d8ef13c4d3b2271c8376deb3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-23 21:07:01.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4403badcd8609dc6a0ae8352cc0c029a34da8fa7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-06 12:39:42.000000000","message":"Patch Set 7: Code-Review+2\n\n+2 but I think we need test coverage for flavor profiles features in the amphora driver.","accounts_in_message":[],"_revision_number":7},{"id":"3df1d047d10129123ede23b546795ee11fadc194","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 19:33:56.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"65f126b239b5b6a471d86016a06f10029324c3c0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-03 17:04:49.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"3dc2e9652eed008d99c394d11b4a23ff7158ab21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:05:05.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"085b3e1a683225a943aaa28d4f36718bc2bcd52f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-03 20:10:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3776634434bf5e887ee4788f71309939961d34ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-03 20:11:06.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"e6138ea1dfa0ebdd633e5b66f42e842096771edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:56:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49885702be514d6d83fe39a8846db58c : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14458b0eb47f4056bac20fe5075671c7 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573709c5798b420285cb37a3e8fe03aa : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968d62ba5b414fb680980a1e2fac9ca4 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/584b0f20943e4baa9acf05e805ec2bf8 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f50e1e45b84b0f9f891116cd3ae4d4 : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d9edcf123ff4bf1a223fdbc9afb68e3 : SUCCESS in 6m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8f36d621b10f4603a3da3ce5771e2d61 : SUCCESS in 9m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/768ce201f748483abaac3657f3eb598d : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/062e5b3a36554a8189819bbccda94876 : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e115a84aee654150a39cc7245ae001e1 : SUCCESS in 47m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca851987e579416787f76716cb2cf638 : SUCCESS in 1h 07m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5fff0ad4ffdf49b6b9204788619e2cb3 : SUCCESS in 45m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a26e2ca203bf4420a810e36670df83b1 : SUCCESS in 57m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/169856e46543448db7e84821cd73558d : SUCCESS in 43m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c4ac09525bbd452b8948b17b76fd0948 : SUCCESS in 1h 32m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73eb883569c8438e84182ce20b75ad41 : SUCCESS in 1h 01m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eab9b74375df4cb69db370ce16c883d4 : SUCCESS in 55m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0939eeb6bed6447aa8a53122484685ae : SUCCESS in 1h 00m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/affafe384c6d4af68bc5d76a4a3e0ecf : SUCCESS in 1h 25m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b5afc95a231a448c878d16d559b7fd44 : SUCCESS in 1h 32m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2854f65381b84dd3b23728cfae61a5a4 : SUCCESS in 1h 09m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ef26bc90706d496d26aae47478ae8252bd5bf742","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-04 08:36:44.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"85d2ddb1ce2e58f2997cc5fb1ea45fe1b227eb6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 08:36:56.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e399eea89cf65ae8d78b319d08220ee5c6e0866d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:34:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"a304620bd9cd9eedffcd19983201ca3f7f1887d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:34:15.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04eb124234d5432d8e1e3a029359e3ab : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43cc70e99951431f93ddf465429572fc : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89ee502af4bc44a9a0b1e256e911d5f6 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d481f42fcf949bfb09c549d06502994 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e465e4dfc1e146299cdc81a4314ce625 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2177a4b44664d4ba6f6e37f06066bbf : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9609596d0a5c492d9a5354b6234b340c : SUCCESS in 8m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb094a441af44d3aa542ca88d9ea0b92 : SUCCESS in 43m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afbb545157c3427aad7ac4114df667d0 : SUCCESS in 50m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e5e9ba48587c44e99c3559f8915c94d6 : SUCCESS in 1h 09m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/95490b5dd8f94cd7a5f519b2e9bbca4a : SUCCESS in 44m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/536aa224518f4760a41986e55adba65c : SUCCESS in 1h 01m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1a3c1b42310c4b9ab9b1ce0d3a10d0a9 : SUCCESS in 1h 52m 16s","accounts_in_message":[],"_revision_number":9},{"id":"bc5422c0f4aadfc19ddd2c294991010331b27085","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:36:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21eee921d24a40b6b56ae378482b3e74 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a22becae34b4eaf8ecf424c0414398e : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"ea0bbd2775778c027fa95ee2a26e46d4a565eb43","revisions":{"63b18ed9a628924b76217f0b2ff5373ddb6f71b0":{"kind":"REWORK","_number":1,"created":"2020-06-23 12:57:47.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/1"}}},"commit":{"parents":[{"commit":"e475734b2ac98d9a37818e0ff9ca2baa5ba42715","subject":"Merge \"Clarify health monitor behavior\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e475734b2ac98d9a37818e0ff9ca2baa5ba42715"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 12:39:55.000000000","tz":120},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63b18ed9a628924b76217f0b2ff5373ddb6f71b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63b18ed9a628924b76217f0b2ff5373ddb6f71b0"}]},"branch":"refs/heads/master"},"6f8db0f856d3888bc4be3bf7cf1453317f062fc1":{"kind":"REWORK","_number":2,"created":"2020-06-25 18:33:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/2"}}},"commit":{"parents":[{"commit":"1a84a2a7c319453d536a9c4319acc39c67f87712","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a84a2a7c319453d536a9c4319acc39c67f87712"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 18:32:56.000000000","tz":120},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nDepends-On: https://review.opendev.org/#/c/738075/\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f8db0f856d3888bc4be3bf7cf1453317f062fc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f8db0f856d3888bc4be3bf7cf1453317f062fc1"}]},"branch":"refs/heads/master"},"1dbe70765f2a4d6dc051afe7729cdb4135e7eb0e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-25 19:33:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/3"}}},"commit":{"parents":[{"commit":"238b2766d64239ea7238f96669649380beed8de0","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/238b2766d64239ea7238f96669649380beed8de0"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:33:24.000000000","tz":120},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nDepends-On: https://review.opendev.org/#/c/738075/\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dbe70765f2a4d6dc051afe7729cdb4135e7eb0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dbe70765f2a4d6dc051afe7729cdb4135e7eb0e"}]},"branch":"refs/heads/master"},"04d2d0ce6bf6d467b3260787495925dc3f0a3390":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-25 19:57:35.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/4"}}},"commit":{"parents":[{"commit":"a5f4f376fd2a6a6c7317e77d794a7b25e236bea6","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f4f376fd2a6a6c7317e77d794a7b25e236bea6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:57:22.000000000","tz":120},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nDepends-On: https://review.opendev.org/#/c/738075/\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04d2d0ce6bf6d467b3260787495925dc3f0a3390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04d2d0ce6bf6d467b3260787495925dc3f0a3390"}]},"branch":"refs/heads/master"},"cdacd5b21b6d167e33c1cf45aaf8a90fc21c2f05":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-26 09:23:17.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/5"}}},"commit":{"parents":[{"commit":"4fb7f3e4ea30209f9494ab3bb2fe96326995c9c1","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4fb7f3e4ea30209f9494ab3bb2fe96326995c9c1"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 09:23:02.000000000","tz":120},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nDepends-On: https://review.opendev.org/#/c/738075/\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdacd5b21b6d167e33c1cf45aaf8a90fc21c2f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdacd5b21b6d167e33c1cf45aaf8a90fc21c2f05"}]},"branch":"refs/heads/master"},"7d4b2c74e9de6cc084d9724a79158545645b598b":{"kind":"REWORK","_number":6,"created":"2020-07-20 14:04:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/6"}}},"commit":{"parents":[{"commit":"4fb7f3e4ea30209f9494ab3bb2fe96326995c9c1","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4fb7f3e4ea30209f9494ab3bb2fe96326995c9c1"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-20 13:48:04.000000000","tz":120},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d4b2c74e9de6cc084d9724a79158545645b598b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d4b2c74e9de6cc084d9724a79158545645b598b"}]},"branch":"refs/heads/master"},"a1fe057f51ba6fe3bdc07ffd16ab0d1cae103c94":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-07-23 12:38:24.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/7"}}},"commit":{"parents":[{"commit":"06047f3f47579b3ad0f282ca37d51de0e052a9ef","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06047f3f47579b3ad0f282ca37d51de0e052a9ef"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-23 12:14:15.000000000","tz":120},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1fe057f51ba6fe3bdc07ffd16ab0d1cae103c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1fe057f51ba6fe3bdc07ffd16ab0d1cae103c94"}]},"branch":"refs/heads/master"},"97e3dfc7f447ed610e90e421c2c063697bd7fb3b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-09-03 17:04:49.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/28/737528/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/8"}}},"commit":{"parents":[{"commit":"f9205e569a2af89b23065da9aa5e6a0f413f56fc","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9205e569a2af89b23065da9aa5e6a0f413f56fc"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-03 17:04:49.000000000","tz":0},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97e3dfc7f447ed610e90e421c2c063697bd7fb3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97e3dfc7f447ed610e90e421c2c063697bd7fb3b"}]},"branch":"refs/heads/master"},"ea0bbd2775778c027fa95ee2a26e46d4a565eb43":{"kind":"REWORK","_number":9,"created":"2020-09-03 20:10:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/28/737528/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/737528/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/737528/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/737528/9"}}},"commit":{"parents":[{"commit":"a422e5a203911e7d39fb7cc7bb66449131a04cbf","subject":"Introduce an image driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a422e5a203911e7d39fb7cc7bb66449131a04cbf"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-23 11:58:00.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-03 20:10:07.000000000","tz":-420},"subject":"Add amphora image tag capability to Octavia flavors","message":"Add amphora image tag capability to Octavia flavors\n\nThis patch adds \u0027amp_image_tag\u0027 to the supported capabilities of the\namphora provider driver. One use case where operators could find this\nrelevant is on multi CPU architecture clouds.\n\nTask: 40156\nStory: 2007854\n\nChange-Id: Id4141e820a8b34e379a8f60b53fc183680c55c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea0bbd2775778c027fa95ee2a26e46d4a565eb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea0bbd2775778c027fa95ee2a26e46d4a565eb43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
