)]}'
{"id":"openstack%2Foctavia~737657","triplet_id":"openstack%2Foctavia~master~Ic7b7c3957a80796f99299473486cded19f15ceb9","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ic7b7c3957a80796f99299473486cded19f15ceb9","subject":"Fix devstack cleanup keypair delete","status":"ABANDONED","created":"2020-06-23 23:13:23.000000000","updated":"2020-06-26 19:09:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8f5bf28a3d0f8b8990a6a779f854a2e85618b28","_number":737657,"virtual_id_number":737657,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"date":"2020-06-24 22:28:53.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-24 01:24:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-1,"date":"2020-06-24 06:45:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-06-25 08:01:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 01:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-24 06:45:12.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-25 08:01:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"18e0047ed8102d424fa220c94b4d0ebf8995816f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-23 23:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c10f3ab958ce34d49e3bd95990f2e9ea89d76c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 01:24:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb75d3cdf10e4ab78e649d1f2de26610 : SUCCESS in 9m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2668ae93730646c798cd51fce5796177 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e37afece464c31bb09c346d2481258 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eb1afa630c04d94be3d73c70c7a01b5 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/992a870088a849de96ef2e78f842166c : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba86eecfe0e4152b3032969e4a208fd : SUCCESS in 13m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ec3ada5b9eb143ea8c15a3d6538a641d : SUCCESS in 8m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/81a39d4d91134c0584fab367c76e1162 : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd6c372e94e94839a14e8954401180d8 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/602569fe24d64a2cbaf107b0cbf31660 : SUCCESS in 48m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7215b8915574008bd3f129ab76749b8 : FAILURE in 2h 03m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/16b52005ef364945934fdaf47d2f8dc5 : SUCCESS in 37m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/15b0d4c07ae845f1aaaea2e3768cdcad : SUCCESS in 50m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3019922fcc64050a805dc74ebc62864 : SUCCESS in 57m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ae349251efb438fac8b54174ee7fcea : FAILURE in 54m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ff1a93d1390d4b00b9ebb3c488e7f8fc : SUCCESS in 1h 51m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e921b17881b4aefbd086b8a746efe03 : SUCCESS in 1h 57m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba294059b91f480baa9a5b8a7dcfd988 : SUCCESS in 1h 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7959165fcfa8ee1db57aa4ab52e45e8011d98eeb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-24 06:45:12.000000000","message":"Patch Set 1: Code-Review-1\n\nDuplicate to https://review.opendev.org/#/c/735510/ which takes a different approach.","accounts_in_message":[],"_revision_number":1},{"id":"bd76c15e1a7f1d09a850c87ab12ffe6f257cf750","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-24 22:28:53.000000000","message":"Patch Set 1:\n\nWell, it\u0027s good that we have the -e check.\nI don\u0027t think this will ever work in the \"clean\" phase of devstack. I think nova is already down by then. This is why I moved it up to the \"unstack\" phase.","accounts_in_message":[],"_revision_number":1},{"id":"3dc0389f18f79d7ebd6e715bd47adbf17526a604","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-25 08:01:41.000000000","message":"Patch Set 1:\n\n\u003e Well, it\u0027s good that we have the -e check.\n \u003e I don\u0027t think this will ever work in the \"clean\" phase of devstack.\n \u003e I think nova is already down by then. This is why I moved it up to\n \u003e the \"unstack\" phase.\n\nThe issue I have is that \"keypair delete\" is failing after a reboot (basically after a reboot, I run: ./unstack.sh; ./clean.sh), moving keypair deletion in the unstack phase doesn\u0027t fix it.","accounts_in_message":[],"_revision_number":1},{"id":"a8f5bf28a3d0f8b8990a6a779f854a2e85618b28","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 19:09:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f94dea29d53ccf06a6eb24bfe38171c6e8b4e171","revisions":{"f94dea29d53ccf06a6eb24bfe38171c6e8b4e171":{"kind":"REWORK","_number":1,"created":"2020-06-23 23:13:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/737657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/737657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/737657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/737657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/737657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/737657/1"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-23 23:11:25.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-23 23:13:22.000000000","tz":-420},"subject":"Fix devstack cleanup keypair delete","message":"Fix devstack cleanup keypair delete\n\nThe devstack plugin had an issue where it was attempting to delete\nthe nova SSH keypair after nova was already not available.\nThis patch moves the keypair deletion from the cleanup phase to the\nstop (unstack) phase to make sure nova is still available to complete\nthe action.\n\nChange-Id: Ic7b7c3957a80796f99299473486cded19f15ceb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f94dea29d53ccf06a6eb24bfe38171c6e8b4e171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f94dea29d53ccf06a6eb24bfe38171c6e8b4e171"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
