)]}'
{"id":"openstack%2Foctavia~737993","triplet_id":"openstack%2Foctavia~master~I03385d93575db4e44a72335c0af4fde490a93b0f","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I03385d93575db4e44a72335c0af4fde490a93b0f","subject":"Add experimental amphorav2 jobs","status":"MERGED","created":"2020-06-25 12:26:15.000000000","updated":"2020-11-19 11:31:35.000000000","submitted":"2020-11-19 11:25:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737993-1605785103299-577aa814","meta_rev_id":"0aa242509674c76708725ad6314ab6d87318ed49","_number":737993,"virtual_id_number":737993,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-17 07:28:46.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 11:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-11-19 08:54:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-11-17 14:36:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-03 09:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-11-19 08:54:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 17:34:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-17 14:36:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-19 08:54:33.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-19 11:25:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71446d2d4d167df508e7baace069496afe10da2b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-25 12:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71bc85f6c6d6c1d41c247c9ee97c46e6f583b78b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-25 12:29:58.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"122b529faa2a06ab22171af03e0031db3d3b183c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:24:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b69961143d1e482490617dac203b1bad : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0355b0d8c9342b780ca3bd07cb33f1c : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e69f08d9220413a933930fc5be96da8 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1378b1db8b5c47a696d42f28be3c2251 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac2054c8e7ed456a9820d93f8734c665 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cb6619b6ed48ba946f5b43eeb44195 : SUCCESS in 12m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/24ab07e8e5b8431391ebd126f8e9d400 : SUCCESS in 7m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5f74520779e44f62ae6cc51533dd7bdc : SUCCESS in 9m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3b7b3830e8f462fb595bcca6f33932f : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3bc2e99eec264e9f94f4260609f2e047 : SUCCESS in 49m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20ecaaa4a26a4081b8ff4a44d1dddbfe : TIMED_OUT in 2h 35m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b51ce7958404926877ab60560d18be9 : SUCCESS in 57m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/23f490c32dc74be6a6a94cfc11e7d2e9 : SUCCESS in 59m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2be210cc649745de92e252abc058f677 : SUCCESS in 37m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/77ff29836f644a2d9afdc926a3c9921b : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/01f2cd3ab13742d5b86d349ee7e4698f : SUCCESS in 1h 43m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5c96374501c442bb8e0e4927e56c5f8 : SUCCESS in 1h 05m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/84521c6c07d24de2bfe6869ae7b47a7d : SUCCESS in 41m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c8587f32469fe20479a94165b74fa626ff8c629","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-26 07:29:46.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"95cc8a3969d4208c16ccaed98c80507b9b7a5831","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 17:29:30.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/056a7517a5ee437eba2c640d3277d135 : FAILURE in 41m 46s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/2094af20d9ae4d19b1c3e876510d8ecb : FAILURE in 1h 49m 22s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/10e16f700f6a4611994dd0d7c165803a : SUCCESS in 57m 49s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5d828b41b428430988a1851d9c7782d7 : FAILURE in 48m 43s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e4df6d1a8d4469d9257aa9e63fed789 : TIMED_OUT in 3h 05m 49s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/07acf7aec3354f19bb57fae3cc296711 : SUCCESS in 1h 47m 29s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/5219293d9c0549f1a35e1c9982908103 : SUCCESS in 1h 59m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0f24637516e9a2975564cc426d20cc8706319d3c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-02 09:23:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c0b04da238f4c942447149953a4337a8bf2e1b2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 11:44:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fae5486961b4c20b8260422509d4a4e : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f91190e71e89493397163198a84d100a : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf179700dd344ce392c046687fc014f3 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b6977eeaef84860a61df8035555f972 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2289b932a2245ddb7a7224c1db68434 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c8eb56d50e49df99762bbce1800d78 : SUCCESS in 14m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd58d468c9d54edbbae7a4fcfb50fe6a : SUCCESS in 9m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/03e5d194f759482f903e68e73e893fd9 : SUCCESS in 10m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c41053f216442acaca1f520a5a299ae : SUCCESS in 9m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc3487eb2457408e8afbaa9192233884 : SUCCESS in 50m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1ba9e971427410d823ef974f91bbd77 : SUCCESS in 1h 00m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8499facb19b34d22b90c05ad4c0e19f7 : SUCCESS in 1h 03m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0893ef29bbb8425fad48fb49ba925b62 : SUCCESS in 1h 00m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8279c8fc97349bbb078c4762b92f2ea : SUCCESS in 1h 02m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbee18e621bb46199d9eb550ce8779ff : SUCCESS in 1h 41m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4cbb6e43193d45319fc08444e6af8723 : SUCCESS in 2h 10m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07e3d5c723194c60b214621031358644 : SUCCESS in 1h 40m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2ae6d0637e9440ff9aaa33023d48d64c : SUCCESS in 1h 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54849d762df5b5e8d314fc05cbee0086fb38b203","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 07:37:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89f915ff56c042aa02af8813639d024fc3f30f19","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 07:43:58.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"932d0b8b02fe886c90181a962f81d3b4ace8b4f4","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:35:52.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/b62391660f6142e2b452c98bb71d6f4f : FAILURE in 50m 24s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/bf2752709a574fffb42bc1b612747b61 : FAILURE in 1h 45m 17s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e6276a9282b419d8bb81996062c5d3e : SUCCESS in 58m 17s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1a0cffdc8c2c44d99f60d43016396808 : FAILURE in 44m 45s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/36a05af421484fc691575845434b742d : FAILURE in 51m 59s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/eeb58a19b3154afb9c2e0cf788651608 : SUCCESS in 1h 27m 42s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/678012e019fd433d86713bfe76f5bb31 : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a1e4fbb9cf82af8d1d76880eb2a610a47ee0fdb0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 09:44:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cf379faf72825d28eccc9727fea6a7f43ea8a483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:45:08.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/octavia/zuul.d/amphorav2-jobs.yaml@master\", line 108, column 11\ncould not find expected \u0027:\u0027\n  in \"openstack/octavia/zuul.d/amphorav2-jobs.yaml@master\", line 109, column 15","accounts_in_message":[],"_revision_number":4},{"id":"8dca091aa7446d4cc3793c7bd00ec3dbbe7298f8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 09:45:14.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"d1fb641d20f8026bf8aa7ad3a5d429c552e6f461","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 09:58:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4d03e01bc51d0e0adbfcd439ea929384361ffe6a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 09:59:29.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"541b54a800253758ff227beeb6f56956ff36c939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 12:06:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e2761e649d44459a68ccfc93309570d : SUCCESS in 10m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d34459ef23e4861bebc07bc92cec8fe : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53dac8db26af469a813fd0e6f4610cb4 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2721cf4404b452abf35a61331d9d9a9 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dcb7c2649c44e639a7742bef553df59 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ecdba08f764af18ee9f73cede9bf89 : SUCCESS in 14m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/617a56272f0140fb8dd2d562be27670e : SUCCESS in 9m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/211de6669c684d2e9468f588f316398c : SUCCESS in 10m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be2779d922394fbe90dbcca25f5dae4b : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ac75289b67b476f8e0dfd021d639095 : SUCCESS in 48m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed35e28967194adfb782010f557361b9 : SUCCESS in 1h 55m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4505fa37c3bf4319b5db932f17bed069 : SUCCESS in 1h 08m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d8941757166747e8ba17d35c6cce6ade : SUCCESS in 58m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a1a7289b2d94ad9a0483adc1e9aac41 : SUCCESS in 1h 06m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d207434f75db40cd919dd5261fa39d10 : SUCCESS in 1h 36m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8985557200ec4de394b944f4f6830601 : SUCCESS in 1h 48m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43671f1f969c4f6e92fe8b2572e5fb38 : POST_FAILURE in 1h 35m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0fe9e151d57d4bf498ce4893be95129d : SUCCESS in 1h 42m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ddda21a778d851325c4ebf3f40fa4a022b88a496","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 12:06:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9680b95c5f3c201ec86247a394b6166e018088cc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 12:30:25.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"a2c23a313e467f5b6d4e69be4cb61b2de7827b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 14:10:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc98d59156a54af2b64423e45b04d2b8 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c80a57db88ee4a66a0286aaee94cdbc0 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1213a34933a44a51aa34fda96654ab39 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d46d56b695d4bb9b72fe08cb76ccfe8 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcc2f7cc209c47f083423ade3df7f3c1 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1377c1d6865d4cc48e522db6dad32d01 : SUCCESS in 12m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/575df179a6934c50a9fa33636109dfbe : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e93459bddd1148159e62f9fb1241e348 : SUCCESS in 8m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a94ae9e1a1a4e45b8c9d985e979ba13 : SUCCESS in 13m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff457fde812144e792be3b8d69da3ac2 : SUCCESS in 44m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8b4632296d94de8be342df73efd5118 : SUCCESS in 1h 38m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2a0e7e0d87f415f97df8fcc63df9c31 : SUCCESS in 59m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b33a53212c7b48cda909e1ba1f2836ca : SUCCESS in 1h 01m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a3758e5e2c14e958e239f3124ccaad9 : SUCCESS in 1h 00m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c4ac7c005fd644b1aa26e99d0e97a4c4 : SUCCESS in 1h 49m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12b2b8cf55004a238ac91955fc132311 : SUCCESS in 1h 46m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d275d32f09d1481981f6fe690a61deee : SUCCESS in 1h 56m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa9ff0db7bf44c0a949fc9d948cafb09 : SUCCESS in 1h 36m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4857970eff77479883c43111f4b89692 : SUCCESS in 1h 49m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fced0e6f9a447c72363ed13f064c795a46dd9150","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 14:51:02.000000000","message":"Patch Set 6:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/9db4b657ecbf4baab32769738eec1d4c : FAILURE in 48m 37s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/fa90080e541a4e4cbc38442b2e79eb16 : FAILURE in 1h 50m 16s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/82fb55f4e0c74204b1e2a9889bf19321 : SUCCESS in 54m 27s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b4d64fe1ffa54fcfa6519106fec7d322 : FAILURE in 41m 26s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/b6ea4c64d50849afb6197ff7c862683a : SUCCESS in 1h 36m 50s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/a3d7a1685acf4ab0956b240525a5b757 : SUCCESS in 2h 07m 12s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/720d91af9a524c6c9e1422d4693a2729 : SUCCESS in 2h 00m 23s","accounts_in_message":[],"_revision_number":6},{"id":"0fc3d25caeccaabaab337130e96300e82a059c0f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-10 09:39:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d4a1cd7a8dfd968ef19539261589f5d284380ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:47:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa852bd339f94ffebc8d0585b4dd59e6 : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/693defbc922c4e0d94e6ea8219fa151c : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82f966a6470c4c19bd1fed9f07f3ed66 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19c811662b3042eda58b83b9ba94254b : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c77de4ec2a7a4611b1a7c30c61256765 : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a16f68ed8f04e95a0815896defdb002 : SUCCESS in 12m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1371e71b2b9e4b2c9b6292d28df15b59 : SUCCESS in 6m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5b8d35d11f4d47c9a96a9a3ed76ae37b : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d79c02c2305f451eb6f2f6421d457f2b : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b846522c23444085aa573e96a6d700e9 : SUCCESS in 48m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b57333e84394f1d8ceb96ee833893fc : FAILURE in 40m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/020e6c6681ad4f0a93776058c2d311e2 : SUCCESS in 1h 03m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f84b645353b748ca8da80fca78eb933e : SUCCESS in 55m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/081720fa63da4d138070efe79ad4acc4 : SUCCESS in 1h 03m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/79987dc9ba4c48c1bceb0d587d17487d : SUCCESS in 1h 59m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/76c06b4b536c4c3b96667e98e76c8453 : SUCCESS in 1h 55m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cd96b69ac31417abbb548aa9d8fb226 : SUCCESS in 1h 02m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5295e3bf952d4b2c8004aa6c9821123e : SUCCESS in 1h 06m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9ca2e8a566a4bc8959e81534e957e37 : SUCCESS in 1h 59m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fdbedf1eebf970c8b5f810f5041f13cc8ce5d485","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-29 08:45:56.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"1651ab79ff9b3691a68a26672c067800f4e34fdb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-29 08:46:13.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"3ca4fd3aae09df49ed019dc9f6c2f7cdb0f06a23","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:35:34.000000000","message":"Patch Set 8:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/a4c6b4703825487093fe91bf5f681680 : FAILURE in 1h 13m 32s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/a5dcabcc26424049b2d3598fcc1cb8d1 : FAILURE in 59m 24s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba3cf861d4e042aaa5d15b8892eee219 : FAILURE in 57m 12s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bcd4f801757b4c1b8faf39b0384b70ed : FAILURE in 1h 18m 57s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/2789cafa18ec438b9d66a94330d00044 : SUCCESS in 1h 37m 47s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/794b57380202420a9ef2889bb1514370 : TIMED_OUT in 2h 34m 17s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/3b8ecbe845304175a659addec99adf47 : TIMED_OUT in 2h 32m 36s","accounts_in_message":[],"_revision_number":8},{"id":"c2d4ddd529c73cd516a2f1f9ce278db697e63d5d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-29 12:17:32.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"954d7d2b96a058197868e395653a0f21116b1cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 12:31:16.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e6e18cc5fda4633bb1c712982d09ed1 : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c89b253ff244d0084dd881fe0fd3982 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ef24ee4b1b4da8b93521d31575e7c0 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f24eca0e0d2849e7b477c06536d44e46 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6282eb71199458f8c4f81fc9dded232 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4e667ad6224ddcbfbf7ad925e27e51 : SUCCESS in 11m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/59b66111cf31402a893d41fc20f29685 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/57b44ac97046429eaf929d24829063a5 : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fbc153cdc30b4c1bb62b118679083cf1 : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b78cf06e38f46b9bf4d306ad9db920c : SUCCESS in 40m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e936f75fcd234f1da3410e76b8b3b588 : TIMED_OUT in 2h 32m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19ffaeaeb0c84dbd9e6782f4c9d2444d : FAILURE in 59m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/243c3442e6bb45e3bd625d053c7c749a : FAILURE in 1h 09m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14554acccdd04d069e5fabcca4098c51 : FAILURE in 58m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2922921780bb4218862739a62834d731 : FAILURE in 2h 17m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0e40b767edab4374a45da8f638f0589e : TIMED_OUT in 2h 34m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e36b6b17627d41539a4c3a78fdee6f14 : TIMED_OUT in 2h 32m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/63276df8824648feb193efcc26433940 : TIMED_OUT in 2h 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/82f41e171e194d0e92dc4b3f4b437938 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ed9a51de907b4cfeb19b567730a99f60 : TIMED_OUT in 2h 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4a0d2608505922eb57cbd454a4925b00899c3ef1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:57:20.000000000","message":"Patch Set 8:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9a5cf9835e1475888f4153d5a240794 : FAILURE in 1h 50m 11s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/2172f1a018de447b8cc7aeb71f7f60dc : FAILURE in 1h 05m 52s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/4e01d0d2f4294595bed236cf6a8341aa : FAILURE in 1h 14m 38s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/013888b7e4d742d59a53b758b593cbb2 : FAILURE in 54m 04s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b8b82d1e2214f349796dff694edd223 : SUCCESS in 1h 21m 40s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/dd751ec2f291441cbad0d2932a692ebf : TIMED_OUT in 2h 34m 13s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/510c43a7057741c4983e429dcd8d43a7 : TIMED_OUT in 2h 32m 20s","accounts_in_message":[],"_revision_number":8},{"id":"38a2e735a008b1bad8ae4437564255e10ca39891","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-31 07:24:55.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"015a7e890dcd4dd538866dfd2667f5990df6d15a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:14:53.000000000","message":"Patch Set 8:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/1f269240997146e3b3ef3c6dec704db7 : FAILURE in 1h 25m 26s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/b42eb5a11e2a436aa9e6ec117d79f28e : FAILURE in 1h 25m 44s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/6e9565cbb9244a9c8d332637b3526106 : FAILURE in 1h 30m 47s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/810bb53255ec492e9a4471bc3aba7ced : SUCCESS in 57m 09s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/735bc09fdf2c44ab9df842e7e271b896 : SUCCESS in 1h 44m 04s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/51dc7da63fce4be9a10e1d865ef84a3e : TIMED_OUT in 2h 34m 56s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/710279a3d3824238847c712ae1198bc5 : SUCCESS in 1h 31m 10s","accounts_in_message":[],"_revision_number":8},{"id":"557249aadf4404f6bf907046bfdcf816e7da9cf5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-27 12:41:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ad9de2e29930b9da4dfa47ba358dfffba5972085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:07:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cf15960ba2a4c85877bb8aeddd6da90 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e123a2be16d0417d9f159cf566e5daa3 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a1b2e292ff40ae880161b4524f4032 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72de3048dbc046d2bee67a7455140768 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6faf30184b04e13bdf70eadf7c7fd7c : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4432ad0e16546e4810623b0bc2fc292 : SUCCESS in 13m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/177658792552484dbd2af1a2543f9cfd : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/55fa288bb869486994aa34a272e092e7 : FAILURE in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6832a454904c421f84875f49513561e4 : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d2685a6b28f410db0055bac67f698ad : FAILURE in 10m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f563aa2946db4344a27c1bb500d60896 : SUCCESS in 1h 10m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f51c5c84cf648bc8846c6a420f42e47 : SUCCESS in 54m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/caf2d3cb09b34e448a0d5ec83d83e237 : FAILURE in 10m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56c2a0893b404d50a9bb084b7d396136 : FAILURE in 10m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/34ee2764eb8c4a9582fb77a9b1e903b3 : FAILURE in 34m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cb37cb2a0cc94af299364ff14734e1b5 : FAILURE in 9m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e115942247de455bbca73cdd0028ebc1 : FAILURE in 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6ecaab6519424614a1e13cfce07ec625 : SUCCESS in 1h 11m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc87cf883cee424da0b01804d82f7c00 : SUCCESS in 1h 03m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/23108e2b58eb47d89b6a081cb6826f22 : FAILURE in 10m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7ae93fe569754679b2ea19b9b9e35841 : SUCCESS in 1h 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"05ead0bb62c56ee21abc4e5bc6eedce96e0ee832","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-31 07:28:00.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"c00ec0a25e1dd5f2b54a6b55d696d40714849991","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-31 07:28:12.000000000","message":"Patch Set 10:\n\ncheck experimental","accounts_in_message":[],"_revision_number":10},{"id":"d62600afd264ec7f6331768b9d6dc2e8ab5af7ce","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 07:56:58.000000000","message":"Patch Set 10:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/c4d1c27b227a4b539363baf2cdd6932e : FAILURE in 14m 31s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/e0ef1f4fc2b4480ba2e3f648299b97bf : FAILURE in 21m 58s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/dcf3b92889f94e54b13ebea8a343551e : FAILURE in 16m 28s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d20e0b5d1c644f35a440ff38de41f744 : FAILURE in 16m 11s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/ca45d55ce7f54bb9b3ce8696a478283e : FAILURE in 26m 30s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/b04a1abb41de4cfda9cab757bcf382c9 : FAILURE in 16m 07s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/76b813acc45b4e848d55ed396cf86c3e : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":10},{"id":"82e4a2eeec45f7d687cefece919b6f4cc76729be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 08:49:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4fa76ba37674ad09ada5ebfaf875629 : FAILURE in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd8d6cb4e4524066b9904bf60a020f33 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2a17833f834517a738cc79ba61fdc7 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c937cdf9c144666bd2a58485be73a86 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0de9dddaec1d4273b32fe7006df04fdb : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9efe9f0ddfb4bcbae06ea07167753e2 : FAILURE in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4cebacac821c4f7a94ea3e35a1350190 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9313e1596b734296921cc705bb9cd9f1 : FAILURE in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ecbf37873064e4d92db3c2be65ebc5a : FAILURE in 5m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31d5eb14b7014d8996d1011a45850cf1 : FAILURE in 18m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8b6dbb22dd647bb8e53bbc0cd0c6566 : FAILURE in 21m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/031f8dfff1b34bfbb5b73122e9f4cf4e : FAILURE in 23m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2cb47e40e2aa48ec8b7b0ee1b51794af : FAILURE in 20m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ce77c47cd904db1800ed74cc53dc93d : FAILURE in 21m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/675e79be5c9d4b359f5b0bc83bc8d3cd : FAILURE in 22m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec4e32d4659b438d8ce868f3269f5b64 : FAILURE in 21m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa8f9bd47fff47fbb643049c7876e6a2 : FAILURE in 23m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/81757d7060ff4b129a11227456c9aaac : FAILURE in 22m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d01c36c5dac4cda92104387855b7c2b : FAILURE in 21m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/68aeee4c6f974d528dae7be05d60c0ce : FAILURE in 15m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/87650b00867b4d19b49f9bdf18fdccab : SUCCESS in 1h 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ecd2384b9d48e5112fd9f15f111fc82fea714536","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-02 07:42:27.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"4d5f91a12c7c59fb72506491d9f8b208209ef29e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-02 07:42:40.000000000","message":"Patch Set 11:\n\ncheck experimental","accounts_in_message":[],"_revision_number":11},{"id":"4054d82c6d4b07cede6d8dc69f6d82572b954459","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 09:16:22.000000000","message":"Patch Set 11:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/02bfa7aec22446e5b89311823bb8e286 : FAILURE in 42m 43s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/26e322f8d7eb470ebeeb809e9e8fd6b5 : SUCCESS in 58m 22s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/01e25195ec904b12ad2de579b9396116 : FAILURE in 40m 25s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8eeeb17707514cf2929ca07451ae8aa5 : FAILURE in 59m 20s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/6558cc607e1c47b8b11c33867f1e643e : SUCCESS in 1h 28m 22s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ad9c44d86c342c589c15243d7994592 : SUCCESS in 54m 16s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/b34fdc2d29fb4c52b22167505aedc4f7 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":11},{"id":"d444234a6acc08f1430b283afd914038579847fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 09:43:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52330fd738244b4ebf7d742b2ab1a920 : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98194e80a728401395ea2ab535388c1b : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c6f745b3d5d454baf89f03832b58ec6 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3911020761da4a25ba62c517cd847711 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f88a185f23b49a8bd9c9461ea089e49 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6580be4bc8ae4fcab7b9ec381eed40cc : SUCCESS in 13m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4caff4d1d1e24bfcb3dfe9477d8b070d : SUCCESS in 6m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/46628dfdcaaf4867aaad94b08b95bce8 : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/625a0a6048f74920bbcbe36a88b6460f : SUCCESS in 9m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1bc965fa97854511bf4baee30af31aa5 : SUCCESS in 1h 01m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26c0559371464d3b96d54007dff2e591 : SUCCESS in 1h 22m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6980e0c0cb9f4eea95de173ca0e6f9cc : SUCCESS in 1h 04m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c19bb2356bc3473e8a60b99522839945 : SUCCESS in 1h 01m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b12be9e68034052a8c95a078f9c49d5 : SUCCESS in 32m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/30c892bd08fc4f7693fae6f7dfd6ef57 : SUCCESS in 1h 56m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae56a51efa9b442199c977dca462921e : SUCCESS in 1h 38m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c758ec1824614e04a6e303e144597f04 : SUCCESS in 57m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/348b35e00ddd41c5a4cad6505305fdee : SUCCESS in 1h 23m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5e40565973b74dc698c81eb8a18515bb : SUCCESS in 59m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4db31942cbdf4777907c1734c4c13146 : SUCCESS in 1h 32m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/74b4ab851d28435bb68d7f3399f144f6 : SUCCESS in 1h 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e9437c79215648ff2e57b5173f78ebe418e52389","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-15 13:08:55.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"734f6e27b4ea9ff2ad8af9250df8aa6172cc206d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 15:26:11.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7be598eb39ae4b02890fc25f353cdfaa : FAILURE in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6708ffe046e49d5b9a960649fbb2701 : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156d7d86a386424bb77b40996d93bc82 : FAILURE in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2c81e5962040eb8304ddf984c511fd : FAILURE in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b28623f984644ce099d5b290cbc5361d : FAILURE in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1cde40d9ec24b1bab6b8c5ca5380261 : FAILURE in 15m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/827508ace3b94b28975029731274efb9 : FAILURE in 3m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aa081128f4594dbd87c0102b0e49c5d0 : FAILURE in 3m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d7d49fb1f82435eafd815ad30c470b5 : FAILURE in 3m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d8251bca4df4b5286e9c5a4a1967a40 : FAILURE in 13m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11f7f7728b0f48b5872b3c12100c439e : SUCCESS in 1h 52m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d30de5fe7b44ee988d33669444ddd84 : SUCCESS in 37m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2bd21f58311147fca25c90469c5b693b : SUCCESS in 1h 09m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb4298e925d84682a11c837a9caa1eac : POST_FAILURE in 11m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f00477e5a169494d93eb3769105234c3 : FAILURE in 16m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ccb2ff8752446d789dfed532e073a66 : POST_FAILURE in 11m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1bc9e3c126d5404e9fdad15c99860083 : FAILURE in 18m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/930e631cbe6a46e393ff370919c361aa : SUCCESS in 2h 11m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2affad01943642528d5c20a66aaa5333 : FAILURE in 12m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a53aae0577ab44c7883a1b84bed1f9de : SUCCESS in 1h 54m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dd95ef8a94ec497985e3fb1efeb89b77 : SUCCESS in 2h 02m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b663e4085940e34aa05368f29dba0274b7676eb0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-18 07:59:57.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1b9c4023d0d34cc5c2b24e486f2b09e82b105e4b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-18 08:00:21.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"f58a9882281fe71f863f1b013ae007ea5481a547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 10:32:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baefd24f223248d1844f5777e67f2083 : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e2d73166d5c40beb45b4d71c38e52b4 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e9458cf605433c97e0196460c27f85 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1177486634a0452093102c22345d1cd2 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11d89fbc72974d3b8d10dcf569609ebe : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ba41f410cc46bf965e48ffb0dba47f : SUCCESS in 19m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e98c23d3fc3343cc85d9418eb5fe2dba : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2aa42ad66f6241e1b3fa5899264dfa9c : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f16bb52076845338c7be5abb0850b33 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6084800e8f1643bea0aeecc86744fbdd : SUCCESS in 2h 03m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3a6d63d522746a5b8804b6b5ab46b80 : SUCCESS in 2h 07m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6d123da5a037434f9f151a1b10f9bbc0 : SUCCESS in 50m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4497382e34df4584a2690cafaa3f9a63 : SUCCESS in 1h 08m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f8450f61bdd4edc80b516131d3f5a59 : SUCCESS in 43m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/788ccaf342484612bb98f4c4829ac71d : SUCCESS in 58m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d32388a4f0494699b92d435453f69a72 : SUCCESS in 2h 25m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/677ad2d207044cb495e9e0e22c5b5b4e : SUCCESS in 2h 02m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fe1e175b854472899aa1102df7f9623 : SUCCESS in 1h 43m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/43159fe3e62b4425b125daada52d8250 : SUCCESS in 1h 49m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/49227c108579437c9ecb6568ac398c74 : SUCCESS in 2h 05m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ea6069ecefda48879aee51c549234748 : SUCCESS in 1h 58m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d8e302f0d170c6f059a5478ec36dab1cb446e341","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 10:42:26.000000000","message":"Patch Set 12:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/51b1fe63da9d415f8ad6c145ec95eea0 : SUCCESS in 2h 21m 36s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/f6ee786e162e4042920c48dff22f9461 : TIMED_OUT in 2h 13m 48s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/0dd207a4473a4bd2b0b907d6a3072c92 : SUCCESS in 43m 52s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/e793127c5e03417793748af353e32c3f : FAILURE in 39m 25s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e21f1bc11d44dd2840e9fd13d0959ec : FAILURE in 44m 40s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/4574260a67264665b6308f8e603f627f : SUCCESS in 58m 03s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/154bd70113cf4642b42cc358590024f2 : TIMED_OUT in 2h 38m 22s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/d6a716e71b6e4984ba4a3fd828ca75a4 : SUCCESS in 1h 59m 09s","accounts_in_message":[],"_revision_number":12},{"id":"910ce3ab66695a22b2d2218cf28e9cd94b852fa3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-05 11:45:30.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"96f8bdba3007636463204bf39f81781a2f1e07ac","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-05 11:45:45.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"dfc8337606f3a0f0d84a4f05c4fa7ce03f2520e9","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:19:40.000000000","message":"Patch Set 13:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/cb14db3b5434406dbf68eecf87f445bb : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a6b18dadb2546b09a3dba693fa9ac7f : FAILURE in 1h 55m 53s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/031945497a304248b4c081a1a856b70b : SUCCESS in 40m 02s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/6ebaeefe1fd5446e94ba44a46af42738 : FAILURE in 39m 22s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e6b038a675604344997ce83a47c25f9f : FAILURE in 48m 03s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/8e09510cf26e4a28b22e38d2a3692459 : SUCCESS in 51m 42s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/7383b0c4d2e946cba13336576bbd0258 : SUCCESS in 2h 13m 09s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d915d9a72f14854bb4c70cda466cfdd : SUCCESS in 2h 29m 23s","accounts_in_message":[],"_revision_number":13},{"id":"af8724b65eb1dee75b239999760fb467ae743518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:33:16.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a77fba4808045b489044c0f047c23bf : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/550f5d663d8c4e07922a9819cbc4ed3c : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc06813ef43a42c780b0973196870296 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feb70154bbf84f1bb2811280b09d1d09 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6fc5b1dcc2047d7b5787b801385ee2a : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10d0cf9df3a41fb9fc70cf657f4dc82 : SUCCESS in 11m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4b78fcdeeea34c308c228fb899d19263 : SUCCESS in 8m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bfa05d71ff1a4dc4860d983904934bb3 : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7536488e3de246c385237f0cc551ea6b : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a6a77c4ada540a0b514b7e283f7822a : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/beaa8079783349c5a2b35ed163d1f231 : SUCCESS in 1h 30m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c37e84f3372b4d4dacb746a61354158e : SUCCESS in 2h 14m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91e886c990e54cb6a6e76f3cf110bde4 : SUCCESS in 45m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/30a53329708848c5a7a5066271003d9b : SUCCESS in 39m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebf247f8551c40cfbc46b0dfe9ce9abc : SUCCESS in 41m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/261121f2c8e4477b975fa2da52577f30 : SUCCESS in 59m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c0e8c913b108423980cff6cc85daacd7 : SUCCESS in 1h 56m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/306f347ab3724bd58b64c9d77cdde51b : SUCCESS in 2h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9742c5cdc49d47b583533f49660598ea : SUCCESS in 1h 59m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c3f7f571db4a445a8da47a80df7877a6 : SUCCESS in 2h 28m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6fc0b870ee374e6a8d6f4649456ac225 : SUCCESS in 1h 55m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4181be9470c240f2ab86ac3f4ad28a19 : SUCCESS in 2h 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"08bb35f7d734c3f8558abb0e5eb1e2278a12f3a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-30 17:34:36.000000000","message":"Patch Set 13: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":13},{"id":"b459b67636cd3a866b3d272a81395a469d68256d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 09:02:09.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"d0164f45b43b9d62e688cd9fa210d79db7c515ac","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 09:02:15.000000000","message":"Patch Set 14:\n\ncheck experimental","accounts_in_message":[],"_revision_number":14},{"id":"6d53e504b3a0a72fa55cbe1523f420aff9ae3f6a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:10:54.000000000","message":"Patch Set 14:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/f5acd432b8674926a3be7ec9596ca7ac : FAILURE in 1h 52m 11s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/500625784d3245fdababb684ba99e575 : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/f327792c29784aab9f855d46ca97c355 : SUCCESS in 39m 40s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/750e3c66d02c400f98bc6bebed90fd90 : FAILURE in 42m 09s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9ba32a0a7ed14700a991435f52832a5b : SUCCESS in 38m 12s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/62ed1a8040a349c3b55b5df595f71008 : FAILURE in 55m 12s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/4dc5a95d44fa43309413411ca77ac9d5 : SUCCESS in 2h 01m 54s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/dcf628ccc7e644d1924e74ba7f9f22e2 : FAILURE in 2h 30m 30s","accounts_in_message":[],"_revision_number":14},{"id":"75a4c49dc3085fff935f3361ab8e7720465f5714","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 12:39:39.000000000","message":"Patch Set 14:\n\noctavia-v2-dsvm-spare-pool-amphora-v2 failing as https://review.opendev.org/#/c/760380/ is missied\noctavia-v2-dsvm-scenario-two-node-amphora-v2 - failed to deploy unrelated.","accounts_in_message":[],"_revision_number":14},{"id":"ba2385e72eab874b29eeca47f778fb467f7898b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 14:21:59.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcd934011fb249ab9531b25d31e85d7a : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daad2d76e8ce4367a95170915cfe269f : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d587c2af3d4b69996b79606e314c73 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5de22c687f1141db92e7e8734d226501 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc863f762b33434a83bc35f879f4c718 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0da32cbfc0149aaba4175d98bb92053 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60ad343e5724483a5c7fd43bfd85cf8 : SUCCESS in 10m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/62873a95e2cf4fb49b57ed17a9a5fbb5 : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/663029d1d1634eb19848481d49ee4903 : SUCCESS in 8m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/899731ce87cb444b9586f90ed02dc3a5 : SUCCESS in 4m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5c98644a87d4cc1a22ba3c8270a33da : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcfc5392db684f4ebcdeb68213db775f : SUCCESS in 1h 14m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b2dd3d1e9e24268b41ee0e97cf31e1f : SUCCESS in 1h 52m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de9733fa600547d29554b77c8e650102 : SUCCESS in 45m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d502354b716241818a5b8d45ab346ae6 : SUCCESS in 43m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2aaec8d4d52f4d4081e92305f7596d7f : SUCCESS in 41m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6cfdbc8c6267403fbe9593b70f8d5052 : FAILURE in 52m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cadd5268cefb443ab86d0027aba4402f : SUCCESS in 2h 09m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b34bb4daa577489d9bd5e3facd9844cf : FAILURE in 2h 01m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/838a5ee1b7054060ba47c41f8587c1c7 : SUCCESS in 1h 56m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/edbb00f2941840b9bdfb9a2d76cecbbc : SUCCESS in 1h 46m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/91d627774da545e28c934b6e7b7484f9 : SUCCESS in 1h 39m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/40d63949fcad4534acd8cb082aefb88b : FAILURE in 34m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"66f14246d11357eed1f79df986a88d9aeb54a8cc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-06 07:07:09.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"7e0f3170069da3037048ed64b0cdcc9775d2dce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 09:24:24.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3991526199143a2b8907245f64cb15b : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/800db8c4456f4c0aad8c892280366db4 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0446cd5fb872405dafc5f984f98e9189 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b7dd0ba70d45aebbb89540744fc591 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db04ccf82d234433828a5de02537823e : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9bc17edc887454980cb1c3fe205243a : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a0cd16edaa4060b32e6b138fb7a5e7 : SUCCESS in 19m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a5d28079147456a9486bdcaa56c1eba : SUCCESS in 5m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/56e1472370654e1fb86fbf9431a00948 : SUCCESS in 9m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/deea6b47e6ee4afba110e005bcb6fc6a : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c1b3f88946c447319c68d9cdac552ad6 : SUCCESS in 6m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/506a871990ba4e8386d1bb795881a61f : SUCCESS in 1h 17m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b97410385df4023be79716bb431711e : SUCCESS in 2h 06m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ceaacccae5554e9daa0545c3458c5696 : SUCCESS in 39m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/893dff3759814fafb13199fac96ccf48 : SUCCESS in 41m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05f82e47012a4066b9b118e9a96f3a54 : SUCCESS in 38m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/11c09b7f9f994c7795bed729b8abdac8 : FAILURE in 53m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/804406718a1f4b46819a171b7328aee8 : SUCCESS in 1h 45m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5919b15b36134de5a32fcba321a58b1c : FAILURE in 2h 11m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc5e3fed540a4b788a4444e1a7892594 : FAILURE in 1h 59m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fffc8772978a4f6d8989a6a63ebd5ba7 : FAILURE in 2h 06m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f82cfd083b8e4f429a06d076a03de619 : SUCCESS in 1h 59m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c682548052fa4a7dbde18162f8efe707 : FAILURE in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"73747fcb06349bad0e020c9feae85ed0de6bdd88","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-09 08:51:31.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"74e7587876675ae80c270468845fcdfd061c7523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 11:53:36.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60f2e938b33548168e558d4b181ce07a : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f41c3f6fd05448c813225e0ef085d4d : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd59867dffe40378c9a413ff7ad0dc6 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be5074bcfce8423d95dc8c363f4faceb : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b36ec363840e4e27a6c309e4a202a115 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/857c1b1ababb41e3948264d89903679e : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642284106c2b42bf8b6a870e9b5216a5 : SUCCESS in 18m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7aaaaff302d2497586fc258219a3ce1c : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8bcfdf6bb3934460bbfd7906f8ba4603 : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9fbbad1ae007429795a2291de700a518 : SUCCESS in 4m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/231d51c6b496493485bb7679d07ad350 : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9aea60b3a92c4f81a4b896ec8cbef4ea : SUCCESS in 1h 05m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad17d6a5043c4b649609770edbfa3ad6 : SUCCESS in 1h 56m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2109dbf45ed54026ae7efbfc740e4c40 : SUCCESS in 40m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1c5dbb05a6f64a39b343ef621a3f0d55 : SUCCESS in 40m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68fb4620808f44448fdd28571a052749 : SUCCESS in 46m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc71ffb62ebf4fa79977bbb931de9432 : FAILURE in 53m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/58ad86cbbbb34d24b3d1360acdb13451 : SUCCESS in 1h 35m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd0a889957f949f6ab66c99b147d6924 : FAILURE in 1h 58m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c1bc4854998645a99aa3286c04e89821 : SUCCESS in 1h 57m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/20676fe4f287413e89f95bbb76e5d1c3 : SUCCESS in 1h 59m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a8aa901edad348bdb9533ab018e5a9ea : SUCCESS in 1h 58m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/79957dbef409440da3b853b4cdea2c9b : FAILURE in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"827ddcc097dd47a6aacbd7894ee1043a9b663539","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-11 08:44:12.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"605802808459f10370912f6b157211071fd5437a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 13:01:01.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/982293d5b74143a9a58a9839bdaa2b4f : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3bb1c9138ea4aacb322c140bd47af78 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa794506f2a4c599c36bb60ea4d76d4 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58bd86468bfb4911b006f06f88d67de7 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7d028d2586482e97c29c4b505cc451 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c5bdb4ca19f40579d814615d1efe070 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/786b7308a8e8477cb4084998818cfba2 : SUCCESS in 18m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7fc04bb0f165404199d0c067a2186244 : SUCCESS in 7m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e77afd9585514d71b8db7d024aeaac67 : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cafabed7b007470fb614f52d39a0c0ec : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b0585f339584a39a2e50b7c2f36bc37 : SUCCESS in 9m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3369a9972e1f4e8992c348aef8e39aea : SUCCESS in 1h 18m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82e0c03b5b9e4f52bd44b37de79c29c6 : SUCCESS in 2h 01m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c95dce8081ce4229a9e0bec5eccf5cca : SUCCESS in 41m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c505549807754926ae448e96eb7656ba : SUCCESS in 40m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aeaacf0bbe584c99ba3c6256b92b190b : SUCCESS in 37m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6e8f2f2d1a443ab9a18293663c60e0b : FAILURE in 56m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a8b09a7fcd194bd38cbca2e66217b6f4 : FAILURE in 2h 15m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5732dce3634e4ea6ad937ec3b93cbeb8 : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ddc16d6d0e9a4c1189293f44192724d8 : SUCCESS in 2h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/18a41b2d17f6449a998c091971b667ad : SUCCESS in 2h 19m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ea17066cf1e94ccb867ca8c5939353ea : SUCCESS in 1h 44m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/83d8661074f949af99db54b87873e89e : FAILURE in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"66c03e4c45e77bd5a141f40f534c15c33de646b9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-17 07:28:46.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"c2a08bdf4e1592512f2f9aefc9950f8b448c8af3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:57:01.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e27a34eb933a44769b0df7b73176ef02 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4b90a5e91fa4aaead3baf1adc2702b7 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c00a87a7b0742d19cebbe437a4e8678 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eae95aea24e240558d296c633b186be8 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aceaa25a839f4a0b841f6caa7bcee3f1 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12e2b491a2274881aade99d8b0a2ca2e : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593623ef5afe4916b9a84f97d128a01a : SUCCESS in 18m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5aa6d990579a44baa808a0da1cd4e1db : SUCCESS in 4m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/172e0078ef974bc6b6ed6a924b6b045d : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cde71ab0df3b4a07a91fb9cb2caaac66 : SUCCESS in 4m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05a6b47345654d6bb42c783a5015bca0 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49d7567c8598406594668c814b7ee3ef : SUCCESS in 1h 17m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27508bb550074e65aff295961f459e57 : SUCCESS in 2h 03m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/84e95f05a9b7440eb955f982d4d319f2 : SUCCESS in 1h 08m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/09bd99ab362a4610912d2214b92f1790 : SUCCESS in 31m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1aa68d54fc8b4fb5b502c4c8abec6556 : SUCCESS in 43m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/177d4e4a6dd14607859eeb7483e394eb : SUCCESS in 59m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/38c488283d3d4acf87a3719a61bb518b : SUCCESS in 1h 57m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d349720528546e498177d5ad484946c : FAILURE in 2h 24m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c33e43f5b8924288949227a6c59ccc26 : SUCCESS in 1h 47m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f47acacc6fcb4b778f45068ebcc5be99 : SUCCESS in 1h 56m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/847fded3d0c54fd39f729ddfcfcdfc0b : SUCCESS in 2h 12m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/91a17d319132400790382bb43e09b015 : FAILURE in 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"03965a3b4c9f2a9caed83a0043642b43e7f83bc5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-17 14:36:18.000000000","message":"Patch Set 14: Code-Review+1\n\ncheck experimental\n\nI think at least the grenade-amphora-v2 should be fixed now.","accounts_in_message":[],"_revision_number":14},{"id":"4c1f9d7d1900d0ad3f69c842021ed4d3624605f9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-19 08:54:33.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"03ca2f51c96ed0ee5a104348630159e1430807bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 08:54:50.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"fd09a1897b742bd035bef404ffa01b049d1c6d2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:25:02.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b616cf7970254b018c9e27ac94fe9287 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a7701862394ec4ab8caf5a6c94385a : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9fb195196b3410186029950636eb51d : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6b8abda0c604093a249853928a8f755 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/697c6eab865c4294b0176471a6046dcf : SUCCESS in 19m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b44fe40d02ab40efab955c979fc8816a : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff41367be86f4cb3a15a1161cea12fdf : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ab7dfd3c70d4aa6b8049d11d23f79d6 : SUCCESS in 1h 20m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f8235f53f6147cc901f4bb9892b1902 : SUCCESS in 2h 22m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f19872a7fa6408a938456a87d8f168d : SUCCESS in 43m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d15da9d4f09d44f7ba772d3409fbb927 : SUCCESS in 44m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ccfd49ade59455f8588b5739e39b5bc : SUCCESS in 37m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4774017efcc341058d934c74558443ed : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":14},{"id":"ad3487aecbee31894d7133d89457592d5e87b38e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:25:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"0aa242509674c76708725ad6314ab6d87318ed49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:31:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0612391edf4d4525ab3febdfd7f39b5a : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"9a6a90000f6f729bee23bc16882a0ee0c08e2097","revisions":{"0d825175c5a52b7d283b74a8576bd768ef19161e":{"kind":"REWORK","_number":1,"created":"2020-06-25 12:26:15.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/1"}}},"commit":{"parents":[{"commit":"41c628a084002017d2003926cf0e25ba3ffeee0c","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41c628a084002017d2003926cf0e25ba3ffeee0c"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d825175c5a52b7d283b74a8576bd768ef19161e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d825175c5a52b7d283b74a8576bd768ef19161e"}]},"branch":"refs/heads/master"},"64140e276e2bd141d51c12d8b778255dd3c6ff2c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-02 09:23:57.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/2"}}},"commit":{"parents":[{"commit":"ec2f89830daf6f31b120bc0d10f6f41b2ea092ba","subject":"Merge \"Fix UDP pool transform in LVS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec2f89830daf6f31b120bc0d10f6f41b2ea092ba"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-30 08:31:02.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64140e276e2bd141d51c12d8b778255dd3c6ff2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64140e276e2bd141d51c12d8b778255dd3c6ff2c"}]},"branch":"refs/heads/master"},"a2531c3b6c4546591a009ed761397bd34905aee9":{"kind":"REWORK","_number":3,"created":"2020-07-09 07:37:38.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/3"}}},"commit":{"parents":[{"commit":"ec2f89830daf6f31b120bc0d10f6f41b2ea092ba","subject":"Merge \"Fix UDP pool transform in LVS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec2f89830daf6f31b120bc0d10f6f41b2ea092ba"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-09 07:37:21.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2531c3b6c4546591a009ed761397bd34905aee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2531c3b6c4546591a009ed761397bd34905aee9"}]},"branch":"refs/heads/master"},"d79e351611dec8c6ae6f5555a5ddbdd750b3307c":{"kind":"REWORK","_number":4,"created":"2020-07-09 09:44:49.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/4"}}},"commit":{"parents":[{"commit":"ec2f89830daf6f31b120bc0d10f6f41b2ea092ba","subject":"Merge \"Fix UDP pool transform in LVS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec2f89830daf6f31b120bc0d10f6f41b2ea092ba"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-09 09:44:32.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d79e351611dec8c6ae6f5555a5ddbdd750b3307c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d79e351611dec8c6ae6f5555a5ddbdd750b3307c"}]},"branch":"refs/heads/master"},"19475f7c26ed0dd6a75c1879f71ad07f3caef2fd":{"kind":"REWORK","_number":5,"created":"2020-07-09 09:58:30.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/5"}}},"commit":{"parents":[{"commit":"ec2f89830daf6f31b120bc0d10f6f41b2ea092ba","subject":"Merge \"Fix UDP pool transform in LVS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec2f89830daf6f31b120bc0d10f6f41b2ea092ba"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-09 09:58:17.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19475f7c26ed0dd6a75c1879f71ad07f3caef2fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19475f7c26ed0dd6a75c1879f71ad07f3caef2fd"}]},"branch":"refs/heads/master"},"b393a706cb8e24e500af4f9f99c3bf3d491299cc":{"kind":"REWORK","_number":6,"created":"2020-07-09 12:06:53.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/6"}}},"commit":{"parents":[{"commit":"ec2f89830daf6f31b120bc0d10f6f41b2ea092ba","subject":"Merge \"Fix UDP pool transform in LVS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec2f89830daf6f31b120bc0d10f6f41b2ea092ba"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-09 12:06:41.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b393a706cb8e24e500af4f9f99c3bf3d491299cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b393a706cb8e24e500af4f9f99c3bf3d491299cc"}]},"branch":"refs/heads/master"},"54b536974799e9f31d2b708f84d0ca995e5580de":{"kind":"REWORK","_number":7,"created":"2020-07-10 09:39:48.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/7"}}},"commit":{"parents":[{"commit":"ec2f89830daf6f31b120bc0d10f6f41b2ea092ba","subject":"Merge \"Fix UDP pool transform in LVS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec2f89830daf6f31b120bc0d10f6f41b2ea092ba"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:37:30.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54b536974799e9f31d2b708f84d0ca995e5580de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54b536974799e9f31d2b708f84d0ca995e5580de"}]},"branch":"refs/heads/master"},"14228bc9d1370fd6bab96b0ebf929f0b6243fb4c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-07-29 08:45:56.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/8"}}},"commit":{"parents":[{"commit":"e35dc81cf4182b4b139c58cd157ebaf694282874","subject":"Update amphora v2 for the failover refactor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e35dc81cf4182b4b139c58cd157ebaf694282874"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-29 08:45:56.000000000","tz":0},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14228bc9d1370fd6bab96b0ebf929f0b6243fb4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14228bc9d1370fd6bab96b0ebf929f0b6243fb4c"}]},"branch":"refs/heads/master"},"eb073da30997c8305e0741eca1a82da8acd32c93":{"kind":"REWORK","_number":9,"created":"2020-08-27 12:41:02.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/9"}}},"commit":{"parents":[{"commit":"0b1d8dd5e76edc4b3058a36b47f6c059f970e516","subject":"Merge \"Update devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1d8dd5e76edc4b3058a36b47f6c059f970e516"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-08-27 12:40:49.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb073da30997c8305e0741eca1a82da8acd32c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb073da30997c8305e0741eca1a82da8acd32c93"}]},"branch":"refs/heads/master"},"6d125b24d464d0a21ff71a3ec3f96b51722361d7":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-08-31 07:28:00.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/10"}}},"commit":{"parents":[{"commit":"6d4d1c92b759c8af3e9f140b13466433fbc60840","subject":"Update amphora v2 for the failover refactor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d4d1c92b759c8af3e9f140b13466433fbc60840"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-08-31 07:28:00.000000000","tz":0},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d125b24d464d0a21ff71a3ec3f96b51722361d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d125b24d464d0a21ff71a3ec3f96b51722361d7"}]},"branch":"refs/heads/master"},"63ae940c06c88ddd6240e953fcc1a555c23b69ac":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-09-02 07:42:27.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/11"}}},"commit":{"parents":[{"commit":"81351cbc4925ec38dd5a048e7ca6de662ade7b70","subject":"Update amphora v2 for the failover refactor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81351cbc4925ec38dd5a048e7ca6de662ade7b70"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-02 07:42:27.000000000","tz":0},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63ae940c06c88ddd6240e953fcc1a555c23b69ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63ae940c06c88ddd6240e953fcc1a555c23b69ac"}]},"branch":"refs/heads/master"},"e2d1b636312b9d81dd873b6b8bf3cbf7030d2c8a":{"kind":"REWORK","_number":12,"created":"2020-09-15 13:08:55.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/12"}}},"commit":{"parents":[{"commit":"a28bc2a55a1e61a45c15b0f9bfcd057369423bbb","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28bc2a55a1e61a45c15b0f9bfcd057369423bbb"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-15 13:08:40.000000000","tz":240},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2d1b636312b9d81dd873b6b8bf3cbf7030d2c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2d1b636312b9d81dd873b6b8bf3cbf7030d2c8a"}]},"branch":"refs/heads/master"},"13c051853bf809fcc34e90eeb6e2b2abe71a477c":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-10-05 11:45:30.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/13"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-05 11:45:30.000000000","tz":0},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13c051853bf809fcc34e90eeb6e2b2abe71a477c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13c051853bf809fcc34e90eeb6e2b2abe71a477c"}]},"branch":"refs/heads/master"},"9a6a90000f6f729bee23bc16882a0ee0c08e2097":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-11-03 09:02:09.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/93/737993/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/737993/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/737993/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/737993/14"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-25 12:23:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-11-03 09:02:09.000000000","tz":0},"subject":"Add experimental amphorav2 jobs","message":"Add experimental amphorav2 jobs\n\nThe following jobs with default amphorav2 added as experimental:\n* octavia-v2-dsvm-noop-api\n* octavia-v2-dsvm-tls-barbican\n* octavia-v2-dsvm-spare-pool\n* octavia-v2-act-stdby-dsvm-scenario\n* octavia-grenade\n* octavia-v2-dsvm-cinder-amphora\n* octavia-v2-dsvm-scenario-two-node\n\nNote: octavia-grenade-amphorav2 will show valid result\nwith renamed alias amphorav2 -\u003e amphora.\n\nChange-Id: I03385d93575db4e44a72335c0af4fde490a93b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6a90000f6f729bee23bc16882a0ee0c08e2097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6a90000f6f729bee23bc16882a0ee0c08e2097"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
