)]}'
{"id":"openstack%2Foctavia~738023","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I0100aaa428477f011bd39a90dd4ec98199b4bebc","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I0100aaa428477f011bd39a90dd4ec98199b4bebc","subject":"Fix netcat option in udp_check.sh for CentOS/RHEL","status":"MERGED","created":"2020-06-25 13:37:39.000000000","updated":"2020-06-26 18:38:48.000000000","submitted":"2020-06-26 18:35:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738023-1593196556838-f0b752bd","meta_rev_id":"f060a016368d994ce30b0c8406a597a852f64247","_number":738023,"virtual_id_number":738023,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 18:35:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-06-26 16:43:52.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-06-25 13:38:46.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":1,"date":"2020-06-26 16:43:52.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 13:38:46.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-26 16:43:52.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-06-26 18:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dea8252e8dc4004e9d68e640e7b2a18458d1087","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-25 13:37:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee39978ad51a23af85dc466ad9358863221c3d93","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 13:38:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9b97b886f5432ae8024610f37ead901a15e774d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 18:27:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed00519dd25c4866a73a6939b9472c92 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/358ea6bf441341b7aa55c7ef7c231723 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac4b786b7c347fe97be28b8b9f3adcf : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b24aa17b376b4e29b167e25ac31b8957 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dc9e541e5894188acf6100ef9695afa : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b351bf72f89b4d299085cade9e238bca : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64bb5c0ece864a0aa77ca8cc90a9370f : SUCCESS in 12m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7cf1fafa29e94b82adb1041c1c1e1a2a : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f3cac59cfb1742b19b5e062835c33be6 : SUCCESS in 10m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a6cf3f3a21f04dbb88b65e0ecbed9146 : SUCCESS in 11m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec1c4af55ace4f0db14e8f9af1045b91 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/275ed8623c284a78aba54d573b92643e : SUCCESS in 57m 32s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/199c4bbb51614e7d81c8190c0dc0c59e : SUCCESS in 55m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e7c158c44ac4c8886ae4e176e79ffc5 : SUCCESS in 1h 48m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/949fed0036364f33b70a0a7d106bd414 : SUCCESS in 1h 50m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b31fe543ee0d4113bee6479ab1b9db0a : FAILURE in 20m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f0ed4c8246954407a1cf14b6ded0108f : FAILURE in 56m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9146194f0df44128be6057e8362ccbe5 : FAILURE in 57m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6dfb5439722b4073a20858ca982ce2b8 : FAILURE in 42m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/809a21ea92b147109e75efac2a862378 : SUCCESS in 1h 02m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39638954694641b5b6ea762f6faad842 : SUCCESS in 1h 55m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22367f8a0578422097b6884676aa8f3b : SUCCESS in 1h 25m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e94c84c7f8f5437b8c4b43570340d5d9 : SUCCESS in 1h 18m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3f1d4a24d7f7444cbe6dbc04d097563f : FAILURE in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cbccee6c9b43ce87322778cddd4b2e7f4724cc1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 19:35:15.000000000","message":"Patch Set 1:\n\nrecheck\n\ngrenade should be okay now","accounts_in_message":[],"_revision_number":1},{"id":"220582a6b63a6a11a1d2bd62f9c4c63088684860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 21:37:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d3996874dd045b08a43a12df3facf64 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fab070a12b6b42a1b2e536d47c5f7096 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc37d55c3634e9f8f78eac48678a1b8 : SUCCESS in 7m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00407c9d135244058334f0f7c3e24e27 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/072fc6c896594e64843f8f784affdd6e : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/274497cb686d469fbd9790f2e2d6a5a5 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe8d3b3e5464fcf8bfe1f490d2fa8b6 : SUCCESS in 11m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/37f5970545424c3eb66f79620b8400a5 : SUCCESS in 10m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/043bd50a2b4a4e97a163ef668be11340 : SUCCESS in 8m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1a431918e1eb4ed3acdd6835c222484f : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d1c874469b04abd8c3037c1ba80d84d : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea87a100bfa8462a8485958ba70a8cfe : SUCCESS in 45m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/40594efb69d54a1e9fbe1b3406a25ac9 : SUCCESS in 44m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c2327ca32d0426e9b2581e01565c896 : SUCCESS in 1h 47m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/975ba5d3b10f4647b625f139dd724ff5 : SUCCESS in 42m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2bc2b9b24fde41e5943d14f1da28af44 : FAILURE in 17m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5427099d3e9a4bdcb005ac93296ba994 : FAILURE in 49m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ad20605fe944e6db10f0f999bfd873b : FAILURE in 54m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7fd21ff79f5b4c5fab713d123681fef3 : SUCCESS in 1h 02m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67640d5dbac443c89b36ab0430e914af : SUCCESS in 34m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6358f9e200714c9e97eb8594ab102a1a : SUCCESS in 1h 54m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d86f8d38c8a484889fbd0595c87f6c5 : SUCCESS in 48m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/56154014e45843a395d4875631875afe : SUCCESS in 43m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ab7da415a5e5482894d44060909c8512 : FAILURE in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5563a881dd2b832242e3c17ad015165a4076601e","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-06-26 16:43:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30bd368875c5e0a93b7a31a1bef6c8eaf0598c09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 16:44:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22bbb30b1b3dc02078475464992117d650d723db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 18:35:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa24612147cb19e85bbe1b718c13ed581b6140bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 18:35:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cf3863ce7144d2a9acacc29b574609a : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c8a077191547cdba2c04b1eaec2e50 : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ae419607f9741b1a890f046fcb5ad6d : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bf6cd32678340e583d68cccbfb03118 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a15098a18dce42da8b90071e3a92e3c7 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a6077429b947f994d55e9514072460 : SUCCESS in 13m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4da34c07ca854bec847e142960df5aaa : SUCCESS in 13m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a3342ad24c5409a98ac6fe37a04e39b : SUCCESS in 9m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e2fb4eea89b4ca5a5a515c2d9485ed4 : SUCCESS in 46m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b6e1599914434850b2628ed3490f71b5 : SUCCESS in 50m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4685e1628a744069816ebb0927813bf1 : SUCCESS in 1h 48m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d6b778cdf3e54359b7463f09dee66268 : SUCCESS in 1h 43m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7164cd45ffe9436eb331f6e64ff940c0 : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f060a016368d994ce30b0c8406a597a852f64247","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 18:38:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42d36f7978514166a9b9ccb4418cb4da : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a4ccf49e94ed9c4c79e4181a81e25b7c3784add","revisions":{"1a4ccf49e94ed9c4c79e4181a81e25b7c3784add":{"kind":"REWORK","_number":1,"created":"2020-06-25 13:37:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/738023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/738023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/738023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/738023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/738023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/738023/1"}}},"commit":{"parents":[{"commit":"e358637a10e6b3c2df336d8dad42cecbd97e3dd5","subject":"Fixed a bug: replace \u0027startwith\u0027 with \u0027startswith\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e358637a10e6b3c2df336d8dad42cecbd97e3dd5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-05-19 09:22:47.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-25 13:37:32.000000000","tz":120},"subject":"Fix netcat option in udp_check.sh for CentOS/RHEL","message":"Fix netcat option in udp_check.sh for CentOS/RHEL\n\n-w (timeout) option doesn\u0027t do anything in nmap-ncat (default netcat in\nCentOS/RHEL) for UDP datagrams, and nmap-ncat has a default idle timeout\nset to 2 seconds.\nWe can get the same behavior as netcap-openbsd (Debian/Ubuntu) by\nsetting that idle timeout (-i) option to 1 second.\n\nThis commit detects the flavor of the netcat binary (nmap vs other) and\nuses it to adapt the parameters.\n\nStory: 2007688\nTask: 39800\n\nChange-Id: I0100aaa428477f011bd39a90dd4ec98199b4bebc\n(cherry picked from commit 6354f92ecc529c3f813733c8437768a432975dbc)\n(cherry picked from commit 209726d9197aebd57acefe29c263aa68bdd811ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a4ccf49e94ed9c4c79e4181a81e25b7c3784add"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a4ccf49e94ed9c4c79e4181a81e25b7c3784add"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
