)]}'
{"id":"openstack%2Foctavia~738096","triplet_id":"openstack%2Foctavia~master~I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1","project":"openstack/octavia","branch":"master","topic":"octavia-arm64","hashtags":[],"change_id":"I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1","subject":"Add aarch64/amd64 support to amphora and devstack","status":"MERGED","created":"2020-06-25 19:54:35.000000000","updated":"2021-03-11 10:30:59.000000000","submitted":"2021-03-11 10:29:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"738096-octavia-arm64","meta_rev_id":"bfdd1b180ad92736f1a856ec3ff5728d6f69cb55","_number":738096,"virtual_id_number":738096,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 10:29:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-11 08:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-10 21:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-11 08:30:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 15:40:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-04 17:05:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-10 21:55:37.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-10 21:56:06.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-11 19:15:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 09:01:24.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2021-03-06 22:58:15.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"e90a7be4683c08813e565d7e593516cacf8c508d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 19:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f90eefe2ea849de8eec17e0157fa5465d0900572","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 19:55:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8a0b05d32201b8dfb9fad472d8f15b612092f233","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-25 20:44:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4c03afa0baaa23aa5565763d8e11d84265926e7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 06:22:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c8202a5a51558df858b6243f09afbf12cf81c07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:05:41.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/a78ec219b9784037bf82561658c0134e : FAILURE in 1h 40m 54s","accounts_in_message":[],"_revision_number":3},{"id":"aae7cf2f7582a4f2289e5c9be7e0a6e7defa5bcf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 08:11:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c313b8270944dd3d9a2ef0f2e9f7e19d9e8525b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 09:22:09.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/c132bff950134ec481a385e5cf696e41 : FAILURE in 1h 09m 27s","accounts_in_message":[],"_revision_number":4},{"id":"823d8eb90f8ebe1ab50cf91d956ebe0347aeb624","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 09:24:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"500cee4e704f9b372750617a7a7a78cf782fec0e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 09:49:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1a2a17cedf5512194a3ad878d9bcd09ba43c902c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 10:56:27.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/7d45e69ad7724b29b4af4c451ef7385b : FAILURE in 1h 03m 28s","accounts_in_message":[],"_revision_number":6},{"id":"ac6c56d405078b46954d709fcb52fa54e81f5564","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 11:22:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b95332f4621ff1c65c8d27c1a678e7edd423066e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:24:30.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/c52c5646e8c042b58298b637f08a47a5 : FAILURE in 1h 01m 06s","accounts_in_message":[],"_revision_number":7},{"id":"f3f359694f1cb98bac1eed36d55b0cb64ba541a3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 12:33:53.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5654d659ef614bba9fcae67becc5a2e63d642615","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 13:37:31.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/d8df5382effe4979b685b944beae6a54 : FAILURE in 1h 02m 22s","accounts_in_message":[],"_revision_number":8},{"id":"93389a2482b921e6b082f1bcaa18f6b32fe0ef74","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 13:46:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cb8b0658b9682f3734152d58c48b152d75a8b71c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:57:22.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/1ec36964247c47c99e4fa2f6ba47a320 : FAILURE in 1h 09m 11s","accounts_in_message":[],"_revision_number":9},{"id":"dbd919f567b9e3cc28789435483444b94c85f7ea","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 15:26:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2ebf5f1a656e8ba3f7b68fa19ac78259997248a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:26:47.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job octavia-v2-dsvm-scenario-arm64 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      # templates:\n      #   - check-requirements\n      #   - periodic-stable-jobs-neutron\n      #   - openstack-cover-jobs\n      #   - openstack-lower-constraints-jobs\n      #   - openstack-python3-victoria-jobs\n      #   - publish-openstack-docs-pti\n      #   - release-notes-jobs-python3\n      #   - octavia-tox-tips\n      # check:\n      #   jobs:\n      #     - openstack-tox-functional-py36:\n      #         irrelevant-files:\n      #           - ^.*\\.rst$\n      #           - ^api-ref/.*$\n      #           - ^doc/.*$\n      #           - ^etc/.*$\n      #           - ^octavia/tests/unit/.*$\n      #           - ^releasenotes/.*$\n      #     - octavia-v2-dsvm-noop-api:\n      #         irrelevant-files: \u0026irrelevant-files\n      #           - ^.*\\.rst$\n      #           - ^api-ref/.*$\n      #           - ^doc/.*$\n      #           - ^octavia/tests/.*$\n      #           - ^releasenotes/.*$\n      #     - octavia-v2-dsvm-scenario:\n      #         irrelevant-files: *irrelevant-files\n      #     - octavia-v2-dsvm-tls-barbican:\n      #         irrelevant-files: *irrelevant-files\n      #     - octavia-v2-dsvm-spare-pool:\n      #         irrelevant-files: *irrelevant-files\n      #     - octavia-v2-act-stdby-dsvm-scenario:\n      #         irrelevant-files: *irrelevant-files\n      #     - octavia-grenade:\n      #         irrelevant-files:\n      #           - ^.*\\.rst$\n      #           - ^api-ref/.*$\n      #           - ^doc/.*$\n      #           - ^octavia/tests/.*$\n      #           - ^releasenotes/.*$\n      #           - ^setup.cfg$\n      #           - ^tools/.*$\n      #           - ^(test-|)requirements.txt$\n      #           - ^tox.ini$\n      #     - octavia-v2-dsvm-cinder-amphora:\n      #         irrelevant-files: *irrelevant-files\n      #         voting: false\n      #     - octavia-v2-dsvm-scenario-two-node:\n      #         irrelevant-files: *irrelevant-files\n      #         voting: false\n      #     - octavia-v2-dsvm-scenario-amphora-v2:\n      #         irrelevant-files: *irrelevant-files\n      #         voting: false\n      check-arm64:\n        jobs:\n          - octavia-v2-dsvm-scenario-arm64\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-v2-dsvm-spare-pool\n          - octavia-v2-act-stdby-dsvm-scenario\n          - octavia-grenade\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos8:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":10},{"id":"2424ef8ef8bf272d4913153446593de658ed8327","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 15:30:27.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"94e8eb33f5da840d8a81110c2fc984a39c8d858e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 15:40:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e0cc9485de18b916012733a61458c36220ec7dc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 16:12:41.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-aarch64 https://zuul.opendev.org/t/openstack/build/16d7b3bce1e642fa83b02f7c3fbe7d24 : FAILURE in 38m 35s","accounts_in_message":[],"_revision_number":11},{"id":"e1c3e7134e473efb53e16e48332a548ed782cd84","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 16:37:15.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5b84844b373c2a8da3a257d8c5f00d74f5820831","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:12:04.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/0a0322da080341a48f5cbb094e0a856c : TIMED_OUT in 2h 33m 42s","accounts_in_message":[],"_revision_number":12},{"id":"4c2dc61ea54c28e36bad8f77aca7194a966ee6b2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 19:43:48.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9523b921280a607d09d7f698a13152e1b4219025","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 20:35:34.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/9c01c69b91e4483db6499407af1bf4d4 : FAILURE in 50m 02s","accounts_in_message":[],"_revision_number":13},{"id":"1fa77dfb940e36ca4c8868eb5d5214716f35d016","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 20:52:20.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2557052ecaada9072011605f01aa15e38ec294f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 22:28:41.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/cde6d895599c4c2a8cd43218ff10d3e1 : FAILURE in 1h 34m 54s","accounts_in_message":[],"_revision_number":14},{"id":"a469400303dcf63d383f7c62294354178d14bec3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-07 11:49:24.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"39f56ba3ad0c9bca8406b47ac6209e7f6d77aa2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 12:15:25.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/2b0ef00ff2174b3b87e4a893b638130b : FAILURE in 23m 36s","accounts_in_message":[],"_revision_number":15},{"id":"c54a9bf4349460132dbd706e80d5fc5500c57c66","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-07 12:16:17.000000000","message":"Patch Set 15:\n\nrecheck\n\n[Operation too slow. Less than 1000 bytes/sec transferred the last 30 seconds]","accounts_in_message":[],"_revision_number":15},{"id":"f36e0443f5669253ca84e7a2edd8613134e25032","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 13:14:13.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/05c75f7471044e37b92726d1f5832b71 : FAILURE in 56m 04s","accounts_in_message":[],"_revision_number":15},{"id":"0cbed6024b6f5bec765c341278a3009f86564e03","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-07 13:24:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"407fb234ec3d49e3f8f8338cefee76a17bc1fbd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:21:50.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/51f73a848c6543f2b3b3fc1813c94a3a : FAILURE in 55m 14s","accounts_in_message":[],"_revision_number":16},{"id":"2508ab44896c0e2dc4bdcc471666a5e5eadb0745","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-13 12:51:08.000000000","message":"Patch Set 16: Workflow-1","accounts_in_message":[],"_revision_number":16},{"id":"23bf1f4d7fd9e7bb5ffb7713661cc09a69206cd3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-22 15:09:03.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"c1680080cd4fd7f03f423ef6add5976f6cdf2e63","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-22 15:39:56.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"b7c82e7497329971b4d7351bb1849c4eb58e7e6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 16:50:56.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-arm64 https://zuul.opendev.org/t/openstack/build/b90b9e2eee9d457a9abb8bdc4e20ec43 : FAILURE in 1h 09m 43s","accounts_in_message":[],"_revision_number":17},{"id":"a415065fec3c5748f965a22513e9a8816a465f8c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-22 20:10:00.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"085a33062cde2ab6f20f48050c01ea897965e40c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 21:20:52.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/4c65a48a5f544d4f87e387dc723820f6 : FAILURE in 1h 09m 42s","accounts_in_message":[],"_revision_number":18},{"id":"c6d9a7e913d3a583595a734c82ee3d71b721621d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-22 21:37:56.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"7d0306daf205fb85feba9b39f968a7279e1c4b26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 22:47:06.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/e5a3f425ff2144e6b8000554b3c8baeb : FAILURE in 1h 08m 00s","accounts_in_message":[],"_revision_number":19},{"id":"6da950e9475d143a8b950b23167487916b06f9f9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-22 23:11:06.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"cb6b4e89a0e0333579367ca0d6c1a56548dae094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 00:05:41.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/37f7db217f0e42edb04c043a61f51276 : RETRY_LIMIT in 17m 29s","accounts_in_message":[],"_revision_number":20},{"id":"d837cb65aeb0a44feef22a792f731bf244fc6813","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-23 08:59:01.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"dbf1966875363f25adea8dc97320f4f551dd5120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 10:07:48.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/2e91a38839124252be45fe13a6ce9579 : FAILURE in 1h 07m 45s","accounts_in_message":[],"_revision_number":21},{"id":"0d9f73c9d43fee0d0838ed0491bf494abf09d4c9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-23 10:10:39.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"7ff9dc0c40d19e4f8dbcc34fd2b9eb48ac3b1ef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 11:16:39.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/925769b822744e77af39df9f65720b6a : FAILURE in 1h 04m 58s","accounts_in_message":[],"_revision_number":22},{"id":"b00e28cf0ba590c7614b49d108cc73784494c48d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-23 11:55:25.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"855355f57d0cb066bc036eebd8f56f1268cbf262","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 12:57:36.000000000","message":"Patch Set 23:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/39b2c3e10fa54004b4f15105a857cba3 : FAILURE in 1h 00m 57s","accounts_in_message":[],"_revision_number":23},{"id":"34e865c238ce04decf70748abbae0681c01f7784","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-23 12:57:56.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"fb1e6765a73d36f9727462d91687e1e0ba8e03dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 14:23:34.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/89dc25e0968946b3a175733fa3e7c7aa : SUCCESS in 1h 23m 54s","accounts_in_message":[],"_revision_number":24},{"id":"de0387d546471a8856b56ca408d23ad16f2b329b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-23 20:14:22.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"00be3117671e439fefd52e97b038819015d48dc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 22:26:55.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/4f78c9b08fb54ca5b239452c02802b6d : FAILURE in 2h 11m 27s","accounts_in_message":[],"_revision_number":25},{"id":"1c6f5498b349873f40d815eb0c8d0f99fc0346f4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 07:52:15.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"d80715a7a6f8a9533da22b4893b12ea36c921d82","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 09:45:10.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"92ec0978bee7e3ff8dce1fef059998783c128897","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:22:40.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/485ad71109cf4e749c4439487c6ac261 : TIMED_OUT in 2h 35m 12s","accounts_in_message":[],"_revision_number":27},{"id":"da2ebb9b8b5573803101f3aec2631dfc349f7248","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 12:47:38.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"673cb804589c634b3459e6868eb925f991b541a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:08:32.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/c18ec541d18b4fb2ad1ce94d661fbcb6 : FAILURE in 2h 17m 42s","accounts_in_message":[],"_revision_number":28},{"id":"d2b11ef787677f3a9a4c1351da46f136044268ff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 17:47:25.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"b6ea74d8d48136934cc4e1a9f4b488475b13831d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 18:44:06.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/83163416f6a94ff8879f48aae5162d92 : FAILURE in 54m 22s","accounts_in_message":[],"_revision_number":29},{"id":"d884e1f141e57c8bfb103c63dc3554ac2e010a17","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 18:47:02.000000000","message":"Patch Set 29:\n\nrecheck\n\nmirror issues","accounts_in_message":[],"_revision_number":29},{"id":"6654aa15eef5f24c62b451a3ed734f8a58366091","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:22:20.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/dab455944a5b4f66b5e719142c4fa7d7 : TIMED_OUT in 2h 33m 26s","accounts_in_message":[],"_revision_number":29},{"id":"7f84465a388c8f20fda8165890c585d79655ed88","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 21:26:33.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"683ad6198c5fcd9e6362d88adbdbaad197b2a067","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 00:02:46.000000000","message":"Patch Set 30:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/067fbe960acb4ad3a93bcdf5d5a555de : TIMED_OUT in 2h 33m 28s","accounts_in_message":[],"_revision_number":30},{"id":"55df405599b6d626388d16a81492bb6b04e7d762","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 19:19:38.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"9c9ec99d50e0b019a1cea9e1ff85f1f4c964dbf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 20:44:42.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48b43531a96c4b3db2ab53a12e29845b : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28c5283811004987be0f11a95b3721d4 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82799f645eb49de8d3c515c7d61b785 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33e9a581527a403caed8ba92388068ae : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e1e1317d96b4fccba91f8b03d8c4419 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9302748da384aecbb6501e4c14e1992 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1616950221c48efb0c320c606f17783 : SUCCESS in 5m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c27a02841e404fa7818c95b34d3f11dd : SUCCESS in 8m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f8c201e7a97c4091b7fa7b42f7bfb74f : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad12696e32be4455ba5b532f744835e5 : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb0faa53a55e427681dd3cca53c05994 : SUCCESS in 47m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e17316f99bcc464289d6f1372e07682d : FAILURE in 50m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1cb8b161ea9e492ba7f90131c57133ba : SUCCESS in 32m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/315d49e6593d4ed3bb1d83846487aed9 : SUCCESS in 32m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1425ad0f7ea54b3eaf9e182515dddc8a : SUCCESS in 32m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8ab7de4ecbc9449784c3a75771293b17 : SUCCESS in 1h 06m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/248b84a4184841bba3f8081035afce22 : SUCCESS in 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/742b5015e43948fb80a154fe03d04076 : SUCCESS in 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d167b84864904f0cb17d94dbd6c115be : SUCCESS in 1h 22m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9ffab3f97604fcdaec23dcf74b1d6cf : SUCCESS in 47m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a6fa180722c34aac9b672394d743590c : FAILURE in 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"190a9e3cb5dca2eda926fd73fa64d5aa6b77f676","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 21:18:33.000000000","message":"Patch Set 31:\n\nrecheck\n\nneutron not applying SG rule?","accounts_in_message":[],"_revision_number":31},{"id":"470abe40c5f3af5693ffec3a935fae4cf9f070ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 01:25:25.000000000","message":"Patch Set 31:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f79201ddc3e4c278be1be6ad29cd332 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5165c309123046129e82c85352a67585 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82001e740d343efb644dfc75e028110 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0d7f76349f84aacb7e33fd6d8efa62e : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05c22ad02de44629b84bf4b56a508daf : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c54aaa70684c65b05e632b5f95d7a1 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/312218f70dd74788bc433c0fad5de3e9 : SUCCESS in 8m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a7c195a49d894feeacf50fb4da209694 : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/37ed3b5771dd4bbfab2753d5421b6b78 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1db455cf8524a6abdf85f9ccf688e79 : SUCCESS in 6m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d3ce4f606c6466abf91d4ce191b4e8e : SUCCESS in 52m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bb0cd94db1a4edb927da8b171505663 : FAILURE in 1h 20m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7201ff1e64e84c76a8d46630ff772961 : SUCCESS in 1h 09m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9facc219f8b54f60b711df32ff4e9f6c : SUCCESS in 39m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b106ae2552f9463c89c45678c8895c5b : SUCCESS in 1h 08m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/676ea5267cf449f8ac765796a95e33fa : SUCCESS in 1h 27m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/734a236df54a43c3be62b373481d4153 : SUCCESS in 1h 32m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e426fcf42e8c4408b9b554d93813dee3 : SUCCESS in 1h 03m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/940162e3ef2d4502b7453b1974ca5a46 : SUCCESS in 1h 25m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a478ce700db94b5393afc9a7f7929472 : SUCCESS in 1h 23m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7db168f8bba1492f91e0ecdc3017ef48 : TIMED_OUT in 2h 36m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/90c1b514e8f540528a19f538659fb0c6 : SUCCESS in 1h 06m 56s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"ff73df3894805ec1c8f32a190c841902c86ddb23","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 16:12:52.000000000","message":"Patch Set 31:\n\nrecheck\n\nudp again...","accounts_in_message":[],"_revision_number":31},{"id":"b6413afd27ebda38ae08c4440ecf9ccb138c6bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 18:41:15.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22737ab349df437d90d351d7a2264a4b : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f816416665d341c8be8bb3abbd1360ef : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12106b8a7e9846b48dba9689ab863103 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e438f81475f944d29dfb46f75348746d : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61b279c137854543b862da1da5dfe86e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2bc33c8e28c4c6596e480b4bf972068 : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bdd6a36a11049b0aac627c1b8f92055 : SUCCESS in 4m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f16a9a42b5a141d4aba33ed89c1ea1f1 : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0df104caf8c74933bd35e3df081ef6ea : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d4a0c737f7b4e88a6014d6ebaa5d86f : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e679b57d7bdc48beba4d3c57450e6669 : SUCCESS in 45m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d00a14f2ad5c4a8ba400795ce2e4bdf2 : SUCCESS in 58m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59bce85a53194a0099db3f59c066c11e : SUCCESS in 42m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cada106622b04a089098d7204c45972b : SUCCESS in 38m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4deebc8edcd249d2995d66040867e872 : SUCCESS in 1h 01m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f9383dbde47411aabb003afdf27ea99 : SUCCESS in 1h 46m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fd960a42a7a84617a45b39187591d6c1 : SUCCESS in 1h 11m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84882f9f2314473c90365c6c8c194040 : SUCCESS in 1h 01m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d9980a19034d4c2b9f0f70a7b96f1db0 : SUCCESS in 1h 45m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a69741cbfb74a79b3408c57d75c4898 : SUCCESS in 57m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d2cfdf4021bb471aa8b6579a86c8d50f : SUCCESS in 1h 00m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8cb319851c5c42668d3ceadfe4f94204 : FAILURE in 37m 10s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"42d64ff25539fb4c9c34828b7ac3838cdd88e9af","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 17:05:22.000000000","message":"Patch Set 31: Code-Review+1","accounts_in_message":[],"_revision_number":31},{"id":"87768a2bd4ccc8b740cff31c65e1664fa3df3ad0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-11 17:04:25.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"557844396d62a98c36492d3fb88c51835e6a768d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:15:23.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9b13b3e552344bba6e194109e58baf7 : SUCCESS in 12m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03eb5630290c430eb5b8c3599c2e0462 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f56c427e8f4f96ab246f1ed940632a : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69abe48404684c0ab9b8a81560165cf2 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dab13bfe3fe41b8aaa41bd4c76cbb74 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea5ecd46a544fec86b083916c13d007 : SUCCESS in 20m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a24cb0cc90bb4be38469bc70d45e5056 : SUCCESS in 5m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4986c1a442884ed8812dd15a95e41341 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/caa4874e681647ea92a5831e98ca41b0 : SUCCESS in 9m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ffbbeb757e104198981ccfef0fac0ebf : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac9db11c676f49b3aac0f22023d4661f : SUCCESS in 36m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a282fa541857446ab7e302d2f990fd88 : SUCCESS in 1h 27m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a0c60e4087c144a58643a40689f5ced7 : SUCCESS in 33m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d0d78ebbf9ed4ea0a57068a8703a31b5 : SUCCESS in 32m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/947e6239a0b340d482ee182b48bca630 : SUCCESS in 42m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0a4e58a3bd944ed09907b972bb2a98e5 : SUCCESS in 52m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e00c0e18cdad43faa0f0e6544c5c1175 : SUCCESS in 2h 02m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/111d4ebefd7e497d819e34ddf32abb1e : SUCCESS in 2h 07m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/972528ec5e194e0eb61834edb87bbdd6 : SUCCESS in 1h 54m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4475d0e987c14b4d992d053bc3f7abfe : SUCCESS in 1h 46m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5f670aad02f34ff190485bc64516081a : SUCCESS in 2h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/241f4ebbf8c8452fa5782c343a0c2922 : SUCCESS in 1h 53m 10s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"dd9a8cd20645a329278c9f2b6af655ab5140cb8e","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-10-08 09:01:24.000000000","message":"Patch Set 32:\n\nHi I create a story [1] in storyboard to track this great multi-arch effort you have.\n\nFeel free to reference the task 41077 in commit message if you like\n\n[1] https://storyboard.openstack.org/#!/story/2008234","accounts_in_message":[],"_revision_number":32},{"id":"bd8b8e6b6706fb2d6286abf0e6990883f4025982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-06 09:39:54.000000000","message":"Patch Set 33: Patch Set 32 was rebased","accounts_in_message":[],"_revision_number":33},{"id":"7d273ba789763d687b8bac179ee9863d3227c840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 11:49:45.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/083a821ca7f14ebdb5440ce157edc763 : SUCCESS in 13m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f9db83735bc4fa5864b3b5c3b679a2f : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d657e53b36484c8c7b83edfcc4330e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5943ae7106444c3bfd64e5ccc142985 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e2999651beb42ff8e4f1cc320f93ad6 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eba7438374f4969b863d4e61c57f382 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd7a050a0be4fafabb5d26cc5243363 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7912351ac8c48adbd01b02ded088782 : SUCCESS in 4m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/717a0340a97c4a04a7997801cdb2fd74 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8094af0019394357aa04116159d9c790 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/02907b8feedd4c618da43507eaf2b42f : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da32fa6203b344389aa7e1f189b338ee : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c220c5d36c0a4c75b991e4e31b3b25d6 : SUCCESS in 1h 22m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/884f5c8d11344aa9b89e00fcc64b6c48 : SUCCESS in 2h 05m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/35d6ee34f6654fb4a5a309d238f039d6 : SUCCESS in 39m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8a643560ab0141e2a5463d6c361d8e0f : SUCCESS in 34m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6a3b1e1fc17455c96590717383d1281 : SUCCESS in 1h 00m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a24abd29034b4449b6e5389d5fd4b6f4 : SUCCESS in 46m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/03e6600df9314f90a6053538c44771ae : SUCCESS in 1h 40m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a52f245ae2ba41a88f5054187536ed93 : FAILURE in 2h 00m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/47de6403bf234ce59ba81bd4445e1b92 : SUCCESS in 1h 18m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd1e4b4b6141453b9f6e2bb7fbf784d9 : SUCCESS in 1h 25m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d9c4ce81678f44d8999644ce385d8ce8 : FAILURE in 1h 38m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c9b1deec50ec4e9a839edb6305a77c0a : FAILURE in 24m 18s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"c44f9eb9d413e3181ce71a7b86d56e3c5fa31fbd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-06 22:58:15.000000000","message":"Patch Set 33: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":33},{"id":"fd39ee1cfcd11ec624e41a4097133e9c53ef0360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-09 09:33:52.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"1d0ad643335ccdd852050c31a94142e1466bf2bd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-09 09:34:00.000000000","message":"Patch Set 33:\n\n(3 comments)","accounts_in_message":[],"_revision_number":33},{"id":"c5fd458e258df3c7decefaab1b1a997daedeef73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:14:04.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce42e8aa61b4372991b0e11686dd12f : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbe76d5981bb41d2a3c06895f8b05a2d : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/147cd3d2eef94db2a90410f0df1c5dc4 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88d2695bf3504563aad2383c4a9ab624 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92d75fa00a744303a8f5f23c4150e4c3 : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f562fb335f443aba57596310c0aa233 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf96033bb3c4c1a9ad0ccae0faef109 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e156aebcabae4c61bdac4b82eccc0355 : SUCCESS in 7m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a3c460a6a6b9495283b2ead20733ff3e : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d80c6d1fc9114a47bd5bd7f10d1e5f83 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a9350466f4e24397b3cc8628a0df729a : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/41b57a0638c249ef8616133382f75f4d : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b5e27257dc5487cacbb224560953735 : SUCCESS in 1h 15m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/222be000feba4b0483fe8987e3b75ede : SUCCESS in 2h 05m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bad78cc69f641288f30db6493c60723 : SUCCESS in 38m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ebf8371e3af47b8bcb70e7994cd5b36 : SUCCESS in 36m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e34415b0e7148cb9a3771ea644bc475 : SUCCESS in 40m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8617942754c40fdbffb94d79de4eb7b : SUCCESS in 59m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7fb1505e60b746f8abd8ccb9db7789f9 : SUCCESS in 2h 09m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/12119ed096f04486bbcde6ec373c4e84 : FAILURE in 2h 31m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/992036b46eff460a8699395aa815efa5 : SUCCESS in 1h 38m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/da3140b261fe4761b8e80ad430666e5e : SUCCESS in 2h 05m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/36926d0d6ccf4e0b86ec90b8e8a8af30 : FAILURE in 1h 36m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/72a78b4474ea483ea0970cd81d4e16f3 : FAILURE in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"ab6ddfcf3eedfd4b4f6bbc72b21608c8786fd9d7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-09 19:20:34.000000000","message":"Patch Set 34: Code-Review-1\n\n(1 comment)\n\nThe image size hack is troubling. This is a \"special\" case thing that will cause headaches.\nCan we at least point to a bug report against DIB for the incorrect image size being created?","accounts_in_message":[],"_revision_number":34},{"id":"6e39cbe9b1415058c57122531cc81c6bf68760bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-10 15:19:47.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"860fc2602c9f0efcfcd31b4d21f113b9a9cb1d64","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-10 15:20:45.000000000","message":"Patch Set 35: Workflow-1","accounts_in_message":[],"_revision_number":35},{"id":"8b671e151230856627f3ccecf6144279494a17a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 17:53:48.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/919e070419244b4f95f3af132d1d8674 : SUCCESS in 11m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc97a05cde6749f79535fd5c93c79c09 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f625fb0df6b743cb86dd9d7a372cf9e2 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/892df0246e0a492988c51a498a60db46 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3cafda8379c4fb39c07479d3d77bf30 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1cbe98e67f944a5a8f08c903db06e4b : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa7703a5e63499dae679a3722c62354 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9377104758f4ba5a9fb14b9dc325e59 : SUCCESS in 5m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ed3b59b0f6534280bfa04d550f16af23 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f0497a892ecd49cbb65f283137876088 : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c7a634dda81e4a2dbd3f5fe1340f2f05 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39f4ad9fd5614a8e8149a5bdf203f9b7 : SUCCESS in 10m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e59ed352ec14304aaa8c43a92d99108 : SUCCESS in 1h 21m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6722e4e59671487fbb0a52d00aaed969 : FAILURE in 1h 48m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b874832eb16d42e098fad4648b10b430 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/98e20d6e562c4361ae6deeb2ef313c3f : SUCCESS in 44m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73c20e82c6014a8194282ade3024a5db : SUCCESS in 1h 13m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0f4b6e83157a4e9ca94f7ba633abf35f : SUCCESS in 1h 02m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de44e303712d420ca54fd58952de4750 : SUCCESS in 1h 43m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6070abc7929435cbccdc084703d5a91 : FAILURE in 49m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3a7a82f162fa4a0bbcf242016b45e0a8 : SUCCESS in 1h 52m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c6de7766623a41908866dc17c8ab8031 : SUCCESS in 1h 48m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b68fc13bda584cf2826c6165ae94c60e : SUCCESS in 2h 19m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1a9072f4ef0f42f78d772cbfbdb2400c : FAILURE in 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"680e0f36efe7ccdb0b09848e5204a97b8adead88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-10 17:55:25.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"aa9ea8e4e10c1e156188954490af696196162c1e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-10 17:55:53.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"3ff82d3119ae282fbbe3b8cc2bd10dc967487119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 20:23:49.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59081b4667654718a6320d2b108b7674 : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8475076415d441f38bd2af274745a74c : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8722c270d0344edbb7175d0f613c460 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87f4130f2094489087a6f50cd9e01303 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa3bc3116124052ab6cc9b0788d5340 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/120f2585af304f83946849198967e8dc : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb33e342deb6458eae622318072f79cf : SUCCESS in 18m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9e71beccab04ffb93f85d0260b8bc38 : SUCCESS in 7m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dba2dcebcecb4a259273c7cf1555e38b : SUCCESS in 8m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/383a8418aad74c0ca2ce9cb9a5b69dd8 : SUCCESS in 10m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2a5383c357244cea8f3afaee6cb6a2aa : SUCCESS in 8m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7bb9e6f376f04d0db07fc6e67c231f12 : SUCCESS in 9m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fbfc76b7f7e45dbb6e959f97b4471d0 : SUCCESS in 1h 21m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31301357fed04253ba4f4babd30d009d : SUCCESS in 1h 56m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c82232e1028426db99e628eef397961 : SUCCESS in 48m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bf6a27bd7d0b4f1c868286976d783274 : SUCCESS in 43m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5db05a99f45d42e19c135ae676794e02 : SUCCESS in 41m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7327a50d1f9a4e09877732617e7f913d : SUCCESS in 1h 08m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b930b0aa20ea4f1b918e49c499f9a7a9 : SUCCESS in 1h 42m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/159711d0ed3844099939ee61cc73984f : FAILURE in 52m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3cc351d1a363499b9414dd789b003557 : SUCCESS in 2h 11m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d3fa7f888498465d91f35ed22220c21b : SUCCESS in 2h 24m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f7ee790367854838b42cc953021cd447 : SUCCESS in 2h 14m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/de6c911fc5bc4f1b95403fefc760df81 : FAILURE in 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"f3bad8cbc309e18ab7254069f8cb410d2a614f5d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-10 21:23:21.000000000","message":"Patch Set 36: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":36},{"id":"bf95a97bbf516678ac5291c93faf4a649a15c796","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-11 08:30:16.000000000","message":"Patch Set 36: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":36},{"id":"b8388c4f6f17c44947817f16655a0da8b67dc6aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 08:30:37.000000000","message":"Patch Set 36: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":36},{"id":"05be5d85835ae206964ea491bcd9dd2721cb4680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:29:05.000000000","message":"Patch Set 36: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d951a795920549489e20614b63c53316 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597f3b5463214115ae6669289f5c4cab : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd6c5d347b5846f4889d0f9dbf1a61fd : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c831bfad1df24da58c56fe8f419e8055 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6ca7dba77c4318b7b0c350b4f5961e : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed047f5b4a0e45c1881a532fed6d322c : SUCCESS in 5m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c300e69ab754583802c5159c9622d99 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22a7c9c83f5c4652bc052a8bb0f5d942 : SUCCESS in 10m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10ccb59b11c345f5863e15b890ee9c51 : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5a60d965b174a60a627fe7bc17dcdf0 : SUCCESS in 1h 56m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa92956ae3ec418a9f4536150236c267 : SUCCESS in 44m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8bd4db063b8140b3b0dce596f5121837 : SUCCESS in 42m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7254811e711f4780a3272f29a5864508 : SUCCESS in 45m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8854eb619d86496396cb3ab8a18545f7 : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":36},{"id":"f78f669d5444ed5d4b6f95fbd05caa77de5dbd14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:29:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":36},{"id":"bfdd1b180ad92736f1a856ec3ff5728d6f69cb55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:30:59.000000000","message":"Patch Set 36:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1024ed21d1b491f96cfa69c4a1f4a90 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4231335deb54604a1b403d4d5a5b9c3 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":36}],"current_revision_number":36,"current_revision":"e99d7b9389d340b73e9b1094c4a0b7e7997a5344","revisions":{"e302db26142f1bc783aa8cbda66e9870a8aa1c23":{"kind":"REWORK","_number":1,"created":"2020-06-25 19:54:35.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/1"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:54:21.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e302db26142f1bc783aa8cbda66e9870a8aa1c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e302db26142f1bc783aa8cbda66e9870a8aa1c23"}]},"branch":"refs/heads/master"},"0ca22279347ef127e6eeba9e5794252f1dee28e1":{"kind":"REWORK","_number":2,"created":"2020-06-25 20:44:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/2"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 20:43:51.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ca22279347ef127e6eeba9e5794252f1dee28e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ca22279347ef127e6eeba9e5794252f1dee28e1"}]},"branch":"refs/heads/master"},"f5de604b73113cddb0592f8b91300821e7f5881c":{"kind":"REWORK","_number":3,"created":"2020-06-26 06:22:44.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/3"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 06:22:34.000000000","tz":0},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5de604b73113cddb0592f8b91300821e7f5881c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5de604b73113cddb0592f8b91300821e7f5881c"}]},"branch":"refs/heads/master"},"97b914f559c985e1bb65b66b81407e5095a09f9b":{"kind":"REWORK","_number":4,"created":"2020-06-26 08:11:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/4"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 08:11:06.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97b914f559c985e1bb65b66b81407e5095a09f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97b914f559c985e1bb65b66b81407e5095a09f9b"}]},"branch":"refs/heads/master"},"4235222193e194306f20ac5a27900e035732fd88":{"kind":"REWORK","_number":5,"created":"2020-06-26 09:24:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/5"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 09:24:28.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4235222193e194306f20ac5a27900e035732fd88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4235222193e194306f20ac5a27900e035732fd88"}]},"branch":"refs/heads/master"},"7f748a9b051ede7ba2906e0e048d81935535c55e":{"kind":"REWORK","_number":6,"created":"2020-06-26 09:49:59.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/6"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 09:49:26.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f748a9b051ede7ba2906e0e048d81935535c55e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f748a9b051ede7ba2906e0e048d81935535c55e"}]},"branch":"refs/heads/master"},"200a097db60814071ceb2f9a5ccfa426bb5ce20c":{"kind":"REWORK","_number":7,"created":"2020-06-26 11:22:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/7"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 11:21:51.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/200a097db60814071ceb2f9a5ccfa426bb5ce20c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/200a097db60814071ceb2f9a5ccfa426bb5ce20c"}]},"branch":"refs/heads/master"},"e0d10329137c3863ced8cc0bb3b8e9a16f6c5292":{"kind":"REWORK","_number":8,"created":"2020-06-26 12:33:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/8"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 12:33:40.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0d10329137c3863ced8cc0bb3b8e9a16f6c5292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0d10329137c3863ced8cc0bb3b8e9a16f6c5292"}]},"branch":"refs/heads/master"},"84051b4bfadd16e83c9474016241980a9800ece4":{"kind":"REWORK","_number":9,"created":"2020-06-26 13:46:50.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/9"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 13:46:35.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84051b4bfadd16e83c9474016241980a9800ece4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84051b4bfadd16e83c9474016241980a9800ece4"}]},"branch":"refs/heads/master"},"4778cae689a7a51ab43fc4e762a61a6cc37d420a":{"kind":"REWORK","_number":10,"created":"2020-06-26 15:26:29.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/10"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 15:26:17.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4778cae689a7a51ab43fc4e762a61a6cc37d420a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4778cae689a7a51ab43fc4e762a61a6cc37d420a"}]},"branch":"refs/heads/master"},"92229e0eb882df09e1536c157fd478e11831a5b6":{"kind":"REWORK","_number":11,"created":"2020-06-26 15:30:27.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/11"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 15:30:16.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92229e0eb882df09e1536c157fd478e11831a5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92229e0eb882df09e1536c157fd478e11831a5b6"}]},"branch":"refs/heads/master"},"6de5cd4318f17a1f82d80c08c3985a824f7d3e5d":{"kind":"REWORK","_number":12,"created":"2020-06-26 16:37:15.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/12"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 16:26:48.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6de5cd4318f17a1f82d80c08c3985a824f7d3e5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6de5cd4318f17a1f82d80c08c3985a824f7d3e5d"}]},"branch":"refs/heads/master"},"0c44f779f2e8ffedc5926d8d46aa97f7f142061f":{"kind":"REWORK","_number":13,"created":"2020-06-26 19:43:48.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/13"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 19:42:35.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c44f779f2e8ffedc5926d8d46aa97f7f142061f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c44f779f2e8ffedc5926d8d46aa97f7f142061f"}]},"branch":"refs/heads/master"},"4f4b77969bbd41ba809953403e64e7869138cc2c":{"kind":"REWORK","_number":14,"created":"2020-06-26 20:52:20.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/14"}}},"commit":{"parents":[{"commit":"89123c0fc117e095c44ed21e360394974d3e15a5","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89123c0fc117e095c44ed21e360394974d3e15a5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-26 20:52:06.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f4b77969bbd41ba809953403e64e7869138cc2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f4b77969bbd41ba809953403e64e7869138cc2c"}]},"branch":"refs/heads/master"},"5dfa812471751ba0ebbda826e72e68d1d038783d":{"kind":"REWORK","_number":15,"created":"2020-08-07 11:49:24.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/15"}}},"commit":{"parents":[{"commit":"d78e992a9ed28f5dca36ae38efc5352b5c11cf32","subject":"Merge \"Fix haproxy user flow log format substitution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d78e992a9ed28f5dca36ae38efc5352b5c11cf32"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-07 11:46:14.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dfa812471751ba0ebbda826e72e68d1d038783d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dfa812471751ba0ebbda826e72e68d1d038783d"}]},"branch":"refs/heads/master"},"b4b90aec4b099add60092d3be15c477182cfa12f":{"kind":"REWORK","_number":16,"created":"2020-08-07 13:24:29.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/16"}}},"commit":{"parents":[{"commit":"d78e992a9ed28f5dca36ae38efc5352b5c11cf32","subject":"Merge \"Fix haproxy user flow log format substitution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d78e992a9ed28f5dca36ae38efc5352b5c11cf32"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-07 13:24:17.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b90aec4b099add60092d3be15c477182cfa12f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b90aec4b099add60092d3be15c477182cfa12f"}]},"branch":"refs/heads/master"},"7232581c7214638f4c95e6f72b165d1e3c375824":{"kind":"REWORK","_number":17,"created":"2020-08-22 15:39:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/17"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-22 15:39:21.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7232581c7214638f4c95e6f72b165d1e3c375824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7232581c7214638f4c95e6f72b165d1e3c375824"}]},"branch":"refs/heads/master"},"8545b759acac4f00c44b50722ba9ccdd9b28290a":{"kind":"REWORK","_number":18,"created":"2020-08-22 20:10:00.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/18"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-22 20:09:48.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8545b759acac4f00c44b50722ba9ccdd9b28290a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8545b759acac4f00c44b50722ba9ccdd9b28290a"}]},"branch":"refs/heads/master"},"236151246886640600bdfa901a1be806199a2483":{"kind":"REWORK","_number":19,"created":"2020-08-22 21:37:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/19"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-22 21:37:44.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/236151246886640600bdfa901a1be806199a2483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/236151246886640600bdfa901a1be806199a2483"}]},"branch":"refs/heads/master"},"e0ac607fca668d7ff344d1304e444cf101fb3fda":{"kind":"REWORK","_number":20,"created":"2020-08-22 23:11:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/20"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-22 23:10:55.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0ac607fca668d7ff344d1304e444cf101fb3fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0ac607fca668d7ff344d1304e444cf101fb3fda"}]},"branch":"refs/heads/master"},"7ba47b36f6ea2b06e3de9645776df5e6d75d8ebf":{"kind":"REWORK","_number":21,"created":"2020-08-23 08:59:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/21"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-23 08:58:03.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ba47b36f6ea2b06e3de9645776df5e6d75d8ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ba47b36f6ea2b06e3de9645776df5e6d75d8ebf"}]},"branch":"refs/heads/master"},"1a97385b712f08381b74d774d33ab6025085468c":{"kind":"REWORK","_number":22,"created":"2020-08-23 10:10:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/22"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-23 10:10:25.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a97385b712f08381b74d774d33ab6025085468c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a97385b712f08381b74d774d33ab6025085468c"}]},"branch":"refs/heads/master"},"39dac2cae96672404613df785e969830244ecefa":{"kind":"REWORK","_number":23,"created":"2020-08-23 11:55:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/23"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-23 11:55:13.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39dac2cae96672404613df785e969830244ecefa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39dac2cae96672404613df785e969830244ecefa"}]},"branch":"refs/heads/master"},"94ee3e4fae50929cc03134f47943bf905c3ed0c9":{"kind":"REWORK","_number":24,"created":"2020-08-23 12:57:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/24"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-23 12:57:45.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94ee3e4fae50929cc03134f47943bf905c3ed0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94ee3e4fae50929cc03134f47943bf905c3ed0c9"}]},"branch":"refs/heads/master"},"2e8a7ac725fd53ff97d156c997418b483eb7753d":{"kind":"REWORK","_number":25,"created":"2020-08-23 20:14:22.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/25"}}},"commit":{"parents":[{"commit":"bf3d5372b9fc670ecd08339fa989c9b738ad8d69","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf3d5372b9fc670ecd08339fa989c9b738ad8d69"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-23 20:14:01.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nIt will fail *at least* to run the x86-compiled test_server\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e8a7ac725fd53ff97d156c997418b483eb7753d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e8a7ac725fd53ff97d156c997418b483eb7753d"}]},"branch":"refs/heads/master"},"7d631a82c49bf04d7d67b6981fe4592788dc7963":{"kind":"REWORK","_number":26,"created":"2020-08-24 07:52:15.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/26"}}},"commit":{"parents":[{"commit":"13faf288e4c1044facc579c8e40e46c524bbe3f8","subject":"Merge \"Fix memory consumption issues with default connection_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13faf288e4c1044facc579c8e40e46c524bbe3f8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 07:52:01.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nDepends-On: https://review.opendev.org/#/c/747629\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d631a82c49bf04d7d67b6981fe4592788dc7963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d631a82c49bf04d7d67b6981fe4592788dc7963"}]},"branch":"refs/heads/master"},"91c0238b5231ec36fc747fa2fc7db9835bef884b":{"kind":"REWORK","_number":27,"created":"2020-08-24 09:45:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/27"}}},"commit":{"parents":[{"commit":"13faf288e4c1044facc579c8e40e46c524bbe3f8","subject":"Merge \"Fix memory consumption issues with default connection_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13faf288e4c1044facc579c8e40e46c524bbe3f8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 09:44:58.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nDepends-On: https://review.opendev.org/#/c/747629\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91c0238b5231ec36fc747fa2fc7db9835bef884b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91c0238b5231ec36fc747fa2fc7db9835bef884b"}]},"branch":"refs/heads/master"},"1157f878b9bf8a088375aaa2f511f1559d20e409":{"kind":"REWORK","_number":28,"created":"2020-08-24 12:47:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/28"}}},"commit":{"parents":[{"commit":"13faf288e4c1044facc579c8e40e46c524bbe3f8","subject":"Merge \"Fix memory consumption issues with default connection_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13faf288e4c1044facc579c8e40e46c524bbe3f8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 12:47:13.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nDepends-On: https://review.opendev.org/#/c/747629\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1157f878b9bf8a088375aaa2f511f1559d20e409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1157f878b9bf8a088375aaa2f511f1559d20e409"}]},"branch":"refs/heads/master"},"a5943bed1d4893f5121a28bb6f5c7bb66009c27c":{"kind":"REWORK","_number":29,"created":"2020-08-24 17:47:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/29"}}},"commit":{"parents":[{"commit":"13faf288e4c1044facc579c8e40e46c524bbe3f8","subject":"Merge \"Fix memory consumption issues with default connection_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13faf288e4c1044facc579c8e40e46c524bbe3f8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 17:47:15.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nDepends-On: https://review.opendev.org/#/c/747629\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5943bed1d4893f5121a28bb6f5c7bb66009c27c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5943bed1d4893f5121a28bb6f5c7bb66009c27c"}]},"branch":"refs/heads/master"},"fb49cbfca318f49991292cd457a92d77ce90332d":{"kind":"REWORK","_number":30,"created":"2020-08-24 21:26:33.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/30"}}},"commit":{"parents":[{"commit":"13faf288e4c1044facc579c8e40e46c524bbe3f8","subject":"Merge \"Fix memory consumption issues with default connection_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13faf288e4c1044facc579c8e40e46c524bbe3f8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 21:26:19.000000000","tz":120},"subject":"DNM: ARM64 support","message":"DNM: ARM64 support\n\nDepends-On: https://review.opendev.org/#/c/747629\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb49cbfca318f49991292cd457a92d77ce90332d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb49cbfca318f49991292cd457a92d77ce90332d"}]},"branch":"refs/heads/master"},"95738635edef35e6f8aeba6afb5a4c67d003ac08":{"kind":"REWORK","_number":31,"created":"2020-08-25 19:19:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/31"}}},"commit":{"parents":[{"commit":"703a07df2150cd33be1dbc6e8013e78b3b71884c","subject":"Merge \"Ensure xtrace is enabled in devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/703a07df2150cd33be1dbc6e8013e78b3b71884c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 19:13:15.000000000","tz":120},"subject":"Add aarch64/amd64 support to amphora and devstack","message":"Add aarch64/amd64 support to amphora and devstack\n\nThe disk image create tool can now build aarch64/arm64 amphora images.\nThe devstack plugin will facilitate image builds and upload to Glance\nfor this CPU architecture.\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95738635edef35e6f8aeba6afb5a4c67d003ac08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95738635edef35e6f8aeba6afb5a4c67d003ac08"}]},"branch":"refs/heads/master"},"9a84ebcb660bc5ed9acab1387e92c7f4f50acca3":{"kind":"REWORK","_number":32,"created":"2020-09-11 17:04:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/96/738096/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/32"}}},"commit":{"parents":[{"commit":"703a07df2150cd33be1dbc6e8013e78b3b71884c","subject":"Merge \"Ensure xtrace is enabled in devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/703a07df2150cd33be1dbc6e8013e78b3b71884c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-11 17:04:24.000000000","tz":-420},"subject":"Add aarch64/amd64 support to amphora and devstack","message":"Add aarch64/amd64 support to amphora and devstack\n\nThe disk image create tool can now build aarch64/arm64 amphora images.\nThe devstack plugin will facilitate image builds and upload to Glance\nfor this CPU architecture.\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a84ebcb660bc5ed9acab1387e92c7f4f50acca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a84ebcb660bc5ed9acab1387e92c7f4f50acca3"}]},"branch":"refs/heads/master"},"281300033b9f13933af9f374d74a9a92766f4dfd":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2021-03-06 09:39:54.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/33"}}},"commit":{"parents":[{"commit":"8d72f571e1fba1ee8db0df5c0d998cd03d73f391","subject":"Merge \"Add HTTP/2 to default lists of ALPN protocols\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d72f571e1fba1ee8db0df5c0d998cd03d73f391"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Gonçalves","email":"cgoncalves@redhat.com","date":"2021-03-06 09:39:54.000000000","tz":0},"subject":"Add aarch64/amd64 support to amphora and devstack","message":"Add aarch64/amd64 support to amphora and devstack\n\nThe disk image create tool can now build aarch64/arm64 amphora images.\nThe devstack plugin will facilitate image builds and upload to Glance\nfor this CPU architecture.\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/281300033b9f13933af9f374d74a9a92766f4dfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/281300033b9f13933af9f374d74a9a92766f4dfd"}]},"branch":"refs/heads/master","description":"Rebase"},"4109c5eaa79b4eb71c50c4701df87b81f6a4d802":{"kind":"REWORK","_number":34,"created":"2021-03-09 09:33:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/34"}}},"commit":{"parents":[{"commit":"ef05f5b571c13e027108058cf6651498c789416a","subject":"Merge \"Small fix in loadbalancer POST api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef05f5b571c13e027108058cf6651498c789416a"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-03-09 09:33:43.000000000","tz":60},"subject":"Add aarch64/amd64 support to amphora and devstack","message":"Add aarch64/amd64 support to amphora and devstack\n\nThe disk image create tool can now build aarch64/arm64 amphora images.\nThe devstack plugin will facilitate image builds and upload to Glance\nfor this CPU architecture.\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4109c5eaa79b4eb71c50c4701df87b81f6a4d802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4109c5eaa79b4eb71c50c4701df87b81f6a4d802"}]},"branch":"refs/heads/master"},"16672d37bf85ba69674d03ba00c80ab4f59a127e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":35,"created":"2021-03-10 15:19:47.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/35"}}},"commit":{"parents":[{"commit":"651f095a818c9a30441ed8df9a39cf726b6e3f5d","subject":"Merge \"Remove Neutron SDN-specific code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/651f095a818c9a30441ed8df9a39cf726b6e3f5d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-03-10 15:19:31.000000000","tz":60},"subject":"Add aarch64/amd64 support to amphora and devstack","message":"Add aarch64/amd64 support to amphora and devstack\n\nThe disk image create tool can now build aarch64/arm64 amphora images.\nThe devstack plugin will facilitate image builds and upload to Glance\nfor this CPU architecture.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/779806\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16672d37bf85ba69674d03ba00c80ab4f59a127e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16672d37bf85ba69674d03ba00c80ab4f59a127e"}]},"branch":"refs/heads/master"},"e99d7b9389d340b73e9b1094c4a0b7e7997a5344":{"kind":"REWORK","_number":36,"created":"2021-03-10 17:55:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/738096/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/738096/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/738096/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/738096/36"}}},"commit":{"parents":[{"commit":"651f095a818c9a30441ed8df9a39cf726b6e3f5d","subject":"Merge \"Remove Neutron SDN-specific code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/651f095a818c9a30441ed8df9a39cf726b6e3f5d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-25 19:49:40.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-03-10 17:54:50.000000000","tz":60},"subject":"Add aarch64/amd64 support to amphora and devstack","message":"Add aarch64/amd64 support to amphora and devstack\n\nThe disk image create tool can now build aarch64/arm64 amphora images.\nThe devstack plugin will facilitate image builds and upload to Glance\nfor this CPU architecture.\n\nChange-Id: I1cebd8a3da58dc56ebbfac22f7802ab7f52585e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e99d7b9389d340b73e9b1094c4a0b7e7997a5344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e99d7b9389d340b73e9b1094c4a0b7e7997a5344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
