)]}'
{"id":"openstack%2Foctavia~738264","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I7b8f7c7d653b37395f9a660be67f954a3a6f26d9","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I7b8f7c7d653b37395f9a660be67f954a3a6f26d9","subject":"Fix neutron subnet lookup ignoring endpoint_type","status":"MERGED","created":"2020-06-26 19:24:32.000000000","updated":"2020-06-27 13:52:46.000000000","submitted":"2020-06-27 13:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738264-1593265881130-995c8465","meta_rev_id":"df1375c969de14d834119664398bf4a68247d2ca","_number":738264,"virtual_id_number":738264,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-27 13:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-06-26 19:27:54.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-06-27 11:59:57.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-06-27 11:59:57.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 19:27:54.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-27 11:59:57.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-27 13:51:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81b05cd16cc302d20e77efc0cea3a027ce890633","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 19:24:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"686c21ffa0f17499596ca72962eeaad262f7ad00","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 19:24:45.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 2a18b6cc996e4c64d650c516071c57358681af3b","accounts_in_message":[],"_revision_number":1},{"id":"76e866c615599607d81a0549107524441dbc3566","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 19:27:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c75bb5a9085b3a3a62d9f5ee225b6cf3f4830c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:21:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dcc1495777444f48ab2d98eeab6522c : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8370459e1a9e4f8683655685ab87d825 : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c6a8b021ca46bab53904496e28f4e7 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d94a703de0f842d3a1ac04569bd853f1 : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af31eb72254f4bff95f25a7ce2a2058f : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d75f36ddc14b8d8a13afe063f8869c : SUCCESS in 12m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ff2f8d6c90214dabb46b6033d3dc2b67 : SUCCESS in 8m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c2e41663f21749ea83c5d5385170fab5 : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2cb2cceb237c41269547edaa4b710348 : SUCCESS in 8m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ca9585cd1ef44c4a29e9413ba5d225c : SUCCESS in 41m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/849e8a8aaf1745b9bc5eabd1baa59a54 : SUCCESS in 1h 22m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/247d583c00744c6a9b793d1bb4c1da1f : SUCCESS in 1h 02m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/029ce47bbb9648dba5413e811b92c35e : SUCCESS in 1h 01m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2adfde95d92f4845907bc5d516f1d1cc : SUCCESS in 1h 04m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8624ec6f50ed44fe8d813bc7589bdb01 : SUCCESS in 1h 50m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff57eb16e5ef43da8bb8087de9c5ed38 : FAILURE in 1h 05m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a333cf6868d9436a8e683021e97ff796 : FAILURE in 44m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8b4215ef523d41578cb8cca37ec13d59 : FAILURE in 1h 13m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afb7c2e62b29dc6530f606aa27c157682ee6fc2c","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-06-27 11:59:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a459df18b038064e8285a37e3dca3aa8bafd0777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 12:00:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ebb1c73e5cb243494042b512e5b55a1360f4701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 13:51:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9233f35f203047279a8819ed59ecde46 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f266591276f4282b2879308030fb536 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ce4256827cc41fba00436cdc31bb890 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11f608db62ca4d00bf8ae733c4d687e9 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afdfc99ac0b8401bbb223c8e00b77e1d : SUCCESS in 10m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb2c78314b584f0db6b126a223253521 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae268134990548cdbef05a3454a517e6 : SUCCESS in 44m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1eb6b41c7af049ee8be7f9f39259e0d8 : SUCCESS in 1h 45m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6777c3fe7e24a0f8a2f5885fc016950 : SUCCESS in 56m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3ce094feab74433bb2bec7693c694f0a : SUCCESS in 53m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55cbe74f38f44a3195e2c028cf5439e2 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"87d6776efce803134d46098b6af90c41f105ea2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 13:51:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df1375c969de14d834119664398bf4a68247d2ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 13:52:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65490811c1ec44b290894a054bb73ea2 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4bc92ebf1a5610cb340299d6230a134f202bfbb","revisions":{"b4bc92ebf1a5610cb340299d6230a134f202bfbb":{"kind":"REWORK","_number":1,"created":"2020-06-26 19:24:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/738264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/738264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/738264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/738264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/738264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/738264/1"}}},"commit":{"parents":[{"commit":"bd2987ba234494aacff6268a14dd2e9ececa0108","subject":"Fix batch member create for v1 amphora driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd2987ba234494aacff6268a14dd2e9ececa0108"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-25 17:05:54.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-26 19:24:32.000000000","tz":0},"subject":"Fix neutron subnet lookup ignoring endpoint_type","message":"Fix neutron subnet lookup ignoring endpoint_type\n\nThe recent patch[1] that adds a neutron subnet lookup using the user\ncontext is not honoring the interface/endpoint_type or region_name\nsettings for neutron in the octavia configuration file.\nThis is causing problems for deployments that use the \"internal\"\nendpoint for neutron and the current code will always return\nthe \"public\" endpoint.\nThis patch corrects this problem by including those filter\nparameters when the neutron endpoint is looked up in keystone.\n\n[1] https://review.opendev.org/726042\n\nChange-Id: I7b8f7c7d653b37395f9a660be67f954a3a6f26d9\nStory: 2007863\nTask: 40173\n(cherry picked from commit 40731e572f2b503a3ccc603642a3721a5ce1cf7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4bc92ebf1a5610cb340299d6230a134f202bfbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4bc92ebf1a5610cb340299d6230a134f202bfbb"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
