)]}'
{"id":"openstack%2Foctavia~738265","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I7b8f7c7d653b37395f9a660be67f954a3a6f26d9","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I7b8f7c7d653b37395f9a660be67f954a3a6f26d9","subject":"Fix neutron subnet lookup ignoring endpoint_type","status":"MERGED","created":"2020-06-26 19:24:45.000000000","updated":"2020-06-28 08:09:42.000000000","submitted":"2020-06-28 08:08:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738265-1593331692995-326f21e0","meta_rev_id":"d27f181b3e74664824b6c7b794750173e9d7d35e","_number":738265,"virtual_id_number":738265,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-06-28 03:46:07.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-28 08:08:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-26 22:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-06-26 19:27:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-06-26 22:26:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 19:27:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-26 22:26:25.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-28 08:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ea67d3a22623810e5558f2fe36f15e113984b43","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 19:24:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"77b8b069a66c77fe2635d0b45bdd8f2541c1e77a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-26 19:27:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6384381f77dc37f8a811c8531d1883828375503f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:33:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/095f3742ec104fe68ec203f549a63b87 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b4eecd34e942588b66ea0e4f3cadf7 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712002c899974a978ecf80ee3cf11ba9 : SUCCESS in 8m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0298f4daf0a44fcd8abfbdf49ad285f2 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3e78424ba24ba1bd4b92f2b908c3d1 : SUCCESS in 11m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4412d19c96ef4281b3f4daa393c4c8da : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e606950779849d898331d52001e98e7 : SUCCESS in 15m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0aa83725833f41269ba3451d9ecb9325 : SUCCESS in 7m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ace2848ab5a4f9e943ea14d90de5673 : SUCCESS in 9m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/392516bcae8249839531aeb9c490817e : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab7054bee2984ea4b4b5ef57290b63bc : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c4b7d3437ff4a6480a2cf3bcb8146ab : SUCCESS in 45m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b548159967ae4cf69a117dcf07113cf5 : SUCCESS in 46m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/569762b0a332404a87603d6f0a9d7bb3 : SUCCESS in 2h 01m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/37e04e038f174cffbb51f821f19bc4e0 : FAILURE in 2h 02m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/192c7ff1ddaa43d69e2b0301a7a5b272 : FAILURE in 20m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/795ddc25fae7490aa4e3bd2d47d0eed2 : FAILURE in 57m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/333df6c1119141de830488c688a7e695 : FAILURE in 49m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/31c4e3af0e444af7a4886eb5f1845213 : SUCCESS in 1h 16m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39b6b01fb4a2401d8dd23a1395173cca : SUCCESS in 59m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0614cab559574b97a25c0c34d4f1f8d1 : SUCCESS in 1h 49m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9d872ba01a545718af575e37a66bc7a : SUCCESS in 1h 00m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e452bdeb5a914811aaf5aa998cca9804 : SUCCESS in 57m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/804db56dc8054400b1498ae27a5fd0bc : FAILURE in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1491a47d211b48e5a1797217d7ddd3acafc38d62","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-26 22:26:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6bf4b96108db9488606f14dafccf3cad256ca917","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-26 22:26:39.000000000","message":"Patch Set 1:\n\nrecheck\nKnown UDP test issue","accounts_in_message":[],"_revision_number":1},{"id":"26cfa1ad5c60d1efe0bbe5f6691b0f13651e1399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 00:41:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d471318c02747c3b490bd110fb0f24a : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af8d0f28ced94d9b9d8f2ebc91b4a314 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45eadb9a3894b79ba7969ac56645ec7 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58b86a2ee7004ddfa2cac2e89686f5df : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46d5aceb0fff4fa2a089a1941da1ac66 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4724db7d00bc429aaf2a45c55af6d95f : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426e7134919d418a8623557ba481ff47 : SUCCESS in 10m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/48e8289fa1d248a8936a5721353c7317 : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9110edcfae05443eb9b57dbc4de88cd7 : SUCCESS in 8m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/11f127a5b9ef4223b9ec0d59612dce20 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d69ed3a3fd9b41269b7a8940197f3178 : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4392035ba274415ab4b9e0f40630c4e7 : SUCCESS in 44m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6a9e0e6e4bc245c8878e45e13f6c8a5f : SUCCESS in 46m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba8845d5aa1d40959ace0374c70c7745 : SUCCESS in 1h 54m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8633330ccf7248f6922e1b27fb4c75ff : FAILURE in 35m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/515a04550355453ebd32e928173d045f : FAILURE in 12m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ab4cfdd1a2445aa94fbe94449d85aa0 : FAILURE in 54m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e77b87013f814ff295614e6788ffbee0 : FAILURE in 55m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95250bcccccb4c26995f911d24e71243 : SUCCESS in 1h 16m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/731590c687ae4f739a484315f43e3909 : SUCCESS in 1h 04m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9def3078603c425f9b90df443a61b37a : SUCCESS in 2h 09m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e2658b812904c53b83d4fd2235f6696 : SUCCESS in 1h 05m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e30fd7612f4c4b04bd50911ffe6a08f4 : SUCCESS in 50m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7f804415968e4df9bbd66ab24aacbdb9 : FAILURE in 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6d23ca7a2342cb2cdd217430362bd9e533d7d4f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-27 02:50:19.000000000","message":"Patch Set 1:\n\nrecheck\nDIB rm error","accounts_in_message":[],"_revision_number":1},{"id":"6e3f1e3efcde96e3f21f27df6aee1cb413da190d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 04:36:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a176f67779a48ffb90a02021cfb51fd : SUCCESS in 9m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a02c40324574733918b32366ad3a7b3 : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a024870354b84768b7fca360635db113 : SUCCESS in 11m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1352589488144411aa66764980b27dac : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ef6668b5b1942808940fdafa966de41 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30608551efcf4ad98e6961418a97f38c : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c9d2a009554909a0ff9835dc9cf5f3 : SUCCESS in 24m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f400564f237c4b64a64b509096f7c46c : SUCCESS in 10m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/06bc1547289f4cc0902f974f8f88e5cf : SUCCESS in 15m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ff656f5f053f471390faa422e469a600 : SUCCESS in 14m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/268798990f484d24927d9b9591f6f1b3 : SUCCESS in 13m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/971e61c5d75d47ad896c3419d459d9bc : SUCCESS in 1h 19m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/34c95ffec34b4d7e8a3193c5903e14e2 : SUCCESS in 1h 15m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/545cdd46d7244cf2a2f841b4bce04ca7 : SUCCESS in 1h 41m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1229700b8acf47e0a7c23a79baf820d2 : SUCCESS in 1h 43m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f7452db8e9414fdb9114c171095d146f : FAILURE in 19m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec9aa7fcb7384ff897f35ec30b811b61 : FAILURE in 1h 29m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/69e36e29e81d49598a345655473f11ca : FAILURE in 58m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9eeb88223aeb4c728e327bdebe969e10 : FAILURE in 1h 17m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01dec7842b4a4e9295f32bfac1f90c97 : SUCCESS in 56m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e982ee290252435ba321c45b94f8b269 : SUCCESS in 1h 38m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/959ececf914342e1aaee45d995ccee44 : SUCCESS in 56m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b1264b1f82784f3c9bd429d2b8d87021 : SUCCESS in 1h 38m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3ead39f30a3f4b759e7d48f3d1ba5059 : FAILURE in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e7de3ce094f22fb77eb1ad0f6c9f5dc39fdeddd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-27 16:38:09.000000000","message":"Patch Set 1:\n\nrecehck","accounts_in_message":[],"_revision_number":1},{"id":"9ca6d0014552467fd5e4dbe0adfd244823369493","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-28 00:21:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6407d42864642afd884e90866c526d3f15d5a4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 02:19:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b7582b904f64017996796e8072cc3cf : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dcacd24987f46e8abf38b0e097b67bc : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcfbe5b33d184a8386fb2cabd7fc08ed : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70a82d605bb142e49a262a3edbe90f56 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f612662fa40e4f50a34e1bec99124e21 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73ba9dfd295d4773885916b028c87f38 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6744d3dc3961450195fef17acfec1662 : SUCCESS in 12m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3e31d9654f4a4ea694d5162dfd93485f : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e6c55431b8774441bf8acd4edcc1f01c : SUCCESS in 8m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fb858881f1f64775a86c439ad2b047db : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23a3471754984a2bb913e95f17fd11ef : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec6d4d80676746c0bb017852e06606ea : SUCCESS in 40m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b4675901bef84416886a30841378a223 : SUCCESS in 44m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3db8197a2ed941c7bcd4c7891212009e : SUCCESS in 56m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7218eacf71f3496b84ce9520368e8f3b : SUCCESS in 1h 00m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/34ebb9604b634de48fc2937427d665b2 : FAILURE in 10m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b73c145cf3443d2b3e2d55a86d28c38 : FAILURE in 46m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/38f1c2f028614e758f93fba03a9bb658 : FAILURE in 43m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b0d2a0d02b54f3c99f245d08be96018 : SUCCESS in 1h 07m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bce9148435fb4ab7862db4078d1d58b2 : SUCCESS in 57m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a0bff2a09e5447afab57ef5755827a5b : SUCCESS in 1h 52m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4f0ef1c8daa4de6ba7e8caf3f160bc6 : SUCCESS in 39m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3851199f7f0e4f41a390dea3e5f480f6 : SUCCESS in 1h 06m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2948352f091c4aa9825921c0c5cd8cdc : FAILURE in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"533914f3f4d21f0de1af625f3ec316cb8bfe7b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 02:19:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe79ba7fe7f580578a8f9e368710a52e1aff7e19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 02:42:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eea6542a7bbf419d84af419609227c6e : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f9b8b7f0e94ce1b613eb0e40e0c4bf : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bb4cca668904658ab93410075af56ce : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f690193bf34623b38a04d4c22a6873 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/202f944788d348a089e464844ebca4be : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009777f03da8419a8cad6cf23fba54fa : SUCCESS in 7m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e44e749782c549c297e35472f9d55433 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b839840cc2794b31898aee10a48fa096 : SUCCESS in 4m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/540d37d6398a4dccb1568f6ab1348129 : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8add7389da5240608288fe94c2868f0c : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f4ba04f28d245eeaa83369dde3ebf4b : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9ed4ecdbe3c24d11a9e237ef7db47cb8 : FAILURE in 21m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28b80bb78af34eb380f8dcfeb316a495 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"1572105e8529388794b693e10a09934008ccb6c2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-28 03:46:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"102bc149838d6c2d804caa676040402dc3250f1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 06:00:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50a700236ebc4c9ba1b7f4c01afb811a : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/958961d7a71c446d91178918a2b64afd : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001046c5ff96408ca91627b14390da7d : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a56a059b632143d38c147663ee0eadea : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1894d920164a4a9e9f842c3c706faf98 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69cdf0626e224ed48dd891a963b8d953 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22822ab2f0184b10b5a9c4b828bdad3d : SUCCESS in 11m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/83bbe88eab52463992f76c3bd681f6b6 : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7348a7d6c91e4331abf3022e766fca95 : SUCCESS in 7m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0ca2a3d95c944c7ba39e302303b6a64b : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66b1b9be1d814fa3a208c544b004eb49 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e09af07934e1403e9fc0bfc78813eb35 : SUCCESS in 42m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/592260526a5f4d799f40f409b9bddf7c : SUCCESS in 40m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/973ba5e0c7954bfea4a86336b22cc638 : SUCCESS in 1h 38m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4041a1d9ace845acb730359d3c480be9 : SUCCESS in 2h 08m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3f1e43a80e02428281a0004012036304 : FAILURE in 10m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9fad378908e434390b47a08f78db9c7 : FAILURE in 1h 00m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/12d3eef1aa6a47aaa2f25b095bf9bcc1 : FAILURE in 1h 00m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b88bb17e3b11492c9598f409b5ac2ea1 : SUCCESS in 1h 10m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7307604a31894ea09b357f8af9897357 : SUCCESS in 41m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0aee40e685764b308a36b3b5887fce23 : SUCCESS in 53m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17c982799928454f987565b864ea1a8e : SUCCESS in 41m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8dc2e2038568451aa12bf17aefee11b9 : SUCCESS in 43m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3ab917f153c748969c9aea6b2a86fd9e : FAILURE in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e52e985970aa90b7f63610dcb4c778b5c1a9a7ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 06:00:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99cd80177805098323f728e611cb45220630926d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 08:08:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc03e78b32f14d5e98720233023ecfd9 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c3e61ae8f3441ec8b85609adb644379 : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4a00b6c29954f2386a2e07a87dac57e : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742f40edb62f4a008ca6aef0447b13cf : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdd913f7bd3b4eafa8010a53e13ccc60 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2db56f386c34b0ab99bd1819227bad0 : SUCCESS in 14m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/afb65cfac5954d8ba646757e895d5384 : SUCCESS in 8m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a75d23ac9984d6589efbf308376e86b : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/131a43ca96a34274b6432415cc5a50b3 : SUCCESS in 46m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/835267a52d20464380b6f1c2f58aacff : SUCCESS in 43m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/955f281584ac47dbb1862ba11083e59e : SUCCESS in 1h 33m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc22f7fcf55341b5a7be041642856edf : SUCCESS in 1h 49m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1550b29ab75341f2bdf31d5814171a70 : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9646dd0a67554e77c11322794d01435f23227260","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 08:08:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d27f181b3e74664824b6c7b794750173e9d7d35e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 08:09:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2df19568393457f9157ea7cf5df4b58 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a18b6cc996e4c64d650c516071c57358681af3b","revisions":{"2a18b6cc996e4c64d650c516071c57358681af3b":{"kind":"REWORK","_number":1,"created":"2020-06-26 19:24:45.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/738265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/738265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/738265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/738265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/738265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/738265/1"}}},"commit":{"parents":[{"commit":"eb14c5acdf85f1137a694e073d0c89caec9ec519","subject":"Merge \"Fix batch member create for v1 amphora driver\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb14c5acdf85f1137a694e073d0c89caec9ec519"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-25 17:05:54.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-26 19:24:45.000000000","tz":0},"subject":"Fix neutron subnet lookup ignoring endpoint_type","message":"Fix neutron subnet lookup ignoring endpoint_type\n\nThe recent patch[1] that adds a neutron subnet lookup using the user\ncontext is not honoring the interface/endpoint_type or region_name\nsettings for neutron in the octavia configuration file.\nThis is causing problems for deployments that use the \"internal\"\nendpoint for neutron and the current code will always return\nthe \"public\" endpoint.\nThis patch corrects this problem by including those filter\nparameters when the neutron endpoint is looked up in keystone.\n\n[1] https://review.opendev.org/726042\n\nChange-Id: I7b8f7c7d653b37395f9a660be67f954a3a6f26d9\nStory: 2007863\nTask: 40173\n(cherry picked from commit 40731e572f2b503a3ccc603642a3721a5ce1cf7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a18b6cc996e4c64d650c516071c57358681af3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a18b6cc996e4c64d650c516071c57358681af3b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
