)]}'
{"id":"openstack%2Foctavia~738381","triplet_id":"openstack%2Foctavia~master~I57a3c528a20943724bdcd36422c689f496068330","project":"openstack/octavia","branch":"master","topic":"sctp_api","hashtags":[],"change_id":"I57a3c528a20943724bdcd36422c689f496068330","subject":"Add SCTP support in API","status":"MERGED","created":"2020-06-29 06:37:21.000000000","updated":"2020-09-10 16:26:57.000000000","submitted":"2020-09-10 16:24:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738381-1599755045779-b51fe808","meta_rev_id":"6e1a7ca9ea4342f13ca1fc53508fb214a51edf78","_number":738381,"virtual_id_number":738381,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 16:24:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-10 11:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-09-10 11:23:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-09-10 11:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-09-10 14:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-2,"date":"2020-09-10 11:23:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 18:32:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-08 18:53:36.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-09 11:17:28.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-10 14:11:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-10 16:24:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cb885b50f41e0bfb68240e54aba5fe6cde439f9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-29 06:37:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3e1d3ef1761a49eb40cf88d24114d18939da2cd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-06-29 06:37:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"567b4c919da0ad04ef681cb68d55a0c1aa9041ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:34:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9faada5427c1490c8d8fd89e446cca46 : SUCCESS in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02531d7ecead45198ec301b5ddd2864d : FAILURE in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db75389033644fcb9a71f96941a61979 : FAILURE in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637c39a6809640f7a825aad6ffdf2b99 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/313bcb548af04574ac0e0526f884d3ab : FAILURE in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a7f996172a54e05bc2f4ce644658584 : FAILURE in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5927a5c3eee840f6bd9987c669785782 : SUCCESS in 14m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ce37b7ca463b410390c7be3111486f31 : FAILURE in 7m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/88a2da95217e42aeb9c54fcef5401c4c : FAILURE in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e9aa3b0198a4d52a5c81005d745868b : FAILURE in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6288af6e78ec4c03b0b35288ba217803 : SUCCESS in 48m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6692fbd36bbd407a8d319b9b8f4ef6b2 : SUCCESS in 1h 08m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d55367e823314ae89bc8ca0734e1aee7 : SUCCESS in 58m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3697c681a9f84b53841aa31ed452e97a : SUCCESS in 40m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15eb05874cf743bdae0194dbf7366f1c : SUCCESS in 40m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08d08f8db6a344ec8dda09e93fe2d39c : SUCCESS in 1h 29m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/150ef9d65933485caf8444c23d1f298c : SUCCESS in 1h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd16900aee5d49d39057ad77dc2bbd42 : SUCCESS in 1h 54m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c2fd91ffde6c4c298e7e0777bc8751b7 : SUCCESS in 1h 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d12ef3a2a5246ae2d8b8bbff8ea906315679ee5a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-29 13:44:27.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"28cd445bf9eb72dcc085bd7edc5e98461ec7dfa5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-01 12:47:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c3bf469e3bd6f3e4a74fe0d5846f625322bd826","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-01 13:08:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2d1ff029cc95ed5df8e1805b297ecb8402e3be51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 15:34:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c1c1ed93d99d438ba4cb3bdaca742920 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d919b980bf0745f29bf654bcede58f50 : FAILURE in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5194486c581f47d78ffc16c650a821f0 : FAILURE in 14m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3964c8529a454066be689a115d28ac91 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aa0442e84104e35861777eb531eb9ed : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea3748e43ffc493699e62487de88be57 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0de9d5a4ca461a8bd46dd5a8b47272 : SUCCESS in 14m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/877502803a2c47788c8bd66a16463181 : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6e9f79292a254b92ae4d5ce2c739ec81 : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f7a27aa450a44dc94d93fe7f0a16021 : FAILURE in 10m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5fa1442c872c48ceb125d8284b92dd83 : SUCCESS in 44m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93189571cd174c618524d47847fffc0c : SUCCESS in 1h 31m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32a09c65b23d45768bacb4e0d6c5f8a5 : SUCCESS in 1h 02m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f247d2dc46e74969abc33f3df1ef35df : SUCCESS in 57m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f20824b7721c4458bc4cd6513e9edd1c : SUCCESS in 57m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7a715d9988f4ec89edbf9595c08d219 : SUCCESS in 1h 47m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/24ca647ad91a43cf872b3ce66ace96cc : SUCCESS in 1h 47m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3a964419882248b78c5ccea3cabbd416 : SUCCESS in 2h 19m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a340bd32739f4d479533fef9699bb445 : SUCCESS in 1h 53m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf1dc32ef21b4a48fcd953181486f416ff077e36","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-07 09:24:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"34f424469cddb80855b4da91f05960edb24982c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 11:47:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50ef700a18854f55884886e024ed410d : SUCCESS in 5m 18s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cb3a6c6fd4ca4e019ab4055927bcf2d2 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abf93095fb3944ea963c524684991e58 : FAILURE in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f87bf1459b54915854babdbfef3c307 : FAILURE in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88f8e0d2ebb4e21b10332b6e90feb6c : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25aa75cdcb6b4a93ae2d68aadcbe33a0 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f338c629c35646f6b080613e2ca8ac0c : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e0116f547149c882a8543d948433b6 : SUCCESS in 15m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a8dd61cc6e834733a6034dd72eaf48fb : SUCCESS in 7m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c6be2daf426f409681b70dd7b2378f19 : SUCCESS in 8m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e118c0d6194d4c48b36c03c381a1247b : FAILURE in 9m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e47f2db982b49208c2eee73b1b25d79 : SUCCESS in 46m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cc7c3abff884fc5927dc92408830741 : SUCCESS in 1h 07m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14404adf3df942f885490b71728bb7dd : SUCCESS in 55m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dcfb2e1b306842ed8cc963e8025664ca : SUCCESS in 56m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24c340c86c3a4a80bbb8c711848704ed : SUCCESS in 1h 04m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f9e3f814602b41f897a38fa92f472bf7 : SUCCESS in 1h 34m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2095fbe60c5e45068644d2caeb1f655e : SUCCESS in 2h 16m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ac2a5e8bd2947b88dc9c8e2a97d3281 : SUCCESS in 1h 49m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5f75807e0ff6480eb2c08688059e5f76 : SUCCESS in 1h 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"146b743032329c4c9d56026fdf4c2f539b6b2758","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-12 11:42:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e8c4c3eb6b446ffdfc492f981532dbd56198fd70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 14:41:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/525ef0dda1034f95bd9c4d3b773a6b19 : FAILURE in 16m 55s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ffcc9568f1b4ab0a675100fe5dce48f : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec18728923854523970be524f4eeacbf : SUCCESS in 10m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/001cad17bee34378ac0a4f95c629993c : FAILURE in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9604ede31a43c8acc56d6aa6ed288d : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cb5675b12e84efb924bc30eb76fbd2a : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02c9beca2eb349cda78bafca1b831157 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/676d396776ae4f6899697ed237a843cf : SUCCESS in 12m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8249b75dff7f4f5583d5adb34acec027 : SUCCESS in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7466cce6a5ed4da1b3b9eaa1091b4d49 : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/627140ac1370401f88be01dbd877a269 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5dc2bee0623949f088e17da8cdd4507f : SUCCESS in 46m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd61400588f24859acc6e267d7469e0e : SUCCESS in 1h 21m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ad76c624cf34f44badc19528b4ce21b : SUCCESS in 1h 05m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b0f9210305149f4b2cc1a1d055c020d : SUCCESS in 1h 04m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dd5887c8b6140bf931c87443234f047 : SUCCESS in 46m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6e7268a9a7974b2e9c83053db64f7308 : SUCCESS in 1h 38m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/047ab2f645464458ba4149c1cf7c41f3 : FAILURE in 19m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4afa13af143742288a3cf263fe41a3f0 : FAILURE in 35m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f5bef720b3294f3c8edc3a0d666f9b2c : FAILURE in 51m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c393b7ab26664d68850e1884b1e13537 : TIMED_OUT in 2h 33m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2642036c72564576b869ca53c0e47ee1 : FAILURE in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"19b7e49d92de6d830830621ab51a5b6e38008800","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-28 13:19:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d186a29a9f8c71800fd2ba1c295422b65efa944c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-28 13:46:21.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)\n\nLGTM. Just a few small things that could be improved.","accounts_in_message":[],"_revision_number":6},{"id":"2ce3e18859db43f72ab49d30e4ac3692b3271e51","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-28 13:53:04.000000000","message":"Patch Set 6:\n\nAlso need to bump octavia-lib to \u003e\u003d2.1.0 in requirements.txt and lower-constraints.txt","accounts_in_message":[],"_revision_number":6},{"id":"47feeaa853b6b3172038a90202fe3e7ae0c8a328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 14:58:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/561c176c04404f3ab8ea4331edb78116 : SUCCESS in 7m 06s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff273c2a01c5442d840a38fddef6e5b0 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e2793af9dfc475bbce246c6190ba4c6 : FAILURE in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9feac886a8844d4bb464635d44d9a65 : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fdeb611681a49799080e5534d8a6223 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7583d41e0498428b9436d018318b32f6 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dca2446450746188dfc3ac0c8b901b2 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38368c8073c3495bade721334ce67849 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea27153424864130b6ba38fa6c101379 : SUCCESS in 6m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d8a2b16f83a44b0bb518cb5fb0f26b0 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cfba5b917e214ffaa2bf7f83e864a7f6 : FAILURE in 10m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ba173c2a9a24da2b52d42da832e47d2 : FAILURE in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3145f361b81c47f29f3184a59ea4f186 : SUCCESS in 37m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eceaf3457f2943518314cb7aa6b67dd1 : SUCCESS in 1h 04m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d8b749969894f7faf542f038080e075 : SUCCESS in 1h 06m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f88d6f9eaba743a48738d39997091db6 : SUCCESS in 1h 01m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9ecc597463f4a52899e22485f8a27c1 : SUCCESS in 35m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d518c4aaaef4cb9996317b69d353420 : SUCCESS in 1h 05m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/23890eaa5fb74d5f9020404c6cdb2606 : SUCCESS in 1h 22m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd88a36ff2e443488713dfa7fcb9a42a : SUCCESS in 1h 29m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e72e88018f2a4b7e93579a481a2fa4fb : SUCCESS in 1h 21m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b7e65f7cfc5246cba7322faf158b3bcc : SUCCESS in 1h 15m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ee3d098d4acf41d1bace2a1d5358866b : SUCCESS in 1h 21m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6194f13f31d64b6384c7019b9c6a5b20 : SUCCESS in 1h 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4215be406b6cfc94f03adce09146291fe99e10c9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-07 08:32:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"44c98ec9f83410bdb7431a6b7ae7580264b8abde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 09:44:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/717f1a3a278f4154b63cf1c5844241cf : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2883f521d9bc4ad1a3d6f65ab9a61cec : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98ada169dd0745a1a6ac370bcab33c71 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3eff2044d3437bb9e23f7202c1bdab : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa8f4471fca44828c95b2d11a801756 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63842038439c49b5a45f66081aa52941 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ffbf86f13f4f72be025596dfbd64c9 : SUCCESS in 11m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e1cf420b92874473884c4eacf14a87a2 : SUCCESS in 5m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9c34c64286c4474e9ed71290eba043e3 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de4139e6a135407784294fe320f772f2 : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/251bc0737a744abf91c68c83f10e63a7 : FAILURE in 24m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d04004f7e44a4ac6b5957b455ac9a855 : FAILURE in 19m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b144e4d7ef0b4beead616a8b8907395b : FAILURE in 34m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/47d94378354349bfa5e41b4ce19d7c8b : FAILURE in 23m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62ca99ed8f7945858c6ae018f9a4d037 : FAILURE in 20m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/986cf314c9bf4da1838476feeecefb10 : FAILURE in 1h 08m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9fcf0a3705f45168379a72bb01313a5 : FAILURE in 28m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6a2cca0f6d64a17aa7ed2c757557da0 : FAILURE in 23m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10938935b66b4e0b8ab446ca1244ff47 : FAILURE in 36m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2dc27deeb13044ffb5f9186d6f2357e9 : FAILURE in 27m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/893b1560a6a44b7989eb2ad8c43a2654 : FAILURE in 38m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c3afbc55a29f459485cb7fb31bc44049 : FAILURE in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e6d44d63d74030a848a32bba02d1687609908ca0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-07 11:36:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"aeab67c6c4a2e435b5757c8df48fcf9d7d516797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 12:55:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d1da8351230c4670adcdd018c64f0902 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2ab8e70da2141a6bad6f838f60a9e98 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5d9c6a926c04bce9a787815ea09e1b6 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9971ff3d9ff34612bcdb6f3b3683902b : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f3826b9ad54503857e026b4b64d2bb : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11651d2a12c04799afe9ed0fe8fd0295 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d87734878844b0fb98b599127ba6791 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9328daeac70b41daaf662ab83680c6ef : SUCCESS in 4m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9afd9426c7ef4a5b873245421b51bbe7 : SUCCESS in 7m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/08d3b4395353468bbd2f75bb58f78d92 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd523881094e4467ad3d398cb8b15703 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0f669788a3d4509865d278a5a3f5110 : SUCCESS in 46m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d15ad88edd5a4e02b665bc1cebb96b8a : SUCCESS in 1h 01m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/741498b93dbc42659722787899fce143 : SUCCESS in 36m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d1ad9cd51c3c4d618019af37c9c45fac : SUCCESS in 39m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b22faa2d16b04842a85b50d8360e5d59 : SUCCESS in 39m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0660e18ec87549e09d99c4a329649ef5 : SUCCESS in 1h 03m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/39109daf74904e77a2b8e6cf6933dda1 : SUCCESS in 49m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c53f27e061d94e7aadf945ff90f9619a : SUCCESS in 1h 15m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d9ca42363ba4b4db2fc21acff4503df : SUCCESS in 46m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c81dd6dbdeec4f6184640fcd09c65ed2 : SUCCESS in 55m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fcf18a5192a24e99a05f155ae85fe52c : SUCCESS in 59m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/570965dfe0fe4b96b401c5f14f57e011 : SUCCESS in 1h 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"49921c07b6950e1911254b610f864e100212180f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-08 17:19:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"20567265435453d596c96b4174e7815a34c80cfa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-08 17:56:43.000000000","message":"Patch Set 9: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"db1a5d8eac56c73172700fd232c12ba9e57bcc46","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-08 18:32:03.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ea0b0c4a5be19aedbf9094ada8dc0fa7d68a432b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-08 18:53:36.000000000","message":"Patch Set 9:\n\nWhere did the amphora implementation go?","accounts_in_message":[],"_revision_number":9},{"id":"46f74876d53bacb1dedb5c15154dc80dc617e788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 19:21:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6c4ec94a9a6846cc94bb5c4b9c237401 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25b8ea5ee5c64f208095fc5f714ba17c : SUCCESS in 10m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f2163d053c42939b3e84a4d059886a : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf08f9d38fad4db6a1b3335acb4eeb24 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bce1c3f5e37493eb4caea31b646b441 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edbe762e5956449fa25b9ad0d7694d0c : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a19ccc1bd94bf7934c5981363a00fe : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f3b853e12b7458983fc95fd34425e0b : SUCCESS in 6m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b7227eaf24e4990bb20a41c789e2a77 : SUCCESS in 8m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cb44ba7b2950446281085544f292012e : SUCCESS in 9m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73e5c211c7d74ec5be071530caca6b8d : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5eab9dd1bd2b4f2eae3356f905b636e6 : SUCCESS in 41m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b18b93b78ba4facbdc1ebbcca189fe5 : SUCCESS in 53m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad9091fa84434333a3b8e86a8c360eb8 : SUCCESS in 33m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2c395077272a499e98b0d0abbae4d0b0 : SUCCESS in 35m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9df07a14666249d8ab4f0a4452b5f194 : SUCCESS in 42m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/93f925e977504070ab09b8e06c625993 : SUCCESS in 1h 55m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45d01d1718864d798e9d1f4ea917f2bf : SUCCESS in 59m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05b5356c277241d1b3732f87baefa5bf : SUCCESS in 56m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9b993a8adfed47a6abb7578cb0f17794 : SUCCESS in 47m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ac43b95ab9dc4952bfe363acf83f33ff : SUCCESS in 56m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/33202e2646e941ca8955af60d4422354 : SUCCESS in 1h 19m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/22448a01b4304cea904897958f6d3adf : SUCCESS in 1h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6116cb470e46f30899e6e0854e51390afd4549bb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-09 11:17:28.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c12eb62088a064ba40688fe23018904de60f33c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:17:41.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"414c43eb1e5ec7a760ef61fe2a36d44a812b0ede","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 14:26:08.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0d2eef8dfac342da941ec6e1489997a1 : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6491c2b566e240b9ae96bbfa0dc4b341 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a2feec2c9c47d8a6c8e19fa4b2a750 : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2734201f8a094aef9e8f28e1a3cc25b0 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01b510ddd71e40d3865ffe8cefc08592 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3a5bf461af48acb294307fb139e935 : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6585b0a287ca4df98e10bb568b9f40cf : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/754343b6710d472b843614cef22a34e9 : SUCCESS in 9m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/921fd764badd422ba50fd9e201471709 : SUCCESS in 48m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1753abac62f404fa893b12013e7784b : SUCCESS in 1h 38m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cfdec6ba7bc64dd8a5730b3844fadd11 : SUCCESS in 46m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d3328c5c4135446eb7e0175798667e88 : SUCCESS in 41m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3487364026b43a1bb0c6283607f565a : SUCCESS in 43m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89785d83d66d44ec8d89ce60b3cff384 : TIMED_OUT in 3h 04m 40s","accounts_in_message":[],"_revision_number":9},{"id":"620e476ba4a1005e793c371fa86cdf6134d05e4a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-09 14:34:52.000000000","message":"Patch Set 9:\n\nrecheck gate","accounts_in_message":[],"_revision_number":9},{"id":"d8215e36b88a4c092833d67f33a2a5ada015718c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 17:51:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/25d5eb2a1cb74076b7d2847d99a0256e : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de220efd89e646a890b98127c9224bbb : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0546edd927094865952d9b20996b27b4 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae77db622ab4328bdbcf4431c125b16 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07c773ec17f942a78cef1af46ef4bce8 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4b570d4d2ab48c0821fb9812910ff8a : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3700f537cd4d15bc6319a0a13e202c : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47615b752a3e481d95bb272c97df1931 : SUCCESS in 6m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/96c471cb21024348969b84fe9829bd6e : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/900cbb012b894e32a0ea381b8419b4e4 : SUCCESS in 9m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74789d2f4ed8429da4f285f0aaf476c5 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/81015a9d75394fb38da517f1edc57e66 : SUCCESS in 43m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2bf05cdeb704ddaae9aee42f216f38e : SUCCESS in 1h 47m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ee955fd91204a93bdc84ce345102579 : SUCCESS in 40m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/836136a2f95e4f888258a33b9b7aecb9 : SUCCESS in 1h 07m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56d06fb37d6f405686ae1397e670e4a9 : SUCCESS in 1h 18m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cf7c79301aa048d790c9257a81b9d7a3 : TIMED_OUT in 3h 03m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a635cad18c404fdc915585afb9382459 : TIMED_OUT in 2h 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/78cb710b5dc34180879544516320a3c6 : SUCCESS in 1h 43m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b7b241f440d24be8a755327a963913d2 : SUCCESS in 1h 31m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/badc22afb6394c9796834ed2d74ea669 : SUCCESS in 2h 30m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/be6bd666990042e1ba9e57c206582105 : SUCCESS in 1h 51m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/327f9e21acfb47b48036a3acce0f7f70 : SUCCESS in 2h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9aefe134c2e3f802537726a7eaac4642ab75b8fb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-10 11:23:04.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"2b4b409f66c68c22bca96c9fe4b3f1269e6fb1f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 14:00:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f3fd24273f0f4a1294b9e0cb6978d532 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a3b12f9405444b08c0734e62dfdc024 : SUCCESS in 12m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e60d411bb09461f847647692f65b2e2 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3c4c51fa09416981c68daca5278e06 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7934fa6040940fbb3ec515af689dc0e : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f32820646c74671b26084d509986d87 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbeb642447042d9b8b6cfe55778fab5 : SUCCESS in 19m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af9cb4e47106403d968a9bb4596c7147 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/89480fdf0d26461294a5e9a78bd34de9 : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/adc1bd7b0ea84c6f94c708de44f6f33f : SUCCESS in 12m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62fac671824748678f0839038bfae188 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/04ea56ab8b6b487ca06ab255a1750f52 : SUCCESS in 44m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95e86b23c6054a8da622cfb4c0718c74 : SUCCESS in 1h 52m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa58baccbcf24c2f979f1f0d31afded9 : SUCCESS in 41m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f365aaffaffc4bd780c7f30395ed2d84 : SUCCESS in 35m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c6fbea146ae4adb8bce20bd0f8cda8b : SUCCESS in 36m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d8188514c1fe4b6b860ba2d931146118 : SUCCESS in 59m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e5a97a1c20bf411d88ec356e5342d90e : SUCCESS in 2h 19m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d51bb5043ed74c728015c65918df168f : SUCCESS in 2h 17m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6d9990e5aa0e4825ab4e2b5e15aacd6e : SUCCESS in 1h 49m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c9fbb906cea429e802131afaaedf1b4 : TIMED_OUT in 2h 34m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/23e91c7389c14c168a6783f6ec059ea0 : SUCCESS in 1h 54m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2f98957cfc6245d9b7024272b25fdff6 : SUCCESS in 1h 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8fa3d6fea65fd61da3df12344c799c3477ae639b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-10 14:11:28.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"41202caf015947e21f33366b7f1acdc67db35c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 14:11:45.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"577ac5e4c92173e5f330d9b633beb736312fd74e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:24:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"d1511d5f42fe9a6ee2b9b5572e384cc969d47247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:24:05.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/db06c4c5132e49909570f78859d60448 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bca6d24ab0348c6b36c77ef9a6ceeda : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9e059341204c7fa4b124616488aae8 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/567b870ffa8d4a79a40dadb97d3bdc8f : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6f2f25caa77470b91fc0cb8ca2ee066 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8598e38599646df82dbbff760e28782 : SUCCESS in 19m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7af53f010c64be58e2e85218183bb58 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84c7198efff145728a61770e0aa66e95 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/efd11dc86cf74e67849b74ec0930131f : SUCCESS in 48m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9306a88659574942884b756b4cd63d43 : SUCCESS in 2h 09m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67234cc9dbde49b89c8880c00291285b : SUCCESS in 38m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a6d9766d16124f0fa05799c7edfb0e1d : SUCCESS in 33m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/634db2a5eaec48dcae2d066dde7d0037 : SUCCESS in 40m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/235efb3b0e1e4481a1df94e4d6999fd5 : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":10},{"id":"fb6e02a2c7077733a636c03882e6510cfaac8e06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:26:57.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/16f722fc902549f494c05bddafc5040a : SUCCESS in 2m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede6620850b14be4a22c31c4c7fcf30a : SUCCESS in 2m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/785f243dd9bc45a39419a5cd21716a06 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"639c11751e07eed48d79e7b81f12bac697eaea73","revisions":{"55c6f257e17adb131d42a483ad54e6822286c771":{"kind":"REWORK","_number":1,"created":"2020-06-29 06:37:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/1"}}},"commit":{"parents":[{"commit":"41c628a084002017d2003926cf0e25ba3ffeee0c","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41c628a084002017d2003926cf0e25ba3ffeee0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-29 06:33:57.000000000","tz":120},"subject":"WIP Add SCTP support in API and Amphora","message":"WIP Add SCTP support in API and Amphora\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55c6f257e17adb131d42a483ad54e6822286c771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55c6f257e17adb131d42a483ad54e6822286c771"}]},"branch":"refs/heads/master"},"fce709f2d45615f1e65c335af0038d4e7ed1b241":{"kind":"REWORK","_number":2,"created":"2020-07-01 12:47:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/2"}}},"commit":{"parents":[{"commit":"41c628a084002017d2003926cf0e25ba3ffeee0c","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41c628a084002017d2003926cf0e25ba3ffeee0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-01 12:22:06.000000000","tz":120},"subject":"WIP Add SCTP support in API and Amphora","message":"WIP Add SCTP support in API and Amphora\n\nDepends-On: https://review.opendev.org/#/c/738834/\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fce709f2d45615f1e65c335af0038d4e7ed1b241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fce709f2d45615f1e65c335af0038d4e7ed1b241"}]},"branch":"refs/heads/master"},"29b344453955bd3b94f91c9ec52044aace3377a9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-01 13:08:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/3"}}},"commit":{"parents":[{"commit":"f68e3aba4944a21eef9de2a395899f426f049e2b","subject":"Merge \"Remove amp_ssH_access_allowed option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68e3aba4944a21eef9de2a395899f426f049e2b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-01 13:07:45.000000000","tz":120},"subject":"WIP Add SCTP support in API and Amphora","message":"WIP Add SCTP support in API and Amphora\n\nDepends-On: https://review.opendev.org/#/c/738834/\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29b344453955bd3b94f91c9ec52044aace3377a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29b344453955bd3b94f91c9ec52044aace3377a9"}]},"branch":"refs/heads/master"},"5d51f130c0edae470a046a525f59a1ec34c9d3ac":{"kind":"REWORK","_number":4,"created":"2020-07-07 09:24:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/4"}}},"commit":{"parents":[{"commit":"f68e3aba4944a21eef9de2a395899f426f049e2b","subject":"Merge \"Remove amp_ssH_access_allowed option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68e3aba4944a21eef9de2a395899f426f049e2b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-07 09:15:17.000000000","tz":120},"subject":"WIP Add SCTP support in API and Amphora","message":"WIP Add SCTP support in API and Amphora\n\nDepends-On: https://review.opendev.org/#/c/738834/\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d51f130c0edae470a046a525f59a1ec34c9d3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d51f130c0edae470a046a525f59a1ec34c9d3ac"}]},"branch":"refs/heads/master"},"ac01f439457b1607f067ea700e09088d626bb220":{"kind":"REWORK","_number":5,"created":"2020-08-12 11:42:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/5"}}},"commit":{"parents":[{"commit":"aa52cb39caf05d9ce91b6569271e0b831c8a589b","subject":"Merge \"Add line terminator to certificate list file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa52cb39caf05d9ce91b6569271e0b831c8a589b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-12 08:43:28.000000000","tz":120},"subject":"WIP Add SCTP support in API and Amphora","message":"WIP Add SCTP support in API and Amphora\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac01f439457b1607f067ea700e09088d626bb220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac01f439457b1607f067ea700e09088d626bb220"}]},"branch":"refs/heads/master"},"b2c7ffc0a90cbc8f8d179ae0a17e7e4c9fc02144":{"kind":"REWORK","_number":6,"created":"2020-08-28 13:19:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/6"}}},"commit":{"parents":[{"commit":"2d97ebcd24a786a58e67d1f7af344d29263adcd9","subject":"Merge \"Allow amphorav2 to run without jobboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d97ebcd24a786a58e67d1f7af344d29263adcd9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-28 09:02:35.000000000","tz":120},"subject":"Add SCTP support in API and Amphora","message":"Add SCTP support in API and Amphora\n\nAdd SCTP support in the API for listeners, pools, health-monitors\nresources.\nAdd SCTP support in the Amphora (with keepalived).\nAdd amphora-health-checker script for customized SCTP health checks\n(INIT/INIT-ACK/ABORT).\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2c7ffc0a90cbc8f8d179ae0a17e7e4c9fc02144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2c7ffc0a90cbc8f8d179ae0a17e7e4c9fc02144"}]},"branch":"refs/heads/master"},"cd71bed4abda30fcc7c2c1072f4f0c8a82616201":{"kind":"REWORK","_number":7,"created":"2020-09-07 08:32:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/7"}}},"commit":{"parents":[{"commit":"46de66b240697b1e6465c06a203fe6dac75b7da1","subject":"Merge \"Update amphora v2 for the failover refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46de66b240697b1e6465c06a203fe6dac75b7da1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-09-07 08:32:25.000000000","tz":120},"subject":"Add SCTP support in API","message":"Add SCTP support in API\n\nAdd SCTP support in the API for listeners, pools, health-monitors\nresources.\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd71bed4abda30fcc7c2c1072f4f0c8a82616201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd71bed4abda30fcc7c2c1072f4f0c8a82616201"}]},"branch":"refs/heads/master"},"4d978d0d51c4f5ee1fdfab4a26ff60842fcd2911":{"kind":"REWORK","_number":8,"created":"2020-09-07 11:36:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/8"}}},"commit":{"parents":[{"commit":"46de66b240697b1e6465c06a203fe6dac75b7da1","subject":"Merge \"Update amphora v2 for the failover refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46de66b240697b1e6465c06a203fe6dac75b7da1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-26 08:24:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-09-07 11:33:32.000000000","tz":120},"subject":"Add SCTP support in API","message":"Add SCTP support in API\n\nAdd SCTP support in the API for listeners, pools, health-monitors\nresources.\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d978d0d51c4f5ee1fdfab4a26ff60842fcd2911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d978d0d51c4f5ee1fdfab4a26ff60842fcd2911"}]},"branch":"refs/heads/master"},"d3e31ecf4383ca21be3fe1105d69f7e04f44f267":{"kind":"REWORK","_number":9,"created":"2020-09-08 17:19:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/9"}}},"commit":{"parents":[{"commit":"ed1ff8fb4ca733069f5e89a8929c8d53a0ad0d4d","subject":"Merge \"Add proxy v2 protocol support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed1ff8fb4ca733069f5e89a8929c8d53a0ad0d4d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-09-08 17:08:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-09-08 17:19:13.000000000","tz":120},"subject":"Add SCTP support in API","message":"Add SCTP support in API\n\nAdd SCTP support in the API for listeners, pools, health-monitors\nresources.\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3e31ecf4383ca21be3fe1105d69f7e04f44f267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3e31ecf4383ca21be3fe1105d69f7e04f44f267"}]},"branch":"refs/heads/master"},"639c11751e07eed48d79e7b81f12bac697eaea73":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-09-10 11:23:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/738381/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/738381/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/738381/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/738381/10"}}},"commit":{"parents":[{"commit":"4260d8a74bae4a46cd9104462eb98fcf30582d37","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4260d8a74bae4a46cd9104462eb98fcf30582d37"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-09-08 17:08:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-09-10 11:23:04.000000000","tz":0},"subject":"Add SCTP support in API","message":"Add SCTP support in API\n\nAdd SCTP support in the API for listeners, pools, health-monitors\nresources.\n\nStory: 2007884\nTask: 40255\n\nChange-Id: I57a3c528a20943724bdcd36422c689f496068330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/639c11751e07eed48d79e7b81f12bac697eaea73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/639c11751e07eed48d79e7b81f12bac697eaea73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
