)]}'
{"id":"openstack%2Foctavia~738402","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ie71db27dc383c93496c1dfd69f486a4fd02b597e","project":"openstack/octavia","branch":"stable/train","topic":"bug/1877818","hashtags":[],"change_id":"Ie71db27dc383c93496c1dfd69f486a4fd02b597e","subject":"add the verify for the session","status":"MERGED","created":"2020-06-29 08:51:41.000000000","updated":"2020-07-17 14:18:54.000000000","submitted":"2020-07-17 14:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738402-1594995398413-9ed56beb","meta_rev_id":"f46b0c0249b9775a46f84b6a89450fff252bc861","_number":738402,"virtual_id_number":738402,"owner":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 14:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2020-07-17 12:48:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-16 21:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-07-17 12:48:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-06-29 08:57:51.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"date":"2020-06-29 08:56:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-29 08:57:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 08:56:40.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-06-29 08:57:51.000000000","updated_by":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"reviewer":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"state":"REVIEWER"},{"updated":"2020-06-29 08:57:53.000000000","updated_by":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"reviewer":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"state":"REVIEWER"},{"updated":"2020-07-16 21:53:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-17 12:48:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-17 14:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83810f0b9476d4fe4e949bc917ca9f8f0090218e","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-06-29 08:51:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"891ad1975e5aa572415871537b0aef95a152df5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 10:56:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c6e627304864f059d3aa3d8987f004e : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b44f0ca469764a01bbcf5973768b28e2 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e4cc4de94d4e9fb2e48b7e2ec78e85 : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a148dd657ab4b1ea34b4433cf4e4d47 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c567efd496b84fe198e5e6baac4a5d80 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7631f6120754b9e8d8fe6920c44d6b4 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d65e7faca54fccb40a347559568666 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/055ff9cccf1c4b86baffce7d3d59d8ed : SUCCESS in 7m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea090576ab184548a17e81ffdf105d01 : SUCCESS in 7m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/42795618a65741509021a892fb5d71cc : SUCCESS in 8m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d94e028887da4c3fb973ba84bc796f8d : SUCCESS in 9m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4bcbb7f72994c0485e2cfb557e87ce7 : SUCCESS in 9m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d61b502efcb4484a82475a53a5d4b82 : SUCCESS in 50m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/21a855d4658a49039ae60daceb18f31a : SUCCESS in 55m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4f099e9862446db8d05a575694dbc7e : SUCCESS in 54m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/15a11de909b84eabb4282b5f055f862d : SUCCESS in 1h 14m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/25e49e47174549b7874e5e9ba0f49f80 : FAILURE in 12m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8380f66e30cb467185cca395c9c4206e : FAILURE in 53m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9969a7d63fda471391e6c8d283063d6c : FAILURE in 58m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee5c4c214a8149e88d9dcf4201068f48 : SUCCESS in 1h 35m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f39b732706b4e009b4438c277e1da7f : SUCCESS in 1h 01m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/216ff9d6a3af4d29bc10adc64fbc277c : SUCCESS in 1h 56m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e117daba8a18417490b05388fe681e45 : SUCCESS in 1h 06m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c181ce85c9b4478b9886e2ca433d5808 : SUCCESS in 53m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2be4caa361e0440abafbead41762e7de : FAILURE in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cbd32348c43b69312623462dfc38f8ccbc87080","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 21:53:29.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"5c5246df6133d15fc129823558723102f00f66da","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-17 12:48:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"989213869b0de51c73f7eadc6af51f2a72459619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 12:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c275dd7653102fa51583a455afc823126ef0f655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 14:16:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e38d0599218459db2a3144b006d48c0 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1aedcea1ff434eb1d5edb99b2efdc7 : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/923cf2eb9a10437d8769adc5c4e0a666 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/990fa6e7df66443aa444e223131f98ba : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff824433208d45adb7a55c4f29930766 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39eda5417dc64e33b809880676ac768a : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f69074aa6ff84f5bb01bfb10279c64c2 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8c9557ab8bb045cda961f2692d97d0d8 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2af0264b8aca4af1997c5e860d2e187d : SUCCESS in 9m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0585e3f9a00c4af492959c7b15a4f82f : SUCCESS in 43m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c5bf48ba0e8c49229e2e74baa7487338 : SUCCESS in 1h 09m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0e200fdac694210b6172d2eca3f5b6d : SUCCESS in 1h 02m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/06af5ca65fc2486282ecd8d2aec68e49 : SUCCESS in 1h 08m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/41d0fe0cfa234ca19467a90a49c28c80 : SUCCESS in 1h 24m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fb3ad229101690df605b501ad1735ea46a8e22e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 14:16:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b0bf456c2c342f03fbdb763afb5abf65b4eba82b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 14:18:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27cc1f21ed2046bbbc0d293ac9b30717 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3be5a3fc9d6247f09a3fb754451a018a : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5954fe677b106d33683285692a6d0ff46f70e775","revisions":{"5954fe677b106d33683285692a6d0ff46f70e775":{"kind":"REWORK","_number":1,"created":"2020-06-29 08:51:41.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/02/738402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/738402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/738402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/738402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/738402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/738402/1"}}},"commit":{"parents":[{"commit":"2a18b6cc996e4c64d650c516071c57358681af3b","subject":"Fix neutron subnet lookup ignoring endpoint_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a18b6cc996e4c64d650c516071c57358681af3b"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-06-29 08:51:35.000000000","tz":480},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nStory: 2007662\nTask:  39754\nCloses-Bug: #1877818\n(cherry picked from commit f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf)\n\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5954fe677b106d33683285692a6d0ff46f70e775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5954fe677b106d33683285692a6d0ff46f70e775"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
