)]}'
{"id":"openstack%2Foctavia~738403","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ie71db27dc383c93496c1dfd69f486a4fd02b597e","project":"openstack/octavia","branch":"stable/ussuri","topic":"bug/1877818","hashtags":[],"change_id":"Ie71db27dc383c93496c1dfd69f486a4fd02b597e","subject":"add the verify for the session","status":"MERGED","created":"2020-06-29 08:55:07.000000000","updated":"2020-07-20 09:37:53.000000000","submitted":"2020-07-20 09:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738403-1595237775972-8f1d0fce","meta_rev_id":"0eb3d958a2c20b04647967eaaa52590254973683","_number":738403,"virtual_id_number":738403,"owner":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 09:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":2,"date":"2020-07-17 12:48:25.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-20 08:14:54.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-07-20 08:14:54.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-06-29 08:59:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2020-06-29 08:58:37.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-29 08:59:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 08:58:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-29 08:59:25.000000000","updated_by":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"reviewer":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"state":"REVIEWER"},{"updated":"2020-06-29 08:59:27.000000000","updated_by":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"reviewer":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"state":"REVIEWER"},{"updated":"2020-07-17 12:48:25.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-20 08:14:54.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-07-20 09:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a331e47b6a02bec51aa08f1861709c9b2134144","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-06-29 08:55:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e39d24beed0c869d0f740f65d0995bb5d546a336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 10:49:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/236c46ae38f64a63ae946ebc8795bf20 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53dcba98a1584f3295c0a824037fc04a : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beffc3e1ea02493499c5f58215d950e7 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f157f82f0f054c5aaae6bfa7268d4548 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6922b15f92904c3c8dae98247b1809b3 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e0c02185544436bf9f682ce4983ecc : SUCCESS in 14m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c0a1cc8714a48ad954e4bed28678120 : SUCCESS in 6m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/08a51676e2f741afaae9b25eb390ca40 : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/41087cb313d145e4911d3129a42b3737 : SUCCESS in 11m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98ecc95c265c4ef486a1fce57fab12e9 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/51136442c950453186afc9352403fec9 : SUCCESS in 45m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9842ad043304478187a19800275403c6 : SUCCESS in 1h 39m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/feab62e23bd04f6d92bc76647979c44b : SUCCESS in 1h 01m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0d2f9aef9aa9449f94ffd86d69f1618f : SUCCESS in 56m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45f5acd83fcb462a9bfb124008f4b1bd : SUCCESS in 1h 07m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c081ede56c24d1ab426203512a99070 : SUCCESS in 1h 51m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc266161e9ac40bd8bc414832214369b : FAILURE in 53m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0b2d93d2fa9428499fecb092801661f : FAILURE in 39m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1612635eea5341fd9145535ffe36703d : FAILURE in 1h 01m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f41e391501f181aa4a5e1fea59aae5acbd09ac9d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-17 12:48:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8bfb9a105ac0c183d490267e26ee69250c47385","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-07-20 08:14:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d2555665c0ebf1802600bdf00905d3708acf9ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 08:15:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4db270c4ae9b5007cbdbeaf6eaf1079e105ba17b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 09:36:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7f3944878e889c4e03f37715a858080de62cca88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 09:36:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec47a17591d6424086f1bfa8c4757050 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce0750dc1cc45b1ada15507ac126222 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2c30ccc6352417db85eb385c98155a0 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d37e54c2c152447dbaf733e033f17619 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8560f04c81041c4a0be891383d991be : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/142446c8304f41968f60a52bc497eb26 : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12a5ed3a95eb4b87bc1d281fa45015f9 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64a838cfa23d489099936fc6bdb332a4 : SUCCESS in 52m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9dbf7f4f04aa4c09a4986063f38557c0 : SUCCESS in 1h 01m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8fda83d2fdc8411cb972abaf32aeac6b : SUCCESS in 32m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e5465a1cde1340daad6974dc3fd49a5c : SUCCESS in 37m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ccd5461054864c81b9bd0f91b6076927 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f3875327ea632f46083142d70ff9b4418e2a632f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 09:37:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6269d9897641442da22bcb5c18dcdb4d : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47a763ce465442d59c9087cc7979d259 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12e30394ad9332ad122a8590c4904e6ea8f48844","revisions":{"12e30394ad9332ad122a8590c4904e6ea8f48844":{"kind":"REWORK","_number":1,"created":"2020-06-29 08:55:07.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/03/738403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/738403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/738403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/738403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/738403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/738403/1"}}},"commit":{"parents":[{"commit":"b4bc92ebf1a5610cb340299d6230a134f202bfbb","subject":"Fix neutron subnet lookup ignoring endpoint_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4bc92ebf1a5610cb340299d6230a134f202bfbb"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-05-10 04:58:02.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-06-29 08:55:02.000000000","tz":480},"subject":"add the verify for the session","message":"add the verify for the session\n\nWe run the octavia scenario test failed when the OpenStack env\nenable TLS. So we need add the verify for the session.\n\nStory: 2007662\nTask:  39754\nCloses-Bug: #1877818\n(cherry picked from commit f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf)\n\nChange-Id: Ie71db27dc383c93496c1dfd69f486a4fd02b597e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12e30394ad9332ad122a8590c4904e6ea8f48844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12e30394ad9332ad122a8590c4904e6ea8f48844"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
