)]}'
{"id":"openstack%2Foctavia~739002","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I04cb2f1f10ec566298834f81df0cf8b100ca916c","project":"openstack/octavia","branch":"stable/ussuri","topic":"failover-refactor","hashtags":[],"change_id":"I04cb2f1f10ec566298834f81df0cf8b100ca916c","subject":"Refactor the failover flows","status":"MERGED","created":"2020-07-02 09:36:29.000000000","updated":"2020-09-08 23:46:20.000000000","submitted":"2020-09-08 23:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":989,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739002-1599608569521-b51fe808","meta_rev_id":"a1a75feefeb41204d9e435cc3120dd0f8035cd09","_number":739002,"virtual_id_number":739002,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 23:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-08 08:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-09-08 08:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-08 08:47:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 20:46:31.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-08-28 19:01:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-08 08:47:07.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-08 23:42:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79bbb77cca321207f6dbfb43f5d0b794c4d4ba02","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-02 09:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f0788b78f8e15c17ce6a35e5b9c9c1aa5709682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 11:51:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e7f6be033c54d8ba7c7ec9e8a457c43 : SUCCESS in 15m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e55457be3a9f47b79791edee002f41bf : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca240d6644348bb943b9c8c4d9b6022 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d532bc4b9c5b47878f55510e029562fb : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0511643d2bb74f5496d16ea5f8194b93 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4d651c117f48ec85ae009e39536bf4 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6dbe067f3884732aec5f366d96012be : SUCCESS in 7m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bee40a396c8649cabd0bea9b91b68d2c : SUCCESS in 11m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7fead395e2f04c3eac52080ac659ab0a : SUCCESS in 10m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccd3a2402d824d23bd404af3ede1f9db : SUCCESS in 10m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/05290e46aaa9498c8265816aa589695d : SUCCESS in 51m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c859ff9d9eba467ca92f0a4b638c634b : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/22f0c7d932704fc0a5e0ca94660bb3d8 : SUCCESS in 1h 03m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8737ec9be95d4975ab91f431b4204a86 : SUCCESS in 57m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96ce93e4a5324dd0b46e61d3c46f53ef : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d5eb320f7644dd090eb2b23882b0457 : SUCCESS in 2h 02m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a77db6b5619840e29e1937293980548b : FAILURE in 56m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/797a12642bf542d5bf78c78b54c74ce2 : FAILURE in 31m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4829c690483346db852801f1494455b3 : SUCCESS in 1h 54m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a72c35e0ddeca8933656b765bd287857dc170a9f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-02 12:44:36.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds to cherry-pick + squash https://review.opendev.org/#/c/737731/ and https://review.opendev.org/#/c/737460/ in this change not to break SDK team.","accounts_in_message":[],"_revision_number":1},{"id":"36729666f126f075372153c45203c06ddb321b5f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-02 14:56:58.000000000","message":"Patch Set 1: -Workflow\n\nOn second thought, will cherry-pick separately.","accounts_in_message":[],"_revision_number":1},{"id":"20fc6bf82c5a4a65b2e87c0c85fc521e45fd99e4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-07 11:19:54.000000000","message":"Patch Set 1:\n\nNote: this will break amphorav2 for active-standby case.","accounts_in_message":[],"_revision_number":1},{"id":"4513c61b174efecf61b1fa8fcadf601349d2c9ca","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-07 11:37:06.000000000","message":"Patch Set 1:\n\nAnn, is there a patch that needs to be backported to fix active-standby in amphorav2 in ussuri? Or have I missed something in this backport patch?","accounts_in_message":[],"_revision_number":1},{"id":"9bde791bb56dbc2be60fdf6fd89003515e9ba864","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-07 11:51:36.000000000","message":"Patch Set 1:\n\nNo, nothing missing - it is WIP https://review.opendev.org/#/c/738292/","accounts_in_message":[],"_revision_number":1},{"id":"5b84860f985aebf47ffc172f79346f618b488c9f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-21 20:46:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e30dab39b1e3b7c0ee31dca41328d457e71e9521","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-22 14:21:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8088a6365c1ec28cb81d5205abc83bfa019696e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 16:22:21.000000000","message":"Patch Set 2: Verified-1\n\n(989 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6be5e7d2efc9430083224330eaecc80c : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cd43933696445ddb2e717f8abef7983 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2cc598f58e49afa5bbdf05ebcec655 : FAILURE in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b25bfeeee7524ea887a689d58557752a : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54d95c2e8687453394fda12cbff229db : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1fa5347913f49759168a358b012a138 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/816c9a48c0284e4bad8a746bcb032883 : SUCCESS in 8m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de18f6eb69744de88949eefa74acf594 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1a5129c40983424cad59d8152060ca0d : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b644bd4d98b4a79aedf29e63f744e5b : SUCCESS in 9m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6920320223ab421e874881046a61de40 : SUCCESS in 53m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ebafe4537ac46f8a7da04082def0c8e : SUCCESS in 1h 19m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d71b83ee6821460ba6323a88dc0e0049 : SUCCESS in 1h 00m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/decc21919efc4964a39b66551e8d560f : SUCCESS in 1h 01m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78d46115be034a3b9b9becb7412c0694 : SUCCESS in 1h 06m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc32641748694c4f9fd528385d055f8b : SUCCESS in 1h 56m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d41e1efe53b348c4ac80932eeae08135 : SUCCESS in 1h 25m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab3c2f7d278e43d083ea06b68642060d : FAILURE in 1h 12m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c146be0554ed405db45d763ee22dd196 : SUCCESS in 1h 22m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/drivers/keepalived/jinja/jinja_cfg.py\n  Comments left for invalid file octavia/amphorae/driver_exceptions/exceptions.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/distributor/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/cmd/agent.py\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/certificates/manager/castellan_mgr.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/controller/queue/v1/consumer.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/cert_task.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/healthmanager/health_drivers/update_db.py\n  Comments left for invalid file octavia/controller/queue/v2/consumer.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/cert_task.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/common/keystone.py\n  Comments left for invalid file octavia/common/stats.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/network_tasks.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/base_taskflow.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_updater.py\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/v2/types/quotas.py\n  Comments left for invalid file octavia/api/common/types.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/availability_zone_profiles.py\n  Comments left for invalid file octavia/api/v2/controllers/provider.py\n  Comments left for invalid file octavia/api/v2/controllers/quotas.py\n  Comments left for invalid file octavia/api/v2/controllers/flavors.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/api/v2/controllers/availability_zones.py\n  Comments left for invalid file octavia/api/v2/controllers/__init__.py\n  Comments left for invalid file octavia/api/v2/controllers/l7rule.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/volume/drivers/cinder_driver.py\n  Comments left for invalid file octavia/volume/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/flavor_profiles.py\n  Comments left for invalid file octavia/api/v2/controllers/amphora.py\n  Comments left for invalid file octavia/api/v2/controllers/l7policy.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py","accounts_in_message":[],"_revision_number":2},{"id":"8060aac54f285973fedc9a7154f2d25fd6ef94ec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 20:23:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"cf71f8ac302382fca0f2339eb426bbaa81242bd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:34:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8684973276d24e76aa42598763076cfd : SUCCESS in 10m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e1cb5718a9243808f413d2af18c970e : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72fa48b4bc24c7e9c948343cb3d6932 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/430e67902b88486a932dbbbcbfa896bc : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e03c3611fd241e082cff306fb842641 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef294bcb5df64719a1c2185a0709e5ec : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3405f076f3944645ae34ac86bae68a2c : SUCCESS in 7m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c3f2656aaa744fce84976cb5bc451f59 : SUCCESS in 7m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b42f6513e5a5414aa27d42c4c0281e50 : SUCCESS in 9m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a9e03e8ed3645e6b0d2b568bdba509c : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/587bc6e004914f299a7d90c1e1f04dfa : SUCCESS in 50m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5f0380293654c048d7e6b1053cd0cd8 : SUCCESS in 1h 10m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d8840e804cc46e0add8f762ab03aa56 : SUCCESS in 40m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c3b4e427130248a79605cfded0450926 : SUCCESS in 51m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18b1b73be072448993772169ca50cb2d : SUCCESS in 46m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b982fb85ea946bc8a9b713fd1edcde2 : SUCCESS in 2h 07m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f6961ece199b42ffa8e57022cb0ff92a : SUCCESS in 1h 06m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a53f4579662540f4bbbaa7536df3a073 : FAILURE in 48m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/786e1a2fe90a4fd8b723b089cc178515 : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"64daa6c73e2ee47c54ab69ce30711e41318c4625","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-28 19:01:23.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"6bd9a7760bc5c77aaca4d7adae2037ca174ae9f2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-07 07:25:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"df56fc61d903d702aa7335ce11033acd61568849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 07:25:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1afaf67f42b127097a8bee74f5b338a22d2603c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:23:10.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/721536e0eafe476b9e49db8f9f65238e : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f10c3a8f5c242efb381dfcf9927e065 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d22ad9e9e0a4ebd82982f44a2591cd0 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c64e94a7cf4c49a0b17c23479874047b : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906e1a86e85d4fcc915e18c76337819c : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bddebb2deea44f19788ba095b8c553a : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/281cfa81df4f4ad195edd28972f814dd : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d2cc63a45774bd187a2a8e5bbead6f2 : SUCCESS in 44m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f760c93a8ba04263aedcbed909117db5 : FAILURE in 53m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6cb6fa28569f454ba0991fc9e13b24ae : SUCCESS in 35m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3ae15cdd416a4790ae78256440ce43bf : SUCCESS in 37m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea63c2c8bbbd4a258a459064f018a4b0 : SUCCESS in 33m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb0577e05ef441e1a709d5742a7c863a : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"f21fa9eb38375638c1dfb03b9e8244b0a19945d6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-07 08:26:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"33d9b5596ca5cd5bbe2345639360ad88819a626b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:23:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afa8dc3efd5c4b06844b5cbfade6acb9 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f096b52258284df18370a0a2d72f2cbc : SUCCESS in 23m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda02ccb821449338a43064c3fbd81e1 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5a4ea030a9b42d39b08ebd9684a6e4c : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dee1611ee11242229f6effcf1abc247b : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c5697b5c1143da9e71afaf38da4c53 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4e3901baddf4b65b2ad1525c88e4009 : SUCCESS in 5m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb9378d6e20e4c9b89bf85784d8aed58 : SUCCESS in 5m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49ef96dc79664c88aa3b5aed20459d43 : SUCCESS in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9302dcbd52a24d9fb96b6515621d870c : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6cbebf88aff14dfe935095f06d3523d5 : SUCCESS in 37m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0219b6e141124975816599d5724a334a : FAILURE in 24m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0524a7db0a4f46ef87478875635f50f4 : SUCCESS in 40m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/08366758b17443b69d1c3b79c22ec61c : SUCCESS in 38m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41492e2bb84b4c9193187f77f52a19d6 : SUCCESS in 35m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f3817de305864ead8be6bd5cd068b3af : SUCCESS in 1h 51m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/58008eec4617445186c5ea0499707ff4 : SUCCESS in 55m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec6de809dbbc40bfbc6b10a099d18630 : FAILURE in 45m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86eb576b3f3b419484af4ce19d14804f : SUCCESS in 1h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77d27a7d19dee606bf51ab68ffc97cc264b38bb2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-08 08:46:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9255588a0a5898cec2d7eb46e85e7d9525803a89","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-08 08:47:07.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a9f255007d0924cc3ae381cba04d5d2a205cac05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 10:58:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc3a1e2680d44611bb9a87d69cf8cfc4 : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b3f5aa9f8a4470dacb516d678688819 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad736b9854464b46a21933d37055ad23 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7a84f83d8be435d9cfdec87c316778b : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e390aba0c1164f8da47a048a23d1d8b1 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9211790a57f54d5a83cdc99bff6a4105 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9365009c6588407e9e42e0c76b3c13a3 : SUCCESS in 5m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1718f32d87214660be527b76a16e3206 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eff5dbe0c9164f3ea00941d383e3e6b1 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/14ba453975ee49e0b017a3b1e1cce3b4 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/36fd903b72884c758df5d8c92946e4ad : SUCCESS in 33m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdcff991eef9402b8220467f874553d8 : SUCCESS in 1h 15m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b3dc440b5db7476c90c0d693c8f5eeca : SUCCESS in 31m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d99e5e5ddd9745f5948bb4eb0891c171 : SUCCESS in 31m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25b2c2f0cd6e468ab9d2a94e0f017be5 : SUCCESS in 40m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a85d878ed43149f698626b3950e73beb : SUCCESS in 2h 01m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ef3f47dc130d49fcaf75a47aa5fd28fd : SUCCESS in 1h 23m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e0dfa427df145e5b27d2a98944ec9a9 : FAILURE in 43m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39010a19cec74281a8ba60ce864b2bed : SUCCESS in 1h 02m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"161ec210d12ed1df9d11a9b0efaa711cf5fee8f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 10:58:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d1f15e8dc5e19d8f07f29927c94e9852e3981058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 12:35:40.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15caf508d6b64a61b9648681de17df55 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f7b0744933446683775b2ffcd567ed : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1414e2f1244c4124b2792e684b62b486 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afdc5efaf37e466e9bacfa979c584e72 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb21833986948c188a2e0496cec1821 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/424bc39939d949998346c6711b6f43d1 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f459f228dcb3477a8cc08812d2008c5b : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6c1f243237945e1a8fd34f2f27cf867 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11dbfcd206a14b31829fe435deedb546 : FAILURE in 1h 34m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcc9d9a5ca9c4d0ba4f8bfa4a733e618 : SUCCESS in 41m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f3f6d46202134959bffed1dc1c9bac39 : SUCCESS in 1h 16m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c0760ae70ed43ad976513e6774a6ba8 : SUCCESS in 41m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1c1fab23031c4e8f9bf63d46166ff18d : CANCELED","accounts_in_message":[],"_revision_number":4},{"id":"17419ce7b5be8e0b6b7626acd9225c794fea909b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-08 12:46:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"72857126fbec40e194be67089076ca06b882c8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 14:41:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b882f006f8248b98cb7c1e9f0ff475a : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02868cb6645c431b9deebcdf4a165dfe : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acff3c6c815e423192670da28b7bcf71 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eff144c1098e48faacbece89f99f8372 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7800bdeb85142ca8ec6e2b99c33342e : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcde78ee815456f915c1243a416bac2 : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/108b20e33141494a8c95ac6d008b4331 : SUCCESS in 7m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/10dccd172120405e83476160c798daa7 : SUCCESS in 7m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d530344e70dc4ca2a710bc72b4956785 : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/738852ee137f42db93d213418aec562b : FAILURE in 10m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8f4d273ab5b4dd5b202b78c3e228a1d : SUCCESS in 47m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27893bef5ceb42799835994d1748367f : SUCCESS in 1h 14m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0c2316325e14140b8e3a5ac4e457054 : SUCCESS in 41m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/afaba2dc74334552b963e8f8e984e997 : SUCCESS in 47m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1781ad604a0f40868700b8868dd4d2d9 : SUCCESS in 40m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0334f70d3e84b149d2baaf97f11b4b5 : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e1a73c1afa4544068f388cb5351327cd : SUCCESS in 57m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35fd39fe99954b7196d54f3e09383cf3 : FAILURE in 54m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f3eb135959704dc88dcff09161a52210 : SUCCESS in 1h 09m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cba2da2fc102d6b0238c6b72020137f9b74aab72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-08 19:59:25.000000000","message":"Patch Set 4:\n\nrecheck\nnova failed to go ACTIVE","accounts_in_message":[],"_revision_number":4},{"id":"15028e75d2baf1958e9952d5f66dc5e1c7b38577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:48:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f19f270c133642c3b6a048934859c731 : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a477826dcd1042e392c2fa22faafc076 : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6e6f8527d24d29ae5ae4e32f6b9017 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a07c7d1e70c94a27a68e056773abd1b8 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a4c00afa9374b5093dc020679e38696 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03edcd2b0d924f3ea29666b2c021b315 : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46d066eec770402a828e774f84ec42e5 : SUCCESS in 4m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/94d540ea04324d50b453ea819e762bc8 : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/12f4f5bae2a342959f263a1c0aceb3b1 : SUCCESS in 11m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f069098aa8bc4823975a79e55b99170c : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb6e21a785af436ab84e9a39b668c847 : SUCCESS in 40m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65a8227fc8544b5da57d87b1d76fb29e : SUCCESS in 51m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a240431bff8345608b964a5aacede1a8 : SUCCESS in 39m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/70def00a518b4c6e94e85bcb22cc38ea : SUCCESS in 37m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2327c2979f214443b0e4339aee03f322 : SUCCESS in 35m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9aee8802b71c4321a9096ca9cacad7cc : SUCCESS in 1h 45m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1643150a3a2240efad87aeb1630ae36b : SUCCESS in 57m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/babf7739f2734f658b2dc5aaab605f97 : FAILURE in 54m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a4d43688dd554db8a9abab0c19c97696 : SUCCESS in 56m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b3cb3ccb0f4b4621550e9250a09e74faa4f7bbc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:48:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b32d80793a957de4f7170a2af853e21a204c2a74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:42:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bb4e27fa58344e8934d0824f12ac9f0 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d739f879943e4ce3bb828fcdef1bd068 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9db27d5f993b45beaabea017fccd78b0 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fa118df84d7495c9f70d0adaaad317d : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37158439100497b99bf8ba2e5ef6bdd : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14b7708b2bdd4501b9f3326bb4e48fb8 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5a6503dee944355b7af88b9ff20b5d3 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f558d60f5b7c468fac3781aa928a2a07 : SUCCESS in 46m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1a5bae800964ec5a8761711328b852d : SUCCESS in 1h 02m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f83625b36d94ba19de5d4e0d1cc3287 : SUCCESS in 50m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e14a489f763748108b7a263f59a7e2c9 : SUCCESS in 30m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbabbc039c7a4d43aa2b2b2a1c2eb7ab : SUCCESS in 31m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b7b921fdc874276a1e1f6a219476911 : SUCCESS in 1h 49m 50s","accounts_in_message":[],"_revision_number":4},{"id":"ccabf7c75148d1ecd494fd7329cb286580a0c511","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:42:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1fceb7ce5280816459eed421e9f0bb6f638bd4d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:46:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c872f0604dcf4ac4a43aac7cf3c03d39 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd5981e89bfe46ccb8f631a223466639 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bfa44b7576c2e4adac85f5a95fe2783bee907dab","revisions":{"2f9dc3693e73fe8c235e0eaaf0f9595576940aae":{"kind":"REWORK","_number":1,"created":"2020-07-02 09:36:29.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/02/739002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/739002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/739002/1"}}},"commit":{"parents":[{"commit":"af902186c029f66540b5299621d6af74b64b5e19","subject":"Merge \"Fix netcat option in udp_check.sh for CentOS/RHEL\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af902186c029f66540b5299621d6af74b64b5e19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-02 09:27:32.000000000","tz":120},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nConflicts:\n    octavia/common/constants.py\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n(cherry picked from commit 955bb8840616d96ed74de3086f8959ad4190a472)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f9dc3693e73fe8c235e0eaaf0f9595576940aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f9dc3693e73fe8c235e0eaaf0f9595576940aae"}]},"branch":"refs/heads/stable/ussuri"},"a5aec3f10a49d449ddfa5bf008fa647a8f990df1":{"kind":"REWORK","_number":2,"created":"2020-08-22 14:21:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/02/739002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/739002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/739002/2"}}},"commit":{"parents":[{"commit":"0462ff457c14dd1a45366dd208aa478b52bfda9f","subject":"Merge \"Fix accepting \u0027insert_headers\u0027 when unsupported\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0462ff457c14dd1a45366dd208aa478b52bfda9f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-22 14:21:25.000000000","tz":120},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nConflicts:\n    octavia/common/constants.py\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n(cherry picked from commit 955bb8840616d96ed74de3086f8959ad4190a472)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5aec3f10a49d449ddfa5bf008fa647a8f990df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5aec3f10a49d449ddfa5bf008fa647a8f990df1"}]},"branch":"refs/heads/stable/ussuri"},"6301cee688ffcbea13b781bb30ce6d6f947a47be":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-25 20:23:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/02/739002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/739002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/739002/3"}}},"commit":{"parents":[{"commit":"5e377fa8a73fa2d9c5b6fb83c585abebbaf2b18e","subject":"Pin pylint for stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e377fa8a73fa2d9c5b6fb83c585abebbaf2b18e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-25 20:23:42.000000000","tz":0},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nConflicts:\n    octavia/common/constants.py\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n(cherry picked from commit 955bb8840616d96ed74de3086f8959ad4190a472)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6301cee688ffcbea13b781bb30ce6d6f947a47be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6301cee688ffcbea13b781bb30ce6d6f947a47be"}]},"branch":"refs/heads/stable/ussuri"},"bfa44b7576c2e4adac85f5a95fe2783bee907dab":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-08 08:46:57.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/02/739002/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/739002/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/739002/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/739002/4"}}},"commit":{"parents":[{"commit":"39eb3286024d1118fd67e80df67bdd3abed7d5b9","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39eb3286024d1118fd67e80df67bdd3abed7d5b9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-08 08:46:57.000000000","tz":0},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nConflicts:\n    octavia/common/constants.py\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n(cherry picked from commit 955bb8840616d96ed74de3086f8959ad4190a472)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfa44b7576c2e4adac85f5a95fe2783bee907dab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfa44b7576c2e4adac85f5a95fe2783bee907dab"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
