)]}'
{"id":"openstack%2Foctavia~739187","triplet_id":"openstack%2Foctavia~master~I7312c8c1057618e909339aa7a4dfeb836f4b8f33","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7312c8c1057618e909339aa7a4dfeb836f4b8f33","subject":"Add requirements for jobboard drivers","status":"MERGED","created":"2020-07-03 09:21:30.000000000","updated":"2020-09-10 22:04:55.000000000","submitted":"2020-09-10 22:02:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739187-1599775349685-9fd2f9ed","meta_rev_id":"b4f7143eae20f906ac5f11e3736949f0c2df1274","_number":739187,"virtual_id_number":739187,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 22:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-10 19:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-03 23:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-09-10 19:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 15:07:55.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-23 21:11:41.000000000","updated_by":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"reviewer":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"state":"REVIEWER"},{"updated":"2020-09-03 23:00:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-10 19:48:52.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-10 22:02:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee312b58ac4936530d824f1bd30f09aa08ea8c51","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-03 09:21:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3636eb81b3615192a550264d15d85ecf6ffc5083","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-03 09:30:19.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"55e64b6ead24d280fae84fe47d6612f41ccc0a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 11:24:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2323356ea4ca4caca36068122b48206e : FAILURE in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ebcbc5ca2744166b2073f04214c6770 : SUCCESS in 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2587c26b9bea48d3ab0093b0d16fad4d : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196fcb74b9dc44f0878066138dc0b4ee : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e0fcfc417614220bcd4da01b277d210 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0243196301d64c238da905b559938397 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42aaa59f410642e29580c4a8677528ca : SUCCESS in 12m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5bea2186afcc4957bca265b3e1fd29e5 : SUCCESS in 7m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/445c6b0f064d4eb5832a8a42330b2947 : SUCCESS in 9m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45e0a8c628854b788b75efa5e9e17d3f : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99d452d790e54b7c9d134699fe1d2d34 : SUCCESS in 47m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/588fa396712a41379176985d5b14b265 : SUCCESS in 1h 51m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/303c0f5d661147e6b83ddc4a4458d129 : SUCCESS in 33m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ba571b14428493aba532a5722158823 : SUCCESS in 34m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/299ff1d878b4436ca7bb1f1513662231 : SUCCESS in 56m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cee6f2f56b0d434f9a5cf3a5f89eb9cc : SUCCESS in 1h 51m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eefa79f46be34e5594068be44a388f3a : SUCCESS in 1h 56m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b020480b66e4361b31351f88e2fba24 : SUCCESS in 1h 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d92c3bf419269dcc89c02e8a4d3a370f39d4eff5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 12:42:12.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/617d69a92a8041ee8301145de78c16b8 : FAILURE in 43m 15s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/13c5cc27c70d41349539b3a0d859ed55 : FAILURE in 57m 51s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc492425fc924cdb9b3844381b56de6a : SUCCESS in 53m 45s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96f2301c7f29459995b4acb53430bf02 : FAILURE in 48m 24s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ff5b95ab40347deb44f16f04d75962e : TIMED_OUT in 3h 05m 32s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/355c7f5a167e41cc96c20829c96fff84 : SUCCESS in 1h 44m 15s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/0131cf14a34f476db46f07cebf6900ef : SUCCESS in 2h 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"af785cd5b3d4bf1123d5d55cc573009ec10f623f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-06 12:39:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4adc4db648f155dea1d2c176c0880ac6da648872","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-06 12:44:15.000000000","message":"Patch Set 2:\n\nCan we add such requirements? \nAt this moment in devstack it is setup separately [1], but I guess the proper way is to have it set in requirements.  \n[1] -  https://github.com/openstack/octavia/blob/master/devstack/plugin.sh#L250","accounts_in_message":[],"_revision_number":2},{"id":"c352ad0cca09fb109bae0202335a2c77fcd58b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 14:47:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a61b7601e264e6d975b2265c6fa6105 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dae50692ff0b4b5fb7be7ca342d0ee46 : SUCCESS in 15m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86b52022a0a84898bade631ca217a524 : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5caeb575d6dd4611b41653245b6bff66 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad153049ee304a4480c1b70eced546dd : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95e1044fbf484e42bac6de5d31618b91 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3917954a05004d01a7fb9023ca4767b6 : SUCCESS in 12m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b1568f4996604c3ab8a0f1a83c721269 : SUCCESS in 6m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3a11bf9a9dbf40b6b1225a4fd6e1e096 : SUCCESS in 8m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4e15d2387224c6ca7e5ae27c854ecec : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc122a071e9245a28c4fc21d651f3d53 : SUCCESS in 43m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cc7c21003b945139684e1989de319e2 : SUCCESS in 2h 00m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49cb7f8546a0423aa2e159d2df1362c2 : SUCCESS in 1h 01m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2a372d6caa324f6d98e555591d8055fa : SUCCESS in 1h 11m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef320e8ee9b245b7b8d5d82df65b511e : SUCCESS in 1h 01m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9acad3e90195487b98e45a2e443d70b9 : SUCCESS in 59m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3bc4794679c94f20b4d719dc3c46b998 : SUCCESS in 46m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd7d4f58317a48638dbbd989ed48ff69 : SUCCESS in 1h 57m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d6e20e61cf994c2e87d1ce2db8b65592 : FAILURE in 1h 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5561214e33f5c8307aed2ebc15abcb285ccb4b5d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-06 15:07:55.000000000","message":"Patch Set 2: Code-Review-1\n\nIs Octavia the right place to add these requirement libs? We do not directly import redis nor zookeeper. I think they should be requirements in taskflow, which appears to be missing.\n\nhttps://github.com/openstack/taskflow/blob/master/taskflow/jobs/backends/impl_redis.py#L31\nhttps://github.com/openstack/taskflow/blob/master/taskflow/jobs/backends/impl_zookeeper.py#L25-L28\nhttps://github.com/openstack/taskflow/blob/master/requirements.txt","accounts_in_message":[],"_revision_number":2},{"id":"da82814eefbec09efc143a70e84d6a8db75a3ec0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-07 06:58:36.000000000","message":"Patch Set 2:\n\nFor some reason it has them only in test-requirement.yml https://github.com/openstack/taskflow/blob/master/test-requirements.txt#L10","accounts_in_message":[],"_revision_number":2},{"id":"155062e18892d8c39bb0c39f2627ddb67abdc6e4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-07 07:33:30.000000000","message":"Patch Set 2: Workflow-1\n\nI proposed https://review.opendev.org/#/c/739670/ to taskflow.\nWe\u0027ll have to bump taskflow minimum version. Leave this change as placeholder for that.","accounts_in_message":[],"_revision_number":2},{"id":"25aece43d4d2d30aff55b18b4263ef9cc5276106","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-07 16:52:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10d700015d28bcc18fe8e298c13525b5460c8bc0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-08 07:15:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4719b0d0eb5b871a1522e3e4f65a0685ec8a1926","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-08 07:34:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff760cfe0e8b6cb52fea8b0cee01a580009232e9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-08 07:36:35.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"5deee6ff34bc90e8112d33fae1ff565882f45b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:34:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/29659fa98f99413cac8a0084b6d4d836 : SUCCESS in 5m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f80c5e1abf754ea59cfc04b99b5553c1 : FAILURE in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e11c1d946a1c46eeb9b612ff03b43137 : SUCCESS in 11m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a821e9e07f8441ab79d69d648f74772 : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3427f0d8a4e74bc0a41ebff76cd7965c : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa0194d4c8474857be66a976610f4614 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602b4bc0a1354d7d88be45c2459fb17f : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3cebf8c07448cabef67f65cdac8767 : SUCCESS in 16m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5e91082678145a5a5877958a29f4099 : SUCCESS in 10m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f759b8c9dfee4e659fcdea6e0e16bcbf : SUCCESS in 12m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c706261dbea43219ceecb3cfc312552 : SUCCESS in 10m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec3d787b9a43418bac644eb21688811d : SUCCESS in 48m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c892654366054be8a017df5926047004 : SUCCESS in 1h 44m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/20bd5ccff0d94fb0a898b47a9eb0ed52 : SUCCESS in 35m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/439dcc2f26334834bea5d70f515a1b8f : SUCCESS in 1h 00m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/879ddafa1e044d5a895946e157a7b17c : SUCCESS in 1h 11m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/014edeba43cd486691fea839be41279e : SUCCESS in 1h 03m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1450f2bab00840a8a4525fd411ce9ea8 : SUCCESS in 52m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1802d9c207774a50b453d20be3977232 : SUCCESS in 1h 51m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4cbca3d059264e29b38d19399bdbefef : SUCCESS in 1h 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b4d03fbbcd9f5d18261571e3c7aa793a2c6c1660","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 10:52:14.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/506392c13c06449ba1d8c94f245f91be : FAILURE in 46m 19s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/739bd578a8bb450a95e12a938cbb7a8c : FAILURE in 1h 40m 09s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/20bc9fd3ba5b4848b0fc193c80be22dc : SUCCESS in 56m 34s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/623d3dbddf2044c5b183d8eaf544f474 : FAILURE in 49m 28s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/42eb30d96bb74606b4fbff60470e0cd8 : TIMED_OUT in 3h 06m 42s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/fa048dba8b3d473a8b1d8d245fbc1adb : SUCCESS in 2h 04m 28s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/46a9e903a1634c3f9f009f68d7669162 : FAILURE in 37m 10s","accounts_in_message":[],"_revision_number":3},{"id":"4de3ab609168ee6bf4332cfce51590ea71ef9cd6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 08:38:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2e5e3c13cc2df322e3d19a54a30454cf7ffc34a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 13:18:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52e5f781fd9e4c6a8fbd6035413a9011 : SUCCESS in 3m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e592090071cd4b33a959e6101dc5dc65 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85111d418e6342a19bf14c49832b4d29 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcfda0de0b07414782cee05b8d1e797e : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8347ca9d0bd94853940242b8ab45ef14 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb728e9955224312974e00bd0ad268ce : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecfa9b307297497c83228ee168251663 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d190c39a60e940c0b1171f072ec3a934 : SUCCESS in 15m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/97c43bae4cda440b841d3d36cc60f30f : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c7f4252646c0436cad982c225797ba2c : SUCCESS in 10m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a88e5bf2827d46ba8236d0d31b46e5a1 : SUCCESS in 9m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42b6466938f44628850980c7a4f67a60 : SUCCESS in 43m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4157b542aadb4024b8afe78fa761be14 : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e55fea95cc1c43fbb231231620831d33 : SUCCESS in 54m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/87c9f58c2c2348f78b70a569af189506 : SUCCESS in 44m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef8386e0860c4c34ae5462957bdd228a : SUCCESS in 49m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d2e8e14725e94b068c70d5d9de4c5ef9 : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c912598d9d8c45e49e730af469e17078 : SUCCESS in 1h 48m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/98c3fcaf758a4c90a824ff4e3ceee1e5 : SUCCESS in 1h 47m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/acdf0eb3359249bda3f32ace70e4efc5 : TIMED_OUT in 2h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8e3de74b59be4cc5a8c6c25414d54a94 : SUCCESS in 1h 50m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8940ed4d8f3d4cda183a48bd8049ee7a856cc6d0","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-23 21:11:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e12108dc634673c6c9e8e9c47958d35b25cf43ea","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-27 08:14:39.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"bbb9b15ed4154b5bc8fe9fd309a489450cdf167a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-27 08:15:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f3b33cab61d5f69743b8fa4234bb75452315e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 10:05:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4bba08a7408341fa85327fb4832874ba : SUCCESS in 3m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5cb988a8050145f084b4cfcf55d86c2f : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e8b90d9fcbc441ba2a30ebb1ed78804 : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/392812d4bbc7465d8117fef017747c26 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c429e2e269ac43229ff383431f0eba25 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c54037f813e04771b0248ab293e4f5ef : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/074b8184bd7e4dd0bd4cfc75c1480a99 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93393cc808a24f85b148d54d17252f25 : SUCCESS in 13m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5c1c4aef9ead4f5891ad4b993e0c5e19 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/17e82f2f4d7549a086c2801a9356f750 : SUCCESS in 8m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f13bd104cbe0470b9c7b2c71749348b1 : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/532b4b6d4b134b2cb8ae0da10387c5f4 : SUCCESS in 45m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55e2c324fa5342c2b1e10d2b7f82e15c : SUCCESS in 50m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9bae0054013645d7b513828ec274c77b : SUCCESS in 40m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d4b0e5abc5b045b1b1c837ded6e3f0f0 : SUCCESS in 37m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca75e26b38f6412282d77dc67e502126 : SUCCESS in 32m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a263b6f25da4b449bd3631cf9821906 : SUCCESS in 1h 44m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/67110ebe78ba43f492c752995ff85af3 : SUCCESS in 54m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/caf864741e374b58908feb23926a2afa : SUCCESS in 1h 42m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5110763486cf44b48614b61221cf06fd : SUCCESS in 1h 12m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/081b3267b3774a26b1ed6136e907642a : SUCCESS in 44m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4a7d05ce2a46484e9c31b4f39d97bc1c : FAILURE in 1h 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ab20e8a93f5af1471a05c92c69483e3c9f0f9565","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 21:14:02.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c3af3d106fb7421e340227f9ed58d1f025d0844","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-27 09:53:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3d8b79a023c39d47cee77dffa2290aa393a97ab","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-27 11:16:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"368d54160d4d9b42466b37846cdc704b558ea3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 13:28:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a861396ab7024aed81ada40a78e6dd20 : SUCCESS in 3m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/80635579c6184e3d9977188673dfa01f : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4793b429762e4eedb17c6770b3ac5773 : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed61f6c070ca4486be8c9ad119d294fc : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d1f99622e94694bab3e2f5b792c924 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1890378124724a09964321bbf4ed62f5 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1616058aca534ff1a17682d423cddb73 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e1be9928e942e3b39fc9c29193e70b : SUCCESS in 14m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7f391ef9206a4d9dbc7f87fa89494a42 : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/096bae820cf44c2bba3f13ce940188c3 : FAILURE in 9m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3414deec533b41e5836005764b56550e : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ef5b16fd45142dda5ad5a83dd3a2b2b : SUCCESS in 49m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad852d2b494c48cb99352dc2b18cccb0 : SUCCESS in 58m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e4c1d98578d84af78049c88f37f9e747 : SUCCESS in 49m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/708cbef1ca0b4a049c8994df08e1a517 : SUCCESS in 41m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23852a0bad304dda85bd29bad17a33eb : SUCCESS in 43m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/15ca5c6024d44ad5b6d790d9f9c29345 : SUCCESS in 1h 54m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1aa7e2f3a85d47b8abb208a7c2daee20 : SUCCESS in 1h 22m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/beeb4c25dd6d4caab051d99a59b2c309 : FAILURE in 15m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a65cc571397244d78a6a394a9ba7ab05 : SUCCESS in 1h 09m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9080620677284946a4dfb0a6657a7cc5 : FAILURE in 10m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4a80730713b546a19bee3191cb8b9dd2 : SUCCESS in 1h 28m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7b335c6d26584f13ad294d6605247c49 : FAILURE in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6897fd392a549304eb38ca88478d16a8de998b82","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-31 07:29:31.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d98be5526a854af8a00737510c757e5ab9da29d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 08:51:06.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ead805c0c3b4ebe9393a5ee0b8e2435 : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e548e26b6e9f425da77cf263b2a50366 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e56647e7d97400db6a9b083ff6f148a : FAILURE in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16a6c76c710e4c7a9725e43520b9984e : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ce88524a9745b788d8766ca866d572 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dfc65e6e9b64984a81ee41851e9f636 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/965f009b33b64726bfc3808ae4dd0a70 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8dcff40d7b4424978ff051e83eda93 : SUCCESS in 12m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e11de398cb884f0789a8cb93bbacde76 : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/030ace878a4f4d1682df394b80dd57ed : FAILURE in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9958bd990bb4aeaa2d243b348388838 : FAILURE in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64f3992f4b824fa0b2029eebdb20fc17 : FAILURE in 19m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17193d1a07a5495aa2c5c60e83796f83 : FAILURE in 15m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf2ecc5a71a941da902a43d6a22ae8a2 : FAILURE in 16m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/da33fb49b76649a6be7cd8eb4758535d : FAILURE in 24m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff3f6898a1c64441a5ff16b72a5384fe : FAILURE in 15m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff87c6a3b95144428374b8a23c2f5576 : FAILURE in 26m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a9c554271ac4ec89c79a2cb33d2e287 : FAILURE in 16m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b736d72f7434bfa833425af54f77855 : FAILURE in 17m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8b024392193b4a4a8cd5d19cad0fe6bf : FAILURE in 11m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/35540d580ef74c7b9c6f341e9a6727be : FAILURE in 25m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/40daeab7dbb243749503817466c9b445 : FAILURE in 15m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bcae8bf5797f40efb9ee76f30a8affa0 : SUCCESS in 1h 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"145081624a57f874b6d85126ebb72bc1ad17ea61","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-03 08:02:01.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"cabf740d3248d3553b78de406c0efc21aaad5daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 09:50:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7761de408eb4486583ae2f9d8902d22f : SUCCESS in 2m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7d6c2825c36f48068a7410859752d657 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79ecce7fee044d1d9a2ae2ec59b8d195 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/189b8484ec76488d89ac93de45c13a33 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892bcfa0058d4aa9b5d89aa92f94cc5c : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b444adfd0db47b2b24dd5b665114990 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/964985f550a0499989bb7b2aa46e1903 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6e37fc070e4b67b417b0e6f8d3d329 : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6e70342710394d7b848e8146b238ff29 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fabb361f3b3a41a09c358818db7fd628 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7afac6fa663c4b6c8963a5d3c1f0fbe5 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e51e2c7b5c04a8ca002e4c51bdbfe8b : SUCCESS in 42m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55abf09c65094030b3cc25925f8b76c9 : SUCCESS in 1h 05m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55ddacee5e594c1d89380440ab9a23b4 : SUCCESS in 48m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c28efb1364ef4d5eb6871a2ccf70c576 : SUCCESS in 41m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18636229833641bcb31e5365faaad579 : SUCCESS in 42m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e29efbc787324f7ea2b312fbe1c5247e : SUCCESS in 1h 32m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/862ba712a51f40e695ecf98cf6df01bc : SUCCESS in 53m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff5ab04804dc443dbd3ce55270d9a250 : SUCCESS in 52m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d7109a726c3041f692efadb9c29e6ebd : SUCCESS in 1h 06m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7f6336f3ee244a78a6876c223b237a89 : SUCCESS in 1h 00m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f3d2cec344e942eaa16552b0ad9427cc : SUCCESS in 55m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7353f2a140d2486b8a296296b74f855f : SUCCESS in 1h 32m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d85879a0d9570454de80a939ae2fac735cba0bc7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-03 23:00:28.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"ecac0e8b4d35ad976ddc32d7200d51edeee703d5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-10 19:48:52.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c0411f01bd77c9c08e4b2884d1f33f09c222fd7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 19:49:04.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e98023d4d593479d807a03afd510950a4a0e6514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 22:02:28.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8477b3e9add24f88b749f48dc372b524 : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb92f1ea35a347419b639b00417aab2b : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8799dd576e14688ae021171c6454d3a : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3eb1c020b6942beadf1114072e39ac5 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f88f6b8c40314bafb75059076ee18db7 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33db227cd47e4a388fd4c1004d150eaf : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db528bc5eed4dbbb8771d68d81b2a71 : SUCCESS in 23m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f0fe2bed0e249a59d0d15857ed97130 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5268c3bc344e48169537b4f8d50ee448 : SUCCESS in 44m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c59e0696829d428395d1668c6bec6048 : SUCCESS in 2h 11m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f9aa61a1feb4bf892668db1d11c62b9 : SUCCESS in 38m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/22ebb7a835be41b684d86f552cfbcef2 : SUCCESS in 40m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34cac67db3c64e119238fb245f9d7235 : SUCCESS in 37m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/590d2bd4f88849fc9879aca9d7140335 : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":7},{"id":"cfd237b753786618d0bdc5d2da020f8de5084ec7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 22:02:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"b4f7143eae20f906ac5f11e3736949f0c2df1274","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 22:04:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237c4f804e7945e68d62eb3cf8ba4360 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"460566e328e48e7d2cfee573b1f6805ebea77024","revisions":{"f747a55e73a0ffbb70a6e44a3268642162a9f4ce":{"kind":"REWORK","_number":1,"created":"2020-07-03 09:21:30.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/87/739187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/739187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/739187/1"}}},"commit":{"parents":[{"commit":"64140e276e2bd141d51c12d8b778255dd3c6ff2c","subject":"Add experimental amphorav2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64140e276e2bd141d51c12d8b778255dd3c6ff2c"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:19:19.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:21:06.000000000","tz":240},"subject":"Test","message":"Test\n\nChange-Id: I7312c8c1057618e909339aa7a4dfeb836f4b8f33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f747a55e73a0ffbb70a6e44a3268642162a9f4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f747a55e73a0ffbb70a6e44a3268642162a9f4ce"}]},"branch":"refs/heads/master"},"f0f618df9c89e7f422af0d163dd97870856fca3e":{"kind":"REWORK","_number":2,"created":"2020-07-06 12:39:56.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/87/739187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/739187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/739187/2"}}},"commit":{"parents":[{"commit":"64140e276e2bd141d51c12d8b778255dd3c6ff2c","subject":"Add experimental amphorav2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64140e276e2bd141d51c12d8b778255dd3c6ff2c"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:19:19.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-06 12:31:25.000000000","tz":240},"subject":"Add requirements for jobboard drivers","message":"Add requirements for jobboard drivers\n\nAs Octavia allows to use RedisTaskFlowDriver or\nZookeeperTaskFlowDriver we should install python clients that\nallows to work with redis and zookeeper backends.\n\nStory: 2007892\n\nChange-Id: I7312c8c1057618e909339aa7a4dfeb836f4b8f33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0f618df9c89e7f422af0d163dd97870856fca3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0f618df9c89e7f422af0d163dd97870856fca3e"}]},"branch":"refs/heads/master"},"e88bed3a5882e055a4c6e28a471727c44e2ca8f6":{"kind":"REWORK","_number":3,"created":"2020-07-08 07:34:36.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/87/739187/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/739187/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/739187/3"}}},"commit":{"parents":[{"commit":"64140e276e2bd141d51c12d8b778255dd3c6ff2c","subject":"Add experimental amphorav2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64140e276e2bd141d51c12d8b778255dd3c6ff2c"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:19:19.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-08 07:33:55.000000000","tz":240},"subject":"Add requirements for jobboard drivers","message":"Add requirements for jobboard drivers\n\nAs Octavia allows to use RedisTaskFlowDriver or\nZookeeperTaskFlowDriver we should install python clients that\nallows to work with redis and zookeeper backends.\n\nStory: 2007892\n\nChange-Id: I7312c8c1057618e909339aa7a4dfeb836f4b8f33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e88bed3a5882e055a4c6e28a471727c44e2ca8f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e88bed3a5882e055a4c6e28a471727c44e2ca8f6"}]},"branch":"refs/heads/master"},"0d5b018eef087a554bd796b3228580c1f93d0959":{"kind":"REWORK","_number":4,"created":"2020-07-09 08:38:13.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/87/739187/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/739187/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/739187/4"}}},"commit":{"parents":[{"commit":"bf196db6bc535de616aac4947405fedeec3d301b","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf196db6bc535de616aac4947405fedeec3d301b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:19:19.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-09 07:47:43.000000000","tz":240},"subject":"Add requirements for jobboard drivers","message":"Add requirements for jobboard drivers\n\nAs Octavia allows to use RedisTaskFlowDriver or\nZookeeperTaskFlowDriver we should install python clients that\nallows to work with redis and zookeeper backends.\n\nStory: 2007892\n\nChange-Id: I7312c8c1057618e909339aa7a4dfeb836f4b8f33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d5b018eef087a554bd796b3228580c1f93d0959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d5b018eef087a554bd796b3228580c1f93d0959"}]},"branch":"refs/heads/master"},"2e7d08a5aa5562bb06c39f6eeb1a15a5018a9d15":{"kind":"REWORK","_number":5,"created":"2020-07-27 08:14:39.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/87/739187/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/739187/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/739187/5"}}},"commit":{"parents":[{"commit":"bf196db6bc535de616aac4947405fedeec3d301b","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf196db6bc535de616aac4947405fedeec3d301b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:19:19.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-27 08:14:34.000000000","tz":0},"subject":"Add requirements for jobboard drivers","message":"Add requirements for jobboard drivers\n\nAs Octavia allows to use RedisTaskFlowDriver or\nZookeeperTaskFlowDriver we should install python clients that\nallows to work with redis and zookeeper backends.\n\nStory: 2007892\n\nChange-Id: I7312c8c1057618e909339aa7a4dfeb836f4b8f33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e7d08a5aa5562bb06c39f6eeb1a15a5018a9d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e7d08a5aa5562bb06c39f6eeb1a15a5018a9d15"}]},"branch":"refs/heads/master"},"c2837e97fe1153d6716388665d2649e290831191":{"kind":"REWORK","_number":6,"created":"2020-08-27 11:16:43.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/87/739187/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/739187/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/739187/6"}}},"commit":{"parents":[{"commit":"0b1d8dd5e76edc4b3058a36b47f6c059f970e516","subject":"Merge \"Update devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1d8dd5e76edc4b3058a36b47f6c059f970e516"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:19:19.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-08-27 11:16:29.000000000","tz":240},"subject":"Add requirements for jobboard drivers","message":"Add requirements for jobboard drivers\n\nAs Octavia allows to use RedisTaskFlowDriver or\nZookeeperTaskFlowDriver we should install python clients that\nallows to work with redis and zookeeper backends.\n\nStory: 2007892\n\nChange-Id: I7312c8c1057618e909339aa7a4dfeb836f4b8f33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2837e97fe1153d6716388665d2649e290831191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2837e97fe1153d6716388665d2649e290831191"}]},"branch":"refs/heads/master"},"460566e328e48e7d2cfee573b1f6805ebea77024":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-03 08:02:01.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/87/739187/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/739187/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/739187/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/739187/7"}}},"commit":{"parents":[{"commit":"4a4a2344de09d61aa253174d9a7ca242d116913e","subject":"Merge \"Add a message for the amphora image size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a4a2344de09d61aa253174d9a7ca242d116913e"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-03 09:19:19.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-03 08:02:01.000000000","tz":0},"subject":"Add requirements for jobboard drivers","message":"Add requirements for jobboard drivers\n\nAs Octavia allows to use RedisTaskFlowDriver or\nZookeeperTaskFlowDriver we should install python clients that\nallows to work with redis and zookeeper backends.\n\nStory: 2007892\n\nChange-Id: I7312c8c1057618e909339aa7a4dfeb836f4b8f33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/460566e328e48e7d2cfee573b1f6805ebea77024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/460566e328e48e7d2cfee573b1f6805ebea77024"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
